)]}'
{"id":"openstack%2Fkeystone~662690","triplet_id":"openstack%2Fkeystone~master~Ic51205c1403f2085d7f921d3ff9d667cfe3f4c87","project":"openstack/keystone","branch":"master","topic":"bp/s","hashtags":[],"change_id":"Ic51205c1403f2085d7f921d3ff9d667cfe3f4c87","subject":"Fix contributor doc of keystone","status":"MERGED","created":"2019-06-03 07:19:51.000000000","updated":"2019-06-19 16:45:09.000000000","submitted":"2019-06-19 16:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662690-1560962709668-91b8215b","meta_rev_id":"81a1d7295b25b5a90ef3d791354c5bee915140ae","_number":662690,"virtual_id_number":662690,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-06-19 16:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2019-06-19 15:53:30.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-06-19 15:55:32.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-06-19 15:55:32.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 05:54:14.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-06-14 05:54:22.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2019-06-18 15:59:33.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-06-19 15:53:30.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-06-19 15:55:32.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-06-19 16:45:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6332718ab58871749c5f0fbe5f80138b93b00e35","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-03 07:19:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbd7f30ee50250ab157be378f3c1b72098225a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 07:29:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662690/1/check/openstack-tox-pep8/f4ae62d/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/90/662690/1/check/openstack-tox-docs/87da833/ : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6a28f899e723a79401c656e041b4ac4e0a295c91","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-10 08:29:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0291fbe4698f9dd842c557bb98ab0c36870d5217","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 08:42:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662690/1/check/openstack-tox-pep8/2be76ea/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/90/662690/1/check/openstack-tox-docs/edc59b3/html/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ab1009945bd3fbf6dff000c4e503b3846e0fa459","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-17 15:46:01.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"139993aec9d2301c0a5f3b990713439f3c710a31","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-17 21:22:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3787edf634fcb8811f76bf8c9cbe063169af3a1d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-18 06:18:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"26d5ca7c873714a9a4f6aa4a83404c8ad7912014","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-18 06:47:03.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3e601d7e0a762dd65bea4df10f397bf639b19046","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:41:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662690/2/check/openstack-tox-pep8/4400fc7/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/90/662690/2/check/openstack-tox-docs/7c07b88/html/ : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":2},{"id":"cfc6c32d72ed8e0b6f45a5f1b5531c39ad5b9f89","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-18 09:03:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"888ed5e7db0d44437c542b9885f60f6e609e0a89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:25:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662690/3/check/openstack-tox-pep8/4e02edd/ : SUCCESS in 6m 11s\n- openstack-tox-docs http://logs.openstack.org/90/662690/3/check/openstack-tox-docs/e90b13b/html/ : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":3},{"id":"aa2ab1dfc3ca378f649564adb134f55b15ed8043","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-18 15:59:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee4c1566cf5cdc1499b18b6cf7ae1f3de25a5526","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-19 06:19:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"84f44f2520cfb3b36ba7cba215eb293d5bb0a583","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 10:31:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662690/4/check/openstack-tox-pep8/2359787/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/90/662690/4/check/openstack-tox-docs/b7ddc7d/html/ : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":4},{"id":"7d474e360637949a8892332ae3b4b126e9d4c3d9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-19 15:53:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7a845d4f4bc5ce327f4bf1c3094f968ff4e3d79f","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-06-19 15:55:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"10c9d8667362befe415894e53f18855bf501ea53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:55:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cbbf8e4a56c697437b5e4157fa249c841d65c0aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:45:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/662690/4/gate/openstack-tox-pep8/5f72243/ : SUCCESS in 6m 36s\n- openstack-tox-docs http://logs.openstack.org/90/662690/4/gate/openstack-tox-docs/e301e65/html/ : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":4},{"id":"c432e75944154b614b85bb4665c166072703c9ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:45:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0a9d7cbec59183f772df9371644ed8639bc55b3b","revisions":{"916bc3695fce3b5414b16263d6456d0095a37a4d":{"kind":"REWORK","_number":1,"created":"2019-06-03 07:19:51.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/90/662690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/90/662690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/90/662690/1"}}},"commit":{"parents":[{"commit":"3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd","subject":"Merge \"Adds caching of credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-05-31 13:46:31.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-05-31 13:46:31.000000000","tz":330},"subject":"Fix contributor doc of keystone","message":"Fix contributor doc of keystone\n\nSince keystone now tracks features as Bugs not\nblueprints. This patch updates the same in the\ncontributor doc.\n\nChange-Id: Ic51205c1403f2085d7f921d3ff9d667cfe3f4c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/916bc3695fce3b5414b16263d6456d0095a37a4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/916bc3695fce3b5414b16263d6456d0095a37a4d"}]},"branch":"refs/heads/master"},"392df2682185be6d292c67c79ff4c5b08f0107d3":{"kind":"REWORK","_number":2,"created":"2019-06-18 06:18:00.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/90/662690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/90/662690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/90/662690/2"}}},"commit":{"parents":[{"commit":"3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd","subject":"Merge \"Adds caching of credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-05-31 13:46:31.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-06-18 06:17:34.000000000","tz":330},"subject":"Fix contributor doc of keystone","message":"Fix contributor doc of keystone\n\nSince keystone now tracks features as Bugs not\nblueprints. This patch updates the same in the\ncontributor doc.\n\nChange-Id: Ic51205c1403f2085d7f921d3ff9d667cfe3f4c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/392df2682185be6d292c67c79ff4c5b08f0107d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/392df2682185be6d292c67c79ff4c5b08f0107d3"}]},"branch":"refs/heads/master"},"ead832dca9047f63e5152772193fd3cd19eca242":{"kind":"REWORK","_number":3,"created":"2019-06-18 09:03:26.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/90/662690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/90/662690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/90/662690/3"}}},"commit":{"parents":[{"commit":"3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd","subject":"Merge \"Adds caching of credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-05-31 13:46:31.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-06-18 09:02:51.000000000","tz":330},"subject":"Fix contributor doc of keystone","message":"Fix contributor doc of keystone\n\nSince keystone now tracks features as Bugs not\nblueprints. This patch updates the same in the\ncontributor doc.\n\nChange-Id: Ic51205c1403f2085d7f921d3ff9d667cfe3f4c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ead832dca9047f63e5152772193fd3cd19eca242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ead832dca9047f63e5152772193fd3cd19eca242"}]},"branch":"refs/heads/master"},"0a9d7cbec59183f772df9371644ed8639bc55b3b":{"kind":"REWORK","_number":4,"created":"2019-06-19 06:19:44.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/90/662690/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/90/662690/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/662690/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/90/662690/4"}}},"commit":{"parents":[{"commit":"3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd","subject":"Merge \"Adds caching of credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d2b293d7edfb0bd4bdec9b33abc63d1308e10bd"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-05-31 13:46:31.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-06-19 06:13:40.000000000","tz":330},"subject":"Fix contributor doc of keystone","message":"Fix contributor doc of keystone\n\nSince keystone now tracks features as Bugs not\nblueprints. This patch updates the same in the\ncontributor doc.\n\nChange-Id: Ic51205c1403f2085d7f921d3ff9d667cfe3f4c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0a9d7cbec59183f772df9371644ed8639bc55b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0a9d7cbec59183f772df9371644ed8639bc55b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
