)]}'
{"id":"openstack%2Fkeystone~665617","triplet_id":"openstack%2Fkeystone~master~Ib1073acf5677a60714d0a386de3bcd14ce6cd134","project":"openstack/keystone","branch":"master","topic":"bug/1832265","hashtags":[],"change_id":"Ib1073acf5677a60714d0a386de3bcd14ce6cd134","subject":"token: consistently decode binary types","status":"MERGED","created":"2019-06-17 09:13:52.000000000","updated":"2019-09-19 03:23:21.000000000","submitted":"2019-09-19 03:21:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665617-1568863282809-d349e51d","meta_rev_id":"d7f8039cbab732e59c8595645fa9af85523e7942","_number":665617,"virtual_id_number":665617,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2019-08-28 19:13:02.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2019-08-05 13:41:10.000000000","_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":2,"date":"2019-09-19 03:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-08 06:01:54.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-07-05 19:19:33.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":2,"date":"2019-09-18 22:56:52.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2019-09-18 22:56:52.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-22 13:36:23.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"state":"REVIEWER"},{"updated":"2019-07-05 19:19:33.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-08 06:01:54.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-08-28 19:13:02.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-18 22:56:52.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-09-19 03:21:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ccd17c1ff6c211c8d8577d1fd6cc9ec0443fe84c","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-06-17 09:13:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6a7afeee678cedd1b1f7d2c3888b6f7831f1539","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:06:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/665617/1/check/openstack-tox-cover/e2cd236/cover/ : SUCCESS in 28m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665617/1/check/openstack-tox-lower-constraints/3c5c8ad/ : SUCCESS in 30m 11s\n- openstack-tox-pep8 http://logs.openstack.org/17/665617/1/check/openstack-tox-pep8/fcd419a/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/17/665617/1/check/openstack-tox-py27/68d7e88/ : SUCCESS in 21m 47s\n- openstack-tox-py36 http://logs.openstack.org/17/665617/1/check/openstack-tox-py36/6d81dda/ : SUCCESS in 28m 55s\n- openstack-tox-py37 http://logs.openstack.org/17/665617/1/check/openstack-tox-py37/6c0857c/ : SUCCESS in 24m 37s\n- openstack-tox-docs http://logs.openstack.org/17/665617/1/check/openstack-tox-docs/3036516/html/ : SUCCESS in 7m 04s\n- tempest-full http://logs.openstack.org/17/665617/1/check/tempest-full/8dce673/ : SUCCESS in 1h 47m 02s\n- neutron-grenade http://logs.openstack.org/17/665617/1/check/neutron-grenade/d2c4267/ : SUCCESS in 1h 02m 16s\n- grenade-py3 http://logs.openstack.org/17/665617/1/check/grenade-py3/e7d279b/ : SUCCESS in 57m 37s\n- tempest-full-py3 http://logs.openstack.org/17/665617/1/check/tempest-full-py3/7a45d65/ : SUCCESS in 1h 32m 59s\n- keystone-dsvm-functional http://logs.openstack.org/17/665617/1/check/keystone-dsvm-functional/2ba5d7b/ : SUCCESS in 33m 26s\n- keystone-dsvm-py3-functional http://logs.openstack.org/17/665617/1/check/keystone-dsvm-py3-functional/d3e847f/ : SUCCESS in 39m 19s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/17/665617/1/check/keystone-dsvm-functional-federation-opensuse15/9ebbee3/ : SUCCESS in 35m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/17/665617/1/check/keystone-dsvm-py3-functional-federation-opensuse15/0d5b41d/ : SUCCESS in 50m 01s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/17/665617/1/check/keystoneclient-devstack-functional/b42ffc8/ : SUCCESS in 17m 35s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/665617/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/72a40cc/ : SUCCESS in 1h 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f40ae8fa4226c271b7a99c073832be800092b5e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-18 18:27:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nPlease include a release note as well.","accounts_in_message":[],"_revision_number":1},{"id":"734fa738b3b04d0c922f02623e33a59219d8f0cf","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-06-19 08:55:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa1d4a113855719a1a88839234e94e79742097f6","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-06-19 08:57:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0afbcd9e6e81f699111fbb9d2ea965329e820000","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 17:15:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/17/665617/2/check/openstack-tox-cover/3285a6f/cover/ : SUCCESS in 34m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665617/2/check/openstack-tox-lower-constraints/4e0b266/ : SUCCESS in 28m 41s\n- openstack-tox-pep8 http://logs.openstack.org/17/665617/2/check/openstack-tox-pep8/c7c7ca3/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/17/665617/2/check/openstack-tox-py27/b95a66e/ : SUCCESS in 23m 07s\n- openstack-tox-py36 http://logs.openstack.org/17/665617/2/check/openstack-tox-py36/937ba0a/ : SUCCESS in 25m 02s\n- openstack-tox-py37 http://logs.openstack.org/17/665617/2/check/openstack-tox-py37/6140c33/ : SUCCESS in 29m 58s\n- openstack-tox-docs http://logs.openstack.org/17/665617/2/check/openstack-tox-docs/fd0515e/html/ : SUCCESS in 7m 16s\n- tempest-full http://logs.openstack.org/17/665617/2/check/tempest-full/a9a1837/ : TIMED_OUT in 2h 07m 03s\n- neutron-grenade http://logs.openstack.org/17/665617/2/check/neutron-grenade/4ded079/ : SUCCESS in 59m 17s\n- grenade-py3 http://logs.openstack.org/17/665617/2/check/grenade-py3/66fadb7/ : SUCCESS in 55m 39s\n- tempest-full-py3 http://logs.openstack.org/17/665617/2/check/tempest-full-py3/1ce91fd/ : SUCCESS in 1h 41m 32s\n- build-openstack-releasenotes http://logs.openstack.org/17/665617/2/check/build-openstack-releasenotes/1eff19c/html/ : SUCCESS in 11m 21s\n- keystone-dsvm-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional/69a02f3/ : SUCCESS in 33m 07s\n- keystone-dsvm-py3-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional/05c71ac/ : SUCCESS in 34m 28s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional-federation-opensuse15/29e9e5b/ : SUCCESS in 40m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional-federation-opensuse15/875104d/ : SUCCESS in 41m 23s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/17/665617/2/check/keystoneclient-devstack-functional/bb4c1a4/ : SUCCESS in 13m 43s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/665617/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/51bc74b/ : SUCCESS in 1h 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5da2340b40ba889317292a81436b4c9f2c504c04","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-06-20 04:11:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"13a81e325393df30e6e742fc39e59f1fa4a1c2ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 06:19:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/17/665617/2/check/openstack-tox-cover/f1c858f/ : TIMED_OUT in 51m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665617/2/check/openstack-tox-lower-constraints/bb9dcd4/ : SUCCESS in 33m 55s\n- openstack-tox-pep8 http://logs.openstack.org/17/665617/2/check/openstack-tox-pep8/a5d0c4c/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/17/665617/2/check/openstack-tox-py27/c6eda6d/ : SUCCESS in 24m 48s\n- openstack-tox-py36 http://logs.openstack.org/17/665617/2/check/openstack-tox-py36/5d56002/ : SUCCESS in 24m 55s\n- openstack-tox-py37 http://logs.openstack.org/17/665617/2/check/openstack-tox-py37/7450171/ : SUCCESS in 25m 32s\n- openstack-tox-docs http://logs.openstack.org/17/665617/2/check/openstack-tox-docs/fd370a3/html/ : SUCCESS in 7m 56s\n- tempest-full http://logs.openstack.org/17/665617/2/check/tempest-full/048b09b/ : SUCCESS in 1h 52m 29s\n- neutron-grenade http://logs.openstack.org/17/665617/2/check/neutron-grenade/59cac20/ : SUCCESS in 57m 40s\n- grenade-py3 http://logs.openstack.org/17/665617/2/check/grenade-py3/f587462/ : SUCCESS in 57m 08s\n- tempest-full-py3 http://logs.openstack.org/17/665617/2/check/tempest-full-py3/43fdcc5/ : SUCCESS in 1h 40m 34s\n- build-openstack-releasenotes http://logs.openstack.org/17/665617/2/check/build-openstack-releasenotes/53f941f/html/ : SUCCESS in 16m 14s\n- keystone-dsvm-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional/cfb5a66/ : SUCCESS in 31m 52s\n- keystone-dsvm-py3-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional/a4aeaf6/ : SUCCESS in 33m 28s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional-federation-opensuse15/99c3737/ : SUCCESS in 36m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional-federation-opensuse15/af160cf/ : SUCCESS in 34m 52s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/17/665617/2/check/keystoneclient-devstack-functional/85872fd/ : SUCCESS in 15m 36s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/665617/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/9c589e8/ : SUCCESS in 1h 53m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d3ccfb0f4ee1763ba74a610b87a9a93ea076839","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-06-20 06:59:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d912a399bfbd30abb6149140461e02f802019086","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 09:15:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/17/665617/2/check/openstack-tox-cover/c1a370e/cover/ : SUCCESS in 32m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665617/2/check/openstack-tox-lower-constraints/2bbb3b1/ : TIMED_OUT in 40m 53s\n- openstack-tox-pep8 http://logs.openstack.org/17/665617/2/check/openstack-tox-pep8/46cafa2/ : SUCCESS in 13m 23s\n- openstack-tox-py27 http://logs.openstack.org/17/665617/2/check/openstack-tox-py27/fa7fbd9/ : SUCCESS in 23m 51s\n- openstack-tox-py36 http://logs.openstack.org/17/665617/2/check/openstack-tox-py36/b3d3c4a/ : SUCCESS in 24m 57s\n- openstack-tox-py37 http://logs.openstack.org/17/665617/2/check/openstack-tox-py37/53086b1/ : SUCCESS in 33m 46s\n- openstack-tox-docs http://logs.openstack.org/17/665617/2/check/openstack-tox-docs/1420509/html/ : SUCCESS in 7m 30s\n- tempest-full http://logs.openstack.org/17/665617/2/check/tempest-full/2cc4bd1/ : SUCCESS in 1h 50m 10s\n- neutron-grenade http://logs.openstack.org/17/665617/2/check/neutron-grenade/05705bb/ : FAILURE in 18m 41s\n- grenade-py3 http://logs.openstack.org/17/665617/2/check/grenade-py3/9bff015/ : SUCCESS in 1h 01m 38s\n- tempest-full-py3 http://logs.openstack.org/17/665617/2/check/tempest-full-py3/640728d/ : SUCCESS in 1h 51m 25s\n- build-openstack-releasenotes http://logs.openstack.org/17/665617/2/check/build-openstack-releasenotes/c6ff0d7/html/ : SUCCESS in 10m 54s\n- keystone-dsvm-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional/278f3c9/ : SUCCESS in 34m 58s\n- keystone-dsvm-py3-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional/b4fdb41/ : TIMED_OUT in 1h 13m 03s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional-federation-opensuse15/260cb63/ : SUCCESS in 37m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional-federation-opensuse15/1e366dd/ : SUCCESS in 48m 22s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/17/665617/2/check/keystoneclient-devstack-functional/edff7a4/ : SUCCESS in 53m 17s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/665617/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/3ae34ff/ : FAILURE in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e764f713eb2dbdeafdbef7e624f49f73375a98f0","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-06-20 13:53:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e4b81e0baf3117b647ea7a0b8f31095687fd1ab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 20:17:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/665617/2/check/openstack-tox-cover/799dbdb/cover/ : SUCCESS in 31m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665617/2/check/openstack-tox-lower-constraints/161589d/ : SUCCESS in 38m 04s\n- openstack-tox-pep8 http://logs.openstack.org/17/665617/2/check/openstack-tox-pep8/60cf87a/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/17/665617/2/check/openstack-tox-py27/f69f8af/ : SUCCESS in 32m 48s\n- openstack-tox-py36 http://logs.openstack.org/17/665617/2/check/openstack-tox-py36/c54a68e/ : SUCCESS in 33m 26s\n- openstack-tox-py37 http://logs.openstack.org/17/665617/2/check/openstack-tox-py37/7d009f5/ : SUCCESS in 22m 34s\n- openstack-tox-docs http://logs.openstack.org/17/665617/2/check/openstack-tox-docs/c8f91fb/html/ : SUCCESS in 7m 45s\n- tempest-full http://logs.openstack.org/17/665617/2/check/tempest-full/5567c4b/ : SUCCESS in 1h 45m 36s\n- neutron-grenade http://logs.openstack.org/17/665617/2/check/neutron-grenade/21a0046/ : SUCCESS in 1h 06m 15s\n- grenade-py3 http://logs.openstack.org/17/665617/2/check/grenade-py3/41222cb/ : SUCCESS in 1h 04m 51s\n- tempest-full-py3 http://logs.openstack.org/17/665617/2/check/tempest-full-py3/77831a4/ : SUCCESS in 1h 31m 25s\n- build-openstack-releasenotes http://logs.openstack.org/17/665617/2/check/build-openstack-releasenotes/9486f94/html/ : SUCCESS in 10m 31s\n- keystone-dsvm-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional/a25d682/ : SUCCESS in 30m 48s\n- keystone-dsvm-py3-functional http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional/f6cc7f5/ : SUCCESS in 39m 52s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-functional-federation-opensuse15/18546d9/ : SUCCESS in 41m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/17/665617/2/check/keystone-dsvm-py3-functional-federation-opensuse15/3bcd212/ : SUCCESS in 43m 28s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/17/665617/2/check/keystoneclient-devstack-functional/564a4ec/ : SUCCESS in 18m 47s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/665617/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/5e03038/ : FAILURE in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af8169460906f8870450c7c5158385b39d2fd7ff","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-20 20:37:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a457af85f62c363b482b69a208732bcfb63ffcce","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2019-06-22 13:36:23.000000000","message":"Patch Set 2: Code-Review+1\n\nI have done manual functional testing of the patch (Rocky) with federated identity, which was also affected by the bug, with positive results, see:\nhttps://bugs.launchpad.net/keystone/+bug/1832265/comments/30","accounts_in_message":[],"_revision_number":2},{"id":"735a56a75b1c82ed97f4eb0a1fbfe966ab7ac688","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 18:21:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLooks good, just one nit on the release note name.","accounts_in_message":[],"_revision_number":2},{"id":"e859c1143434b3e3b49bb03362da8c8d48a94553","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-07-04 06:56:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"97ad8f3aab62ff6a19068d3bfdcfc0dff6bf5364","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-07-04 06:57:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b924b1b5e9fe9677288857b36f0ceca52d32b165","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:14:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/665617/3/check/openstack-tox-cover/bc21ebb/cover/ : SUCCESS in 33m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665617/3/check/openstack-tox-lower-constraints/7c76c86/ : SUCCESS in 31m 50s\n- openstack-tox-pep8 http://logs.openstack.org/17/665617/3/check/openstack-tox-pep8/7d99bde/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/17/665617/3/check/openstack-tox-py27/800d7f6/ : SUCCESS in 24m 05s\n- openstack-tox-py36 http://logs.openstack.org/17/665617/3/check/openstack-tox-py36/c44552a/ : SUCCESS in 27m 27s\n- openstack-tox-py37 http://logs.openstack.org/17/665617/3/check/openstack-tox-py37/ba7b653/ : SUCCESS in 29m 04s\n- openstack-tox-docs http://logs.openstack.org/17/665617/3/check/openstack-tox-docs/e110cd5/html/ : SUCCESS in 10m 16s\n- tempest-full http://logs.openstack.org/17/665617/3/check/tempest-full/9ee028d/ : SUCCESS in 1h 50m 44s\n- neutron-grenade http://logs.openstack.org/17/665617/3/check/neutron-grenade/32101f6/ : SUCCESS in 1h 03m 17s\n- grenade-py3 http://logs.openstack.org/17/665617/3/check/grenade-py3/2323cac/ : SUCCESS in 1h 05m 01s\n- tempest-full-py3 http://logs.openstack.org/17/665617/3/check/tempest-full-py3/3c1e925/ : SUCCESS in 1h 41m 39s\n- build-openstack-releasenotes http://logs.openstack.org/17/665617/3/check/build-openstack-releasenotes/0d777ff/html/ : SUCCESS in 11m 01s\n- keystone-dsvm-functional http://logs.openstack.org/17/665617/3/check/keystone-dsvm-functional/69c01eb/ : SUCCESS in 33m 27s\n- keystone-dsvm-py3-functional http://logs.openstack.org/17/665617/3/check/keystone-dsvm-py3-functional/962e29f/ : SUCCESS in 45m 05s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/17/665617/3/check/keystone-dsvm-functional-federation-opensuse15/1202692/ : SUCCESS in 37m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/17/665617/3/check/keystone-dsvm-py3-functional-federation-opensuse15/759caee/ : SUCCESS in 51m 39s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/17/665617/3/check/keystoneclient-devstack-functional/84ff5f9/ : SUCCESS in 16m 28s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/665617/3/check/legacy-tempest-dsvm-ldap-domain-specific-driver/982071b/ : SUCCESS in 1h 57m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5ef1e76620aa30cb43c69a6a6a1b6c813aae8546","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-05 19:19:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18759c930f4cbd7317d2c57fb08c0aa3bf36469a","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-08-05 13:41:10.000000000","message":"Patch Set 3:\n\n@lance - any chance of a second review on this fix?","accounts_in_message":[],"_revision_number":3},{"id":"e564fc8cae48bf4ab60630a5545d8be7d5c7f983","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-08-08 06:01:54.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"42d581f052781b68ba1035b6089c4ae781114468","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-08-28 19:03:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"010e3a58e1c8ba1ef7856ab3df387a411451af97","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-08-28 19:13:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17985e3746b2d2a8ab4f076b781f2dc747b429ff","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-18 22:56:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"085b24d930cf785a33d6e06b924ea7512409957e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 22:57:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5e5bbd108ac3985ba1b6c829104f5eb298bd9b94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:21:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67331ba5d4554ac09da59f11103c6509 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eae8c3997df47c98ccfe766c6a30f70 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d26411f22db94ea69592befaee61ae83 : SUCCESS in 14m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7adf331272db44e7abc88e7324eee57a : SUCCESS in 13m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db8d1d9ade90411c9a298f0ed7a4fa88 : SUCCESS in 1h 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed153cb15fdd4e1cbb2fe6fbba695334 : SUCCESS in 18m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d4493037dc74e91adedfea6392469b5 : SUCCESS in 1h 41m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27ef7a1dc1cc4b7d9f007d707d257630 : SUCCESS in 1h 02m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1c4ea737b29843d5a57f3840b0adcf22 : SUCCESS in 1h 01m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e0fc8bfd70b47cdae32bc9284442462 : SUCCESS in 1h 27m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8c3f33d431c4da3a383b75d685050d2 : SUCCESS in 15m 17s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b78090f8a2c4a5ca8f3775cfc53418b : SUCCESS in 36m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c8212f87c1124f11967a1e4b67ce5f40 : SUCCESS in 42m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b8f4d36ad454ff982a8b019f90b89b2 : SUCCESS in 1h 05m 19s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/5228c2a98a2b47d8aee03cf7ea62d11d : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":3},{"id":"91ab28564bb316df0ba8a6d09c96614e04e46c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:21:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7f0a42d321ec696917476bdacf951c673a750510","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:23:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c17cfae970460487841b4e4b3dfb0c : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25cb76f1bad14574971a44ec6a57b254 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ffa0918f5a92fd18c86703916d768012b0bea61b","revisions":{"922ec35bb19981167db3d7ebc22deb10755c1baf":{"kind":"REWORK","_number":1,"created":"2019-06-17 09:13:52.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/17/665617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/665617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/665617/1"}}},"commit":{"parents":[{"commit":"f35e71fd84e21ebc9c27df21a4976783edc3d7be","subject":"Merge \"Add flake8 ignore list to fast8 script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f35e71fd84e21ebc9c27df21a4976783edc3d7be"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2019-06-17 08:56:11.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2019-06-17 09:12:08.000000000","tz":60},"subject":"token: consistently decode binary types","message":"token: consistently decode binary types\n\nEnsure that any binary types unpacked from message payloads\nare correctly converted from binary to text type.\n\nUnder Python 3 msgpack returns the serialized input as a\nbyte string. Similar to other msgpack\u0027d values in the payload,\nwe need to explicitly decode it to a string value.\n\nThis is specifically more of an issue under Python 3; however\nthe decode operation is safe back to Python 2 so there is no\nneed to limit the decode codepath to just Python 3.\n\nChange-Id: Ib1073acf5677a60714d0a386de3bcd14ce6cd134\nCloses-Bug: 1832265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/922ec35bb19981167db3d7ebc22deb10755c1baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/922ec35bb19981167db3d7ebc22deb10755c1baf"}]},"branch":"refs/heads/master"},"d8262030840aa2da0a0d2bc5aef13091f85fed91":{"kind":"REWORK","_number":2,"created":"2019-06-19 08:55:59.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/17/665617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/665617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/665617/2"}}},"commit":{"parents":[{"commit":"f35e71fd84e21ebc9c27df21a4976783edc3d7be","subject":"Merge \"Add flake8 ignore list to fast8 script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f35e71fd84e21ebc9c27df21a4976783edc3d7be"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2019-06-17 08:56:11.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2019-06-19 08:54:02.000000000","tz":60},"subject":"token: consistently decode binary types","message":"token: consistently decode binary types\n\nEnsure that any binary types unpacked from message payloads\nare correctly converted from binary to text type.\n\nUnder Python 3 msgpack returns the serialized input as a\nbyte string. Similar to other msgpack\u0027d values in the payload,\nwe need to explicitly decode it to a string value.\n\nThis is specifically more of an issue under Python 3; however\nthe decode operation is safe back to Python 2 so there is no\nneed to limit the decode codepath to just Python 3.\n\nChange-Id: Ib1073acf5677a60714d0a386de3bcd14ce6cd134\nCloses-Bug: 1832265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d8262030840aa2da0a0d2bc5aef13091f85fed91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d8262030840aa2da0a0d2bc5aef13091f85fed91"}]},"branch":"refs/heads/master"},"ffa0918f5a92fd18c86703916d768012b0bea61b":{"kind":"REWORK","_number":3,"created":"2019-07-04 06:56:47.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/17/665617/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/665617/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/665617/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/665617/3"}}},"commit":{"parents":[{"commit":"f35e71fd84e21ebc9c27df21a4976783edc3d7be","subject":"Merge \"Add flake8 ignore list to fast8 script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f35e71fd84e21ebc9c27df21a4976783edc3d7be"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2019-06-17 08:56:11.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2019-07-04 06:55:43.000000000","tz":60},"subject":"token: consistently decode binary types","message":"token: consistently decode binary types\n\nEnsure that any binary types unpacked from message payloads\nare correctly converted from binary to text type.\n\nUnder Python 3 msgpack returns the serialized input as a\nbyte string. Similar to other msgpack\u0027d values in the payload,\nwe need to explicitly decode it to a string value.\n\nThis is specifically more of an issue under Python 3; however\nthe decode operation is safe back to Python 2 so there is no\nneed to limit the decode codepath to just Python 3.\n\nChange-Id: Ib1073acf5677a60714d0a386de3bcd14ce6cd134\nCloses-Bug: 1832265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ffa0918f5a92fd18c86703916d768012b0bea61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ffa0918f5a92fd18c86703916d768012b0bea61b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
