)]}'
{"id":"openstack%2Fkeystone~666600","triplet_id":"openstack%2Fkeystone~master~I539769bc4af0bfa04112cbfd0fccd6b78b4e7c87","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"I539769bc4af0bfa04112cbfd0fccd6b78b4e7c87","subject":"Update misleading comment about fernet credential encryption","status":"MERGED","created":"2019-06-20 14:53:42.000000000","updated":"2019-06-21 09:59:55.000000000","submitted":"2019-06-21 09:59:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666600-1561111195156-420ce86b","meta_rev_id":"e321ae2cc38a0f56d3063103d986636c75e5c5f1","_number":666600,"virtual_id_number":666600,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-06-21 09:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-06-20 15:45:57.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-06-20 21:14:49.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-06-21 03:57:16.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 15:45:57.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-06-21 03:57:16.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-06-21 09:59:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"339152d2c2007df2f0beff1f8d57fd0842177a50","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-20 14:53:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4d839421d60908bc8ebd153edf3463a4bfd0d5c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-20 15:02:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b35b297a17d118ae25d4a20170107cbd56267efd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-20 15:13:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"891779d5755013776ce152aa65b430882184ce00","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-20 15:22:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"031e39bc9c8f0bed50cbf8b1f45ab8f89b7a042d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-20 15:45:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25583de89e785c42fda34ec53db615d61785a42b","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-06-20 21:14:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56afe04b9c0c17fd44a5cacd1e979fba00014812","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:07:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/00/666600/2/check/openstack-tox-cover/86e43ce/cover/ : SUCCESS in 34m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/666600/2/check/openstack-tox-lower-constraints/870498c/ : SUCCESS in 36m 29s\n- openstack-tox-pep8 http://logs.openstack.org/00/666600/2/check/openstack-tox-pep8/03507e7/ : SUCCESS in 13m 45s\n- openstack-tox-py27 http://logs.openstack.org/00/666600/2/check/openstack-tox-py27/dc8d138/ : SUCCESS in 27m 28s\n- openstack-tox-py36 http://logs.openstack.org/00/666600/2/check/openstack-tox-py36/1b133ac/ : SUCCESS in 25m 05s\n- openstack-tox-py37 http://logs.openstack.org/00/666600/2/check/openstack-tox-py37/1ddf453/ : SUCCESS in 28m 23s\n- openstack-tox-docs http://logs.openstack.org/00/666600/2/check/openstack-tox-docs/9775216/html/ : SUCCESS in 8m 27s\n- tempest-full http://logs.openstack.org/00/666600/2/check/tempest-full/67203ba/ : SUCCESS in 1h 50m 43s\n- neutron-grenade http://logs.openstack.org/00/666600/2/check/neutron-grenade/3173a2a/ : SUCCESS in 55m 45s\n- grenade-py3 http://logs.openstack.org/00/666600/2/check/grenade-py3/a8e043c/ : SUCCESS in 59m 54s\n- tempest-full-py3 http://logs.openstack.org/00/666600/2/check/tempest-full-py3/a34b9cf/ : TIMED_OUT in 2h 04m 46s\n- keystone-dsvm-functional http://logs.openstack.org/00/666600/2/check/keystone-dsvm-functional/d5e971c/ : TIMED_OUT in 1h 13m 33s\n- keystone-dsvm-py3-functional http://logs.openstack.org/00/666600/2/check/keystone-dsvm-py3-functional/4dfe0a2/ : SUCCESS in 33m 53s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/00/666600/2/check/keystone-dsvm-functional-federation-opensuse15/7653b2d/ : SUCCESS in 36m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/00/666600/2/check/keystone-dsvm-py3-functional-federation-opensuse15/1cd9ed6/ : SUCCESS in 40m 27s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/00/666600/2/check/keystoneclient-devstack-functional/f58d07f/ : SUCCESS in 16m 56s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/00/666600/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/35937f6/ : SUCCESS in 1h 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ddcc47be146b66439a0e18ba422d16313063616","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-06-20 22:07:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"758050d97409b2b38ae5fc01f79427d9916598b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 03:54:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/00/666600/2/check/openstack-tox-cover/d74c891/cover/ : SUCCESS in 38m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/666600/2/check/openstack-tox-lower-constraints/734af83/ : SUCCESS in 41m 20s\n- openstack-tox-pep8 http://logs.openstack.org/00/666600/2/check/openstack-tox-pep8/ecce53b/ : SUCCESS in 7m 11s\n- openstack-tox-py27 http://logs.openstack.org/00/666600/2/check/openstack-tox-py27/2f5cc94/ : SUCCESS in 29m 51s\n- openstack-tox-py36 http://logs.openstack.org/00/666600/2/check/openstack-tox-py36/0a405fd/ : SUCCESS in 31m 51s\n- openstack-tox-py37 http://logs.openstack.org/00/666600/2/check/openstack-tox-py37/ed7b043/ : SUCCESS in 25m 52s\n- openstack-tox-docs http://logs.openstack.org/00/666600/2/check/openstack-tox-docs/c35e1cd/html/ : SUCCESS in 8m 24s\n- tempest-full http://logs.openstack.org/00/666600/2/check/tempest-full/99ad355/ : SUCCESS in 2h 02m 26s\n- neutron-grenade http://logs.openstack.org/00/666600/2/check/neutron-grenade/aedf016/ : SUCCESS in 1h 08m 32s\n- grenade-py3 http://logs.openstack.org/00/666600/2/check/grenade-py3/2703940/ : SUCCESS in 1h 04m 28s\n- tempest-full-py3 http://logs.openstack.org/00/666600/2/check/tempest-full-py3/7763821/ : SUCCESS in 1h 49m 42s\n- keystone-dsvm-functional http://logs.openstack.org/00/666600/2/check/keystone-dsvm-functional/de8ef2f/ : SUCCESS in 29m 57s\n- keystone-dsvm-py3-functional http://logs.openstack.org/00/666600/2/check/keystone-dsvm-py3-functional/76b8a89/ : SUCCESS in 33m 22s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/00/666600/2/check/keystone-dsvm-functional-federation-opensuse15/4d17619/ : SUCCESS in 35m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/00/666600/2/check/keystone-dsvm-py3-functional-federation-opensuse15/7b85030/ : SUCCESS in 50m 32s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/00/666600/2/check/keystoneclient-devstack-functional/eea55a7/ : SUCCESS in 51m 51s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/00/666600/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/07a418d/ : SUCCESS in 1h 55m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41237f5a683f4b8f92c6f2e23db455b45a20a215","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-06-21 03:57:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1dad6f2af0160f121d48b688fba83ddd433c570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 03:58:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"199b4a3f664253c61c665afc2a0a9aedf4830646","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 09:59:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/00/666600/2/gate/openstack-tox-lower-constraints/d181e83/ : SUCCESS in 35m 43s\n- openstack-tox-pep8 http://logs.openstack.org/00/666600/2/gate/openstack-tox-pep8/36c0a07/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/00/666600/2/gate/openstack-tox-py27/c0bd761/ : SUCCESS in 24m 08s\n- openstack-tox-py36 http://logs.openstack.org/00/666600/2/gate/openstack-tox-py36/b8c90d0/ : SUCCESS in 27m 27s\n- openstack-tox-py37 http://logs.openstack.org/00/666600/2/gate/openstack-tox-py37/2f20ee9/ : SUCCESS in 23m 13s\n- openstack-tox-docs http://logs.openstack.org/00/666600/2/gate/openstack-tox-docs/c85dee7/html/ : SUCCESS in 8m 22s\n- tempest-full http://logs.openstack.org/00/666600/2/gate/tempest-full/1a0c0ce/ : SUCCESS in 1h 25m 04s\n- neutron-grenade http://logs.openstack.org/00/666600/2/gate/neutron-grenade/c25fb6e/ : SUCCESS in 57m 14s\n- grenade-py3 http://logs.openstack.org/00/666600/2/gate/grenade-py3/0b9ceae/ : SUCCESS in 1h 00m 33s\n- tempest-full-py3 http://logs.openstack.org/00/666600/2/gate/tempest-full-py3/19da58e/ : SUCCESS in 1h 27m 12s\n- keystone-dsvm-functional http://logs.openstack.org/00/666600/2/gate/keystone-dsvm-functional/1cc9e99/ : SUCCESS in 36m 19s\n- keystone-dsvm-py3-functional http://logs.openstack.org/00/666600/2/gate/keystone-dsvm-py3-functional/eb21e0a/ : SUCCESS in 28m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e321ae2cc38a0f56d3063103d986636c75e5c5f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 09:59:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8817a26ee217ae9d57d53d0e47eb7bcabe007f7b","revisions":{"397bb27446a6eb6158a9e573e877f72878cbc3e0":{"kind":"REWORK","_number":1,"created":"2019-06-20 14:53:42.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/00/666600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/00/666600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/00/666600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/00/666600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/00/666600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/00/666600/1"}}},"commit":{"parents":[{"commit":"2b1a917e95b10064e1871ea215b0b8658e1d0c48","subject":"Merge \"Fix contributor doc of keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b1a917e95b10064e1871ea215b0b8658e1d0c48"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-20 14:51:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-20 14:53:40.000000000","tz":0},"subject":"Update misleading comment about fernet credential encryption","message":"Update misleading comment about fernet credential encryption\n\nIf credential keys are over-rotated, the credentials will be\nunrecoverable. Also clarify the usage of fernet for tokens and the\nusage of fernet for credential encryption.\n\nChange-Id: I539769bc4af0bfa04112cbfd0fccd6b78b4e7c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/397bb27446a6eb6158a9e573e877f72878cbc3e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/397bb27446a6eb6158a9e573e877f72878cbc3e0"}]},"branch":"refs/heads/master"},"8817a26ee217ae9d57d53d0e47eb7bcabe007f7b":{"kind":"REWORK","_number":2,"created":"2019-06-20 15:13:13.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/00/666600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/00/666600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/00/666600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/00/666600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/00/666600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/00/666600/2"}}},"commit":{"parents":[{"commit":"2b1a917e95b10064e1871ea215b0b8658e1d0c48","subject":"Merge \"Fix contributor doc of keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b1a917e95b10064e1871ea215b0b8658e1d0c48"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-20 14:51:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-20 15:13:06.000000000","tz":0},"subject":"Update misleading comment about fernet credential encryption","message":"Update misleading comment about fernet credential encryption\n\nIf credential keys are over-rotated, the credentials will be\nunrecoverable. Also clarify the usage of fernet for tokens and the\nusage of fernet for credential encryption.\n\nChange-Id: I539769bc4af0bfa04112cbfd0fccd6b78b4e7c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8817a26ee217ae9d57d53d0e47eb7bcabe007f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8817a26ee217ae9d57d53d0e47eb7bcabe007f7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
