)]}'
{"id":"openstack%2Fkeystone~666861","triplet_id":"openstack%2Fkeystone~stable%2Fstein~I1332df34dd6ee59537d35d93d5c35c2fe96561b2","project":"openstack/keystone","branch":"stable/stein","topic":"bug/1825111-stable/stein","hashtags":[],"change_id":"I1332df34dd6ee59537d35d93d5c35c2fe96561b2","subject":"[docs] remove deprecated ubuntu package from installation","status":"MERGED","created":"2019-06-21 15:09:24.000000000","updated":"2019-07-20 05:00:09.000000000","submitted":"2019-07-20 05:00:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666861-1563598809637-31ef7618","meta_rev_id":"f26538b7d65842ed29f1deff0175b132e63b44ad","_number":666861,"virtual_id_number":666861,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"date":"2019-07-09 16:33:02.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-07-20 05:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2019-06-21 21:18:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":1,"date":"2019-06-21 16:49:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-06-24 06:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-07-20 02:28:10.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2019-06-21 15:09:24.000000000","_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-07-20 02:28:10.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 15:09:24.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2019-06-21 16:49:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-06-21 21:18:16.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-06-24 06:19:32.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-07-20 02:28:10.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-07-20 05:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88e05f25975f5b6533e22ca935c071ad43b27e0d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-06-21 15:09:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fb06021dcce8a2b88ca2b74d9c5fd6015efe6704","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-21 16:49:45.000000000","message":"Patch Set 1: Code-Review+1\n\nClean cherry-pick.","accounts_in_message":[],"_revision_number":1},{"id":"43dbb94cea0130eac659841487ee0b0509a5914b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 18:58:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/666861/1/check/openstack-tox-pep8/2307d67/ : FAILURE in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/61/666861/1/check/openstack-tox-docs/a329da2/html/ : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a27dd81b955afdd3d3eb12fb004a59f7ec56ca1d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 21:17:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9656f7056e1124fbe26c34e40cbcc5376932cd57","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 21:18:16.000000000","message":"Patch Set 1: Code-Review+1\n\nClean cherry pick, already merged in master","accounts_in_message":[],"_revision_number":1},{"id":"54849ee3cc637b6ea2ca620e23e90d8ab281d884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 23:47:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/666861/1/check/openstack-tox-pep8/66ee0ad/ : FAILURE in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/61/666861/1/check/openstack-tox-docs/fd33b97/html/ : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9305281ae42c3bc0ec60bb22f7567d89930a6464","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-06-24 06:19:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"06419340e6b8db173d6f2916381f71dc053fde16","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-07-09 16:33:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cdf57209f67066c89491e1e42810b8e20c1d5ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 18:56:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/666861/1/check/openstack-tox-pep8/2ade61d/ : FAILURE in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/61/666861/1/check/openstack-tox-docs/0ac7d6f/html/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1bd7d3021e8e623122e66dde022d37b15aa7aaca","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-07-10 05:06:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92bbb8b698588fc9f473247c5fc14a482c443bd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 05:14:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/666861/1/check/openstack-tox-pep8/5e7a90b/ : FAILURE in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/61/666861/1/check/openstack-tox-docs/b0166de/html/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e1147ba1cb095d2e3500fafeac0bcfe273b9d9c1","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-10 21:31:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"972f7f1464fc19a5752d3b7437329b406a35b3d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 00:55:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/666861/1/check/openstack-tox-pep8/fc5c74c/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/61/666861/1/check/openstack-tox-docs/7484279/html/ : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3ea64cf15751280332e0ff90efbc3a4182aff1a9","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-07-20 02:28:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"030458fdcbcf32346905a64e6fb46c5abc3bfaf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 02:28:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"140f27173605376ad48e9bfe1c5db5200163fbe9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 05:00:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/666861/1/gate/openstack-tox-pep8/8e8ab92/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/61/666861/1/gate/openstack-tox-docs/f6ec2df/html/ : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"38485ce58c89cd529b56d68bdcc1c54caf1c585a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 05:00:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7f5e7a91bc9bfd8879fb2e4a4c0905b508dd3e5","revisions":{"a7f5e7a91bc9bfd8879fb2e4a4c0905b508dd3e5":{"kind":"REWORK","_number":1,"created":"2019-06-21 15:09:24.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/61/666861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/61/666861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/666861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/666861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/666861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/61/666861/1"}}},"commit":{"parents":[{"commit":"1828d0612cf2c51427773077dc25bd8b659eb549","subject":"Blacklist bandit 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1828d0612cf2c51427773077dc25bd8b659eb549"}]}],"author":{"name":"Frank Kloeker","email":"f.kloeker@telekom.de","date":"2019-05-02 20:25:09.000000000","tz":120},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-06-21 15:09:24.000000000","tz":0},"subject":"[docs] remove deprecated ubuntu package from installation","message":"[docs] remove deprecated ubuntu package from installation\n\nkeystone package depends already on apache2 and wsgi package:\n\nkeystone\n  Depends: adduser\n |Depends: apache2\n  Depends: \u003chttpd\u003e\n    tntnet\n    aolserver4-core\n    aolserver4-daemon\n    apache2\n    ebhttpd\n    lighttpd\n    micro-httpd\n    mini-httpd\n    nginx-core\n    nginx-extras\n    nginx-full\n    nginx-light\n    webfs\n    yaws\n  Depends: libapache2-mod-wsgi-py3\n  Depends: python3-keystone\n  Depends: ssl-cert\n[...]\n\nChange-Id: I1332df34dd6ee59537d35d93d5c35c2fe96561b2\nRelated-Bug: #1825111\n(cherry picked from commit 0f2b36b2ad50694efa92a39649b778546cca3b31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a7f5e7a91bc9bfd8879fb2e4a4c0905b508dd3e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a7f5e7a91bc9bfd8879fb2e4a4c0905b508dd3e5"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
