)]}'
{"id":"openstack%2Fkeystone~668861","triplet_id":"openstack%2Fkeystone~master~I283ee3eb587199339987bb6c2c5a54d10cebaabc","project":"openstack/keystone","branch":"master","topic":"log_regex_master","hashtags":[],"change_id":"I283ee3eb587199339987bb6c2c5a54d10cebaabc","subject":"Incorrect behavior of validate_password method","status":"MERGED","created":"2019-07-03 10:50:07.000000000","updated":"2019-07-05 08:12:36.000000000","submitted":"2019-07-05 08:12:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668861-1562314356773-5004bf0a","meta_rev_id":"2b227d541909425beb4e97cec1c849b99f23f0ec","_number":668861,"virtual_id_number":668861,"owner":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-07-05 08:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-07-03 16:56:57.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-07-05 05:42:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-05 06:11:16.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-05 06:11:16.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 16:56:57.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-05 05:42:08.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-07-05 06:11:16.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-07-05 08:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ad0d697d77ddff81245a0f0cb52372ed16e3969","author":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"date":"2019-07-03 10:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02208fa8e551abe27f8d9b107c0d688d00ba846f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:53:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/61/668861/1/check/openstack-tox-cover/e79d635/cover/ : SUCCESS in 34m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/668861/1/check/openstack-tox-lower-constraints/e3e70e6/ : SUCCESS in 38m 18s\n- openstack-tox-pep8 http://logs.openstack.org/61/668861/1/check/openstack-tox-pep8/48f9c3e/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/61/668861/1/check/openstack-tox-py27/416b621/ : SUCCESS in 25m 46s\n- openstack-tox-py36 http://logs.openstack.org/61/668861/1/check/openstack-tox-py36/1492b0e/ : SUCCESS in 27m 46s\n- openstack-tox-py37 http://logs.openstack.org/61/668861/1/check/openstack-tox-py37/d0c8220/ : SUCCESS in 34m 58s\n- openstack-tox-docs http://logs.openstack.org/61/668861/1/check/openstack-tox-docs/83c7d43/html/ : SUCCESS in 7m 12s\n- tempest-full http://logs.openstack.org/61/668861/1/check/tempest-full/1510e4e/ : SUCCESS in 1h 55m 26s\n- neutron-grenade http://logs.openstack.org/61/668861/1/check/neutron-grenade/a2c18f0/ : SUCCESS in 1h 01m 45s\n- grenade-py3 http://logs.openstack.org/61/668861/1/check/grenade-py3/f549fb0/ : SUCCESS in 1h 07m 58s\n- tempest-full-py3 http://logs.openstack.org/61/668861/1/check/tempest-full-py3/db94777/ : SUCCESS in 1h 42m 08s\n- keystone-dsvm-functional http://logs.openstack.org/61/668861/1/check/keystone-dsvm-functional/ef76f77/ : SUCCESS in 28m 09s\n- keystone-dsvm-py3-functional http://logs.openstack.org/61/668861/1/check/keystone-dsvm-py3-functional/e5ad0ea/ : SUCCESS in 29m 34s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/61/668861/1/check/keystone-dsvm-functional-federation-opensuse15/770804e/ : SUCCESS in 33m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/61/668861/1/check/keystone-dsvm-py3-functional-federation-opensuse15/436d963/ : SUCCESS in 34m 36s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/61/668861/1/check/keystoneclient-devstack-functional/6ddefb5/ : SUCCESS in 16m 37s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/61/668861/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/bad099e/ : SUCCESS in 1h 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"755472f45c6513e593ed5fab66d69be1bd4bc0cb","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-03 16:56:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5463bd8673affdbfdf10446e7e19302575288223","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-07-05 05:42:08.000000000","message":"Patch Set 1: Code-Review+1\n\nMakes sense. Thanks","accounts_in_message":[],"_revision_number":1},{"id":"22cee8811f02a5c230d6fc7e913d022c7d30084f","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-07-05 06:11:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6740f783728521fe58a8830d6724be958aa75f98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:11:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c96380c4edbc163e372756b315e74758232ee732","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 08:12:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/61/668861/1/gate/openstack-tox-lower-constraints/e4831ba/ : SUCCESS in 30m 52s\n- openstack-tox-pep8 http://logs.openstack.org/61/668861/1/gate/openstack-tox-pep8/0ed433b/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/61/668861/1/gate/openstack-tox-py27/3ed022c/ : SUCCESS in 27m 52s\n- openstack-tox-py36 http://logs.openstack.org/61/668861/1/gate/openstack-tox-py36/2c2d6e6/ : SUCCESS in 24m 24s\n- openstack-tox-py37 http://logs.openstack.org/61/668861/1/gate/openstack-tox-py37/6b241a7/ : SUCCESS in 29m 07s\n- openstack-tox-docs http://logs.openstack.org/61/668861/1/gate/openstack-tox-docs/768eadd/html/ : SUCCESS in 9m 15s\n- tempest-full http://logs.openstack.org/61/668861/1/gate/tempest-full/3288c7c/ : SUCCESS in 1h 54m 30s\n- neutron-grenade http://logs.openstack.org/61/668861/1/gate/neutron-grenade/2f3b41a/ : SUCCESS in 1h 05m 40s\n- grenade-py3 http://logs.openstack.org/61/668861/1/gate/grenade-py3/bd11165/ : SUCCESS in 1h 05m 23s\n- tempest-full-py3 http://logs.openstack.org/61/668861/1/gate/tempest-full-py3/e8c52e0/ : SUCCESS in 1h 40m 44s\n- keystone-dsvm-functional http://logs.openstack.org/61/668861/1/gate/keystone-dsvm-functional/1e73bcd/ : SUCCESS in 38m 34s\n- keystone-dsvm-py3-functional http://logs.openstack.org/61/668861/1/gate/keystone-dsvm-py3-functional/c5a914d/ : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a239a0ea4c38615945cea3dc103446177670816a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 08:12:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72af26168752b58b8c236b79b20d195083f6302d","revisions":{"72af26168752b58b8c236b79b20d195083f6302d":{"kind":"REWORK","_number":1,"created":"2019-07-03 10:50:07.000000000","uploader":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"ref":"refs/changes/61/668861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/61/668861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/668861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/668861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/668861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/61/668861/1"}}},"commit":{"parents":[{"commit":"163f4e4a6e0b5ae9d1060a1779745fab949a6088","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/163f4e4a6e0b5ae9d1060a1779745fab949a6088"}]}],"author":{"name":"BubaVV","email":"vmarkov@mirantis.com","date":"2019-07-01 14:00:44.000000000","tz":180},"committer":{"name":"BubaVV","email":"vmarkov@mirantis.com","date":"2019-07-03 10:41:37.000000000","tz":180},"subject":"Incorrect behavior of validate_password method","message":"Incorrect behavior of validate_password method\n\nLogging fails when security_compliance is turn on and password_regex is\ninvalid.\n\nChange-Id: I283ee3eb587199339987bb6c2c5a54d10cebaabc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72af26168752b58b8c236b79b20d195083f6302d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72af26168752b58b8c236b79b20d195083f6302d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
