)]}'
{"id":"openstack%2Fkeystone~672351","triplet_id":"openstack%2Fkeystone~stable%2Frocky~I87a3bfa94b5907ce4c6b4eb8e124ec948b390bf2","project":"openstack/keystone","branch":"stable/rocky","topic":"bug/1782922-stable-rocky","hashtags":[],"change_id":"I87a3bfa94b5907ce4c6b4eb8e124ec948b390bf2","subject":"Fixing dn_to_id function for cases were id is not in the DN","status":"MERGED","created":"2019-07-23 20:05:56.000000000","updated":"2019-08-29 18:45:16.000000000","submitted":"2019-08-29 18:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672351-1567104316193-1d71bfeb","meta_rev_id":"323122042a3c18f1213abc4fbeb6ee79b7dfe54f","_number":672351,"virtual_id_number":672351,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-08-29 18:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"all":[{"value":1,"date":"2019-08-05 12:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":2,"date":"2019-08-29 14:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2019-08-14 13:26:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":1,"date":"2019-08-20 06:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-08-29 15:41:45.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 20:13:03.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2019-08-01 17:16:23.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-08-14 13:26:12.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-08-20 06:29:25.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-08-29 14:10:02.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-08-29 15:41:45.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-08-29 18:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7708a00d3047c03bd227c1fe74cb8e78151f43f0","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-23 20:05:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"c37a7f44e97a0c6dc5f3791b02e51d8d708c3622","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-07-23 20:13:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"41bfda9044ee38f1b57bfbe769b891a3c9a31c76","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-23 20:38:27.000000000","message":"Patch Set 1: Workflow-1\n\nThere might be an issue in this patch with DN encoding. I just commented on the master review. WF-1 for now.","accounts_in_message":[],"_revision_number":1},{"id":"3cecbab800c010e4566f55736a5a5e877b93a4b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 22:40:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/672351/1/check/openstack-tox-pep8/c8538a6/ : FAILURE in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/51/672351/1/check/openstack-tox-py27/7fe19eb/ : SUCCESS in 22m 18s\n- openstack-tox-py35 http://logs.openstack.org/51/672351/1/check/openstack-tox-py35/d9f5be5/ : SUCCESS in 23m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/51/672351/1/check/build-openstack-sphinx-docs/05f8525/html/ : SUCCESS in 8m 56s\n- tempest-full http://logs.openstack.org/51/672351/1/check/tempest-full/c04f714/ : SUCCESS in 1h 46m 18s\n- neutron-grenade http://logs.openstack.org/51/672351/1/check/neutron-grenade/efa29c5/ : SUCCESS in 53m 57s\n- tempest-full-py3 http://logs.openstack.org/51/672351/1/check/tempest-full-py3/f284b03/ : SUCCESS in 1h 23m 34s\n- build-openstack-releasenotes http://logs.openstack.org/51/672351/1/check/build-openstack-releasenotes/b6ab377/html/ : SUCCESS in 10m 46s\n- keystone-dsvm-functional http://logs.openstack.org/51/672351/1/check/keystone-dsvm-functional/78c5306/ : SUCCESS in 35m 48s\n- keystone-dsvm-functional-v3-only http://logs.openstack.org/51/672351/1/check/keystone-dsvm-functional-v3-only/21d4c71/ : SUCCESS in 36m 36s (non-voting)\n- keystone-dsvm-py35-functional-v3-only http://logs.openstack.org/51/672351/1/check/keystone-dsvm-py35-functional-v3-only/236c833/ : SUCCESS in 34m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/672351/1/check/openstack-tox-lower-constraints/cf0311b/ : SUCCESS in 12m 49s\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/51/672351/1/check/legacy-keystoneclient-dsvm-functional/390f6fb/ : SUCCESS in 29m 35s (non-voting)\n- openstack-tox-cover-keystone-xenial http://logs.openstack.org/51/672351/1/check/openstack-tox-cover-keystone-xenial/4313016/cover/ : SUCCESS in 12m 51s\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/51/672351/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/51063c3/ : SUCCESS in 1h 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d748a405263b82a483dd30cd1346c4858a96f31d","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-08-01 13:22:20.000000000","message":"Patch Set 1:\n\nIssue fixed here: https://review.opendev.org/#/c/674028/ I think we\u0027re good to go with this patch now","accounts_in_message":[],"_revision_number":1},{"id":"40f3e2eaf52925862d17a67e8772ad72019c32df","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-08-01 14:42:37.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"d715be7b4626871243ba31662067f2253294d36d","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-08-01 14:43:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aaf63eead1fd143365c713293dba7c40fa72ce38","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-08-01 17:16:23.000000000","message":"Patch Set 1:\n\nThis is weird. Why\u0027s bandit suddenly failing on this branch?","accounts_in_message":[],"_revision_number":1},{"id":"91203fc84e2060fc0acfa4a09fe39adbe183ad1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 20:46:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/51/672351/1/check/openstack-tox-pep8/ccf35f8/ : FAILURE in 6m 45s\n- openstack-tox-py27 https://logs.opendev.org/51/672351/1/check/openstack-tox-py27/c081f65/ : SUCCESS in 14m 09s\n- openstack-tox-py35 https://logs.opendev.org/51/672351/1/check/openstack-tox-py35/15fb819/ : SUCCESS in 12m 32s\n- build-openstack-sphinx-docs https://logs.opendev.org/51/672351/1/check/build-openstack-sphinx-docs/ddf035c/html/ : SUCCESS in 6m 51s\n- tempest-full https://logs.opendev.org/51/672351/1/check/tempest-full/4d1aa28/ : SUCCESS in 1h 37m 51s\n- neutron-grenade https://logs.opendev.org/51/672351/1/check/neutron-grenade/29324d8/ : SUCCESS in 57m 49s\n- tempest-full-py3 https://logs.opendev.org/51/672351/1/check/tempest-full-py3/6917cec/ : SUCCESS in 1h 30m 55s\n- build-openstack-releasenotes https://logs.opendev.org/51/672351/1/check/build-openstack-releasenotes/310d022/html/ : SUCCESS in 12m 16s\n- keystone-dsvm-functional https://logs.opendev.org/51/672351/1/check/keystone-dsvm-functional/2fcf2b0/ : SUCCESS in 34m 57s\n- keystone-dsvm-functional-v3-only https://logs.opendev.org/51/672351/1/check/keystone-dsvm-functional-v3-only/330db2f/ : SUCCESS in 31m 54s (non-voting)\n- keystone-dsvm-py35-functional-v3-only https://logs.opendev.org/51/672351/1/check/keystone-dsvm-py35-functional-v3-only/a75fc4a/ : SUCCESS in 32m 01s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/51/672351/1/check/openstack-tox-lower-constraints/f495c6e/ : SUCCESS in 16m 48s\n- legacy-keystoneclient-dsvm-functional https://logs.opendev.org/51/672351/1/check/legacy-keystoneclient-dsvm-functional/250c30e/ : SUCCESS in 31m 37s (non-voting)\n- openstack-tox-cover-keystone-xenial https://logs.opendev.org/51/672351/1/check/openstack-tox-cover-keystone-xenial/065c86a/cover/ : SUCCESS in 16m 31s\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/51/672351/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/936967f/ : SUCCESS in 1h 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"053ce9f9a88664b5c1112c331635b2aeac281a43","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-08-02 14:06:30.000000000","message":"Patch Set 1:\n\nI\u0027m getting this pep8 issue running on stable/rocky (without this patch) locally. This issue doesn\u0027t seem related to this change at all.","accounts_in_message":[],"_revision_number":1},{"id":"2805fce92c22328409a30538d613def81919f32e","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-08-02 14:10:46.000000000","message":"Patch Set 1:\n\n@Guang, I believe we need to approve this revert change to unblock this pep8 job: https://review.opendev.org/#/c/667675/","accounts_in_message":[],"_revision_number":1},{"id":"2403b08839dfcd0116d46c59a4eb713d183626fd","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-08-02 14:45:08.000000000","message":"Patch Set 1:\n\nInvestigating a bit more, I saw this issue happening on stable/rocky, even reverting the latest change on it, but this is not happening on master branch. I would not able to find any related fix on master regarding this issue, but we need to do further investigation on it.","accounts_in_message":[],"_revision_number":1},{"id":"1c2adf7289601e529da0c47e7fa2496ce5e98f8e","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-08-05 12:12:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"40e214d49512a8a574370cc0d21d0df3468eb913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:52:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/51/672351/2/check/openstack-tox-pep8/1004e71/ : SUCCESS in 5m 58s\n- openstack-tox-py27 https://logs.opendev.org/51/672351/2/check/openstack-tox-py27/ad61117/ : SUCCESS in 11m 34s\n- openstack-tox-py35 https://logs.opendev.org/51/672351/2/check/openstack-tox-py35/cb42137/ : SUCCESS in 10m 31s\n- build-openstack-sphinx-docs https://logs.opendev.org/51/672351/2/check/build-openstack-sphinx-docs/5b00717/html/ : SUCCESS in 5m 58s\n- tempest-full https://logs.opendev.org/51/672351/2/check/tempest-full/e853997/ : SUCCESS in 1h 34m 39s\n- neutron-grenade https://logs.opendev.org/51/672351/2/check/neutron-grenade/9f5367f/ : SUCCESS in 59m 11s\n- tempest-full-py3 https://logs.opendev.org/51/672351/2/check/tempest-full-py3/c6b3b97/ : SUCCESS in 1h 29m 05s\n- build-openstack-releasenotes https://logs.opendev.org/51/672351/2/check/build-openstack-releasenotes/1944f03/html/ : SUCCESS in 12m 07s\n- keystone-dsvm-functional https://logs.opendev.org/51/672351/2/check/keystone-dsvm-functional/675f3be/ : SUCCESS in 35m 36s\n- keystone-dsvm-functional-v3-only https://logs.opendev.org/51/672351/2/check/keystone-dsvm-functional-v3-only/abf16c9/ : FAILURE in 27m 41s (non-voting)\n- keystone-dsvm-py35-functional-v3-only https://logs.opendev.org/51/672351/2/check/keystone-dsvm-py35-functional-v3-only/fb7f525/ : SUCCESS in 28m 44s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/51/672351/2/check/openstack-tox-lower-constraints/974b7af/ : SUCCESS in 12m 44s\n- legacy-keystoneclient-dsvm-functional https://logs.opendev.org/51/672351/2/check/legacy-keystoneclient-dsvm-functional/fdc318e/ : SUCCESS in 27m 42s (non-voting)\n- openstack-tox-cover-keystone-xenial https://logs.opendev.org/51/672351/2/check/openstack-tox-cover-keystone-xenial/0565765/cover/ : SUCCESS in 15m 16s\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/51/672351/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/4bb30bd/ : SUCCESS in 1h 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"916f3c03f53d40b92e5b21b46f446bc9f5238cb3","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-08-14 13:26:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ec62a787db7beb7ed757760a259cedc148564bc6","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-20 06:29:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"771e71a27afcc86b2d9c82e9dc9049de93568223","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-08-29 14:10:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"42cecfaa06345796c06565d5361c5c5193af3bbb","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-29 15:41:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa568d81c2e787cbef6313f0c052927f6613ee61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 15:51:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cbe86ea2174bbe323af19084f39f2c9c58c28de7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 18:45:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27e59d1710c4e2c9228e140bc1fe962 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3bdaa1f88ea4447aad04c271a060abf : SUCCESS in 16m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb84d973632f4756a290e99d7a2d234e : SUCCESS in 13m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6df995c845b5498bb8771c44edf20a95 : SUCCESS in 13m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4da52a569fa04cb3828609969dcf8bdb : SUCCESS in 1h 41m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c2944afe1d7745619ccb9a928ca1763f : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1687559444ad40aab772f23dac7c2e04 : SUCCESS in 1h 26m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ab3a1c098004721b655fc1851f843f2 : SUCCESS in 15m 10s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9b96b0b0ae6435b91190231a3c3ccc6 : SUCCESS in 34m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00bce79d670455b873dc493488c46fc : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":2},{"id":"20e7f1d9b549887e145e7d1b056c55d6b0e38b21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 18:45:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b84e9fcf1c259fe32def301f5e94a2ded845533","revisions":{"b870705ebd3b51971b93245a9415737c596bd357":{"kind":"REWORK","_number":1,"created":"2019-07-23 20:05:56.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/51/672351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/672351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/672351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/672351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/672351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/672351/1"}}},"commit":{"parents":[{"commit":"00242bd1977de4b1be62aaae2b47853a534c4d32","subject":"Blacklist bandit 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/00242bd1977de4b1be62aaae2b47853a534c4d32"}]}],"author":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-04-01 19:48:07.000000000","tz":-180},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-23 20:05:56.000000000","tz":0},"subject":"Fixing dn_to_id function for cases were id is not in the DN","message":"Fixing dn_to_id function for cases were id is not in the DN\n\nThe more common scenario to return the uid as part of the RDN in a DN,\nHowever, it\u0027s a valid case to not have the uid in the RDN, so we need to\nsearch in the LDAP based on the DN and return the uid in the entire object.\n\nAlso, we do not support multivalued attribute id on DN, so the test case\ncovering this case, it was adjusted for raise NotFound.\n\nCloses-Bug: 1782922\nChange-Id: I87a3bfa94b5907ce4c6b4eb8e124ec948b390bf2\n(cherry picked from commit a1dc21f3d34ae34bc6a5c9acebc0eb752495ae7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b870705ebd3b51971b93245a9415737c596bd357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b870705ebd3b51971b93245a9415737c596bd357"}]},"branch":"refs/heads/stable/rocky"},"7b84e9fcf1c259fe32def301f5e94a2ded845533":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-05 12:12:46.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/51/672351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/672351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/672351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/672351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/672351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/672351/2"}}},"commit":{"parents":[{"commit":"c220cc450cabb652c4b0afab265b93c7a7aa2afd","subject":"Revert \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c220cc450cabb652c4b0afab265b93c7a7aa2afd"}]}],"author":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-04-01 19:48:07.000000000","tz":-180},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-08-05 12:02:53.000000000","tz":-180},"subject":"Fixing dn_to_id function for cases were id is not in the DN","message":"Fixing dn_to_id function for cases were id is not in the DN\n\nThe more common scenario to return the uid as part of the RDN in a DN,\nHowever, it\u0027s a valid case to not have the uid in the RDN, so we need to\nsearch in the LDAP based on the DN and return the uid in the entire object.\n\nAlso, we do not support multivalued attribute id on DN, so the test case\ncovering this case, it was adjusted for raise NotFound.\n\nCloses-Bug: 1782922\nChange-Id: I87a3bfa94b5907ce4c6b4eb8e124ec948b390bf2\n(cherry picked from commit a1dc21f3d34ae34bc6a5c9acebc0eb752495ae7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7b84e9fcf1c259fe32def301f5e94a2ded845533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7b84e9fcf1c259fe32def301f5e94a2ded845533"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
