)]}'
{"id":"openstack%2Fkeystone~672629","triplet_id":"openstack%2Fkeystone~master~I5de14b5bd2d96c2f78152eda48842d388109e02b","project":"openstack/keystone","branch":"master","topic":"bug/1838037","hashtags":[],"change_id":"I5de14b5bd2d96c2f78152eda48842d388109e02b","subject":"Deprecate keystone.conf.memcache socket_timeout","status":"MERGED","created":"2019-07-25 05:56:02.000000000","updated":"2019-08-01 21:47:09.000000000","submitted":"2019-08-01 21:47:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672629-1564696029993-f6c17697","meta_rev_id":"9c90a02cfc4a539a13e7b9982304cc40f622e539","_number":672629,"virtual_id_number":672629,"owner":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-08-01 21:47:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-08-01 14:46:31.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-01 17:14:25.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 14:31:31.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-08-01 14:46:31.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-08-01 17:14:25.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-08-01 21:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"297c1df649fcfdcf7f0fd4be0bb606dab40fd290","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-07-25 05:56:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b09654ad2c844fcb5fc590eadfc06d391d1f376","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 07:10:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/29/672629/1/check/openstack-tox-cover/d66b665/ : FAILURE in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/672629/1/check/openstack-tox-lower-constraints/f56dcdf/ : FAILURE in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/29/672629/1/check/openstack-tox-pep8/254ccbd/ : FAILURE in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/29/672629/1/check/openstack-tox-py27/00b44dd/ : FAILURE in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/29/672629/1/check/openstack-tox-py36/9a83a8d/ : FAILURE in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/29/672629/1/check/openstack-tox-py37/a46ccf0/ : FAILURE in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/29/672629/1/check/openstack-tox-docs/ccb4a48/ : FAILURE in 5m 21s\n- tempest-full http://logs.openstack.org/29/672629/1/check/tempest-full/25adde3/ : FAILURE in 29m 26s\n- neutron-grenade http://logs.openstack.org/29/672629/1/check/neutron-grenade/0dd0a2f/ : FAILURE in 39m 32s\n- grenade-py3 http://logs.openstack.org/29/672629/1/check/grenade-py3/468bd2e/ : FAILURE in 33m 46s\n- tempest-full-py3 http://logs.openstack.org/29/672629/1/check/tempest-full-py3/e5071a2/ : FAILURE in 29m 59s\n- build-openstack-releasenotes http://logs.openstack.org/29/672629/1/check/build-openstack-releasenotes/39ff447/html/ : SUCCESS in 11m 48s\n- keystone-dsvm-functional http://logs.openstack.org/29/672629/1/check/keystone-dsvm-functional/25d02d0/ : FAILURE in 32m 20s\n- keystone-dsvm-py3-functional http://logs.openstack.org/29/672629/1/check/keystone-dsvm-py3-functional/0eb9038/ : FAILURE in 33m 18s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/29/672629/1/check/keystone-dsvm-functional-federation-opensuse15/b56562b/ : FAILURE in 41m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/29/672629/1/check/keystone-dsvm-py3-functional-federation-opensuse15/4ca625c/ : FAILURE in 31m 27s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/29/672629/1/check/keystoneclient-devstack-functional/9231283/ : FAILURE in 16m 26s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/29/672629/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/8974426/ : FAILURE in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69688aa54b5145ae64576fc1e1a11c0539d8ca5f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-07-25 07:20:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f575b819819f9f80be0ffc195177aa9aff9d41e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-07-25 08:58:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb62259a0bf3d8ac77f80c4e3cd2e4a143ef468c","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-07-26 11:57:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ac35b51e393be60d14e453f32f707076422e918","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-07-26 11:59:25.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"51df1717bbb0a090709139fcfbf840086013901e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 13:15:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/29/672629/2/check/openstack-tox-cover/4713526/ : FAILURE in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/672629/2/check/openstack-tox-lower-constraints/b14df94/ : FAILURE in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/29/672629/2/check/openstack-tox-pep8/8fcc99c/ : FAILURE in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/29/672629/2/check/openstack-tox-py27/aaa6dad/ : FAILURE in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/29/672629/2/check/openstack-tox-py36/7f344fc/ : FAILURE in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/29/672629/2/check/openstack-tox-py37/b0f76e1/ : FAILURE in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/29/672629/2/check/openstack-tox-docs/6fb106e/ : FAILURE in 6m 51s\n- tempest-full http://logs.openstack.org/29/672629/2/check/tempest-full/7b6d977/ : FAILURE in 24m 13s\n- neutron-grenade http://logs.openstack.org/29/672629/2/check/neutron-grenade/90463b1/ : FAILURE in 50m 53s\n- grenade-py3 http://logs.openstack.org/29/672629/2/check/grenade-py3/4acb5b2/ : FAILURE in 52m 25s\n- tempest-full-py3 http://logs.openstack.org/29/672629/2/check/tempest-full-py3/f32e5fe/ : FAILURE in 20m 16s\n- build-openstack-releasenotes http://logs.openstack.org/29/672629/2/check/build-openstack-releasenotes/e34b3a4/html/ : SUCCESS in 15m 34s\n- keystone-dsvm-functional http://logs.openstack.org/29/672629/2/check/keystone-dsvm-functional/52e6761/ : FAILURE in 19m 59s\n- keystone-dsvm-py3-functional http://logs.openstack.org/29/672629/2/check/keystone-dsvm-py3-functional/46ec7b1/ : FAILURE in 19m 15s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/29/672629/2/check/keystone-dsvm-functional-federation-opensuse15/a8d4936/ : FAILURE in 21m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/29/672629/2/check/keystone-dsvm-py3-functional-federation-opensuse15/f27b9df/ : FAILURE in 22m 21s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/29/672629/2/check/keystoneclient-devstack-functional/f2268c9/ : FAILURE in 16m 05s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/29/672629/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/333ec7d/ : FAILURE in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"743e40f8d528126959e3b9669ab80c0aaf435f3f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-07-26 14:31:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c207875072c0b8a1d4204e6d9abb9a00432ebb7","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-07-29 04:09:09.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\n \u003e (1 comment)\n\nPlease link [1].\n\n[1] https://bugs.launchpad.net/keystone/+bug/1829454","accounts_in_message":[],"_revision_number":2},{"id":"e7582cdeda96f0ba124e9f602444fc14b865563f","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-08-01 04:08:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5494a045d973b86e209c9889be55dd57b045f23a","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-08-01 04:09:03.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"0bf9c9cbe86f10eecea53d8358dfb7d2b8ab616d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 05:49:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/29/672629/3/check/openstack-tox-cover/c8d20ea/cover/ : SUCCESS in 34m 26s\n- openstack-tox-lower-constraints https://logs.opendev.org/29/672629/3/check/openstack-tox-lower-constraints/fcdb76e/ : SUCCESS in 38m 52s\n- openstack-tox-pep8 https://logs.opendev.org/29/672629/3/check/openstack-tox-pep8/c860c36/ : SUCCESS in 4m 54s\n- openstack-tox-py27 https://logs.opendev.org/29/672629/3/check/openstack-tox-py27/248de69/ : SUCCESS in 28m 35s\n- openstack-tox-py36 https://logs.opendev.org/29/672629/3/check/openstack-tox-py36/be3033e/ : SUCCESS in 31m 09s\n- openstack-tox-py37 https://logs.opendev.org/29/672629/3/check/openstack-tox-py37/9dcb7cb/ : SUCCESS in 34m 37s\n- openstack-tox-docs https://logs.opendev.org/29/672629/3/check/openstack-tox-docs/8f15001/html/ : SUCCESS in 7m 54s\n- tempest-full https://logs.opendev.org/29/672629/3/check/tempest-full/1b312f2/ : SUCCESS in 1h 38m 29s\n- neutron-grenade https://logs.opendev.org/29/672629/3/check/neutron-grenade/85617b1/ : SUCCESS in 1h 05m 40s\n- grenade-py3 https://logs.opendev.org/29/672629/3/check/grenade-py3/07693b2/ : SUCCESS in 56m 21s\n- tempest-full-py3 https://logs.opendev.org/29/672629/3/check/tempest-full-py3/2b677df/ : SUCCESS in 1h 26m 24s\n- build-openstack-releasenotes https://logs.opendev.org/29/672629/3/check/build-openstack-releasenotes/7ece1a6/html/ : SUCCESS in 10m 39s\n- keystone-dsvm-functional https://logs.opendev.org/29/672629/3/check/keystone-dsvm-functional/60b8ff8/ : SUCCESS in 33m 09s\n- keystone-dsvm-py3-functional https://logs.opendev.org/29/672629/3/check/keystone-dsvm-py3-functional/8a3247f/ : SUCCESS in 26m 43s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/29/672629/3/check/keystone-dsvm-functional-federation-opensuse15/f8d653f/ : SUCCESS in 46m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/29/672629/3/check/keystone-dsvm-py3-functional-federation-opensuse15/28361ed/ : SUCCESS in 33m 40s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/29/672629/3/check/keystoneclient-devstack-functional/530a053/ : SUCCESS in 18m 11s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/29/672629/3/check/legacy-tempest-dsvm-ldap-domain-specific-driver/e86c350/ : SUCCESS in 1h 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9169d57e6c929c85aa50b4a40d0c893e8cef402","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-01 14:46:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b900170a1c55c56e61fd9f79d90867000e8a256a","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-08-01 17:14:25.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c8d2b16d11f80f159d1f64008bd51aae416e6b43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:52:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34ad4c1c8690b3ea129188654364a5438e80aeb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:47:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a58e7f8f9b54287924c2502f165850194035b93b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:47:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/29/672629/3/gate/openstack-tox-lower-constraints/a00cacd/ : SUCCESS in 39m 02s\n- openstack-tox-pep8 https://logs.opendev.org/29/672629/3/gate/openstack-tox-pep8/b1472bc/ : SUCCESS in 6m 34s\n- openstack-tox-py27 https://logs.opendev.org/29/672629/3/gate/openstack-tox-py27/bdc107c/ : SUCCESS in 29m 56s\n- openstack-tox-py36 https://logs.opendev.org/29/672629/3/gate/openstack-tox-py36/bc4a683/ : SUCCESS in 38m 27s\n- openstack-tox-py37 https://logs.opendev.org/29/672629/3/gate/openstack-tox-py37/64547e6/ : SUCCESS in 29m 37s\n- openstack-tox-docs https://logs.opendev.org/29/672629/3/gate/openstack-tox-docs/d52cddb/html/ : SUCCESS in 9m 30s\n- tempest-full https://logs.opendev.org/29/672629/3/gate/tempest-full/d924c13/ : SUCCESS in 1h 33m 55s\n- neutron-grenade https://logs.opendev.org/29/672629/3/gate/neutron-grenade/8d2cf4f/ : SUCCESS in 58m 08s\n- grenade-py3 https://logs.opendev.org/29/672629/3/gate/grenade-py3/708d75d/ : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://logs.opendev.org/29/672629/3/gate/tempest-full-py3/769574b/ : SUCCESS in 1h 23m 17s\n- build-openstack-releasenotes https://logs.opendev.org/29/672629/3/gate/build-openstack-releasenotes/941c4bb/html/ : SUCCESS in 11m 50s\n- keystone-dsvm-functional https://logs.opendev.org/29/672629/3/gate/keystone-dsvm-functional/6f43543/ : SUCCESS in 38m 49s\n- keystone-dsvm-py3-functional https://logs.opendev.org/29/672629/3/gate/keystone-dsvm-py3-functional/1b09400/ : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4b747fa083ad09ae8ba92dadc63ac0b830c7e562","revisions":{"914f6ef2a8a509ec48d0d83361d87d44edcf4f72":{"kind":"REWORK","_number":1,"created":"2019-07-25 05:56:02.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/29/672629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/29/672629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/29/672629/1"}}},"commit":{"parents":[{"commit":"85ae3c4d096d6d63ac003848ce868af7265bc930","subject":"Merge \"Fixing dn_to_id function for cases were id is not in the DN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/85ae3c4d096d6d63ac003848ce868af7265bc930"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-07-25 05:53:21.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-07-25 05:53:21.000000000","tz":480},"subject":"Deprecate keystone.conf.memcache socket_timeout","message":"Deprecate keystone.conf.memcache socket_timeout\n\nChange-Id: I5de14b5bd2d96c2f78152eda48842d388109e02b\nCloses-Bug: #1837407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/914f6ef2a8a509ec48d0d83361d87d44edcf4f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/914f6ef2a8a509ec48d0d83361d87d44edcf4f72"}]},"branch":"refs/heads/master"},"f46593d389c4a8d1c169cffac445a0aafcc53746":{"kind":"REWORK","_number":2,"created":"2019-07-26 11:57:39.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/29/672629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/29/672629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/29/672629/2"}}},"commit":{"parents":[{"commit":"85ae3c4d096d6d63ac003848ce868af7265bc930","subject":"Merge \"Fixing dn_to_id function for cases were id is not in the DN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/85ae3c4d096d6d63ac003848ce868af7265bc930"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-07-25 05:53:21.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-07-26 11:57:26.000000000","tz":480},"subject":"Deprecate keystone.conf.memcache socket_timeout","message":"Deprecate keystone.conf.memcache socket_timeout\n\nChange-Id: I5de14b5bd2d96c2f78152eda48842d388109e02b\nCloses-Bug: #1837407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f46593d389c4a8d1c169cffac445a0aafcc53746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f46593d389c4a8d1c169cffac445a0aafcc53746"}]},"branch":"refs/heads/master"},"4b747fa083ad09ae8ba92dadc63ac0b830c7e562":{"kind":"REWORK","_number":3,"created":"2019-08-01 04:08:55.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/29/672629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/29/672629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/672629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/29/672629/3"}}},"commit":{"parents":[{"commit":"85ae3c4d096d6d63ac003848ce868af7265bc930","subject":"Merge \"Fixing dn_to_id function for cases were id is not in the DN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/85ae3c4d096d6d63ac003848ce868af7265bc930"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-07-25 05:53:21.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-08-01 04:07:38.000000000","tz":480},"subject":"Deprecate keystone.conf.memcache socket_timeout","message":"Deprecate keystone.conf.memcache socket_timeout\n\nChange-Id: I5de14b5bd2d96c2f78152eda48842d388109e02b\nPartial-Bug: #1838037\nCloses-Bug: #1837407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b747fa083ad09ae8ba92dadc63ac0b830c7e562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b747fa083ad09ae8ba92dadc63ac0b830c7e562"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
