)]}'
{"id":"openstack%2Fkeystone~675228","triplet_id":"openstack%2Fkeystone~master~I4d99f630cb16e1d58261012e59d3a92c7035734c","project":"openstack/keystone","branch":"master","topic":"immutable-resources","hashtags":[],"change_id":"I4d99f630cb16e1d58261012e59d3a92c7035734c","subject":"Add --immutable-roles flag to bootstrap command","status":"MERGED","created":"2019-08-07 23:23:24.000000000","updated":"2019-09-14 01:13:40.000000000","submitted":"2019-09-14 01:10:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675228-1568423425399-4c93791d","meta_rev_id":"1585fec295984bf1e12fb52a998ca6b462384996","_number":675228,"virtual_id_number":675228,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-13 01:39:05.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-09-14 01:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-09-10 16:18:02.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-11 19:07:24.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":1,"date":"2019-09-13 23:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-11 19:07:24.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 17:42:55.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-10 16:18:02.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-09-11 19:07:24.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-09-13 23:58:20.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 01:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e38e6cd1d6caa2454a54effbc8cc39829c82c637","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-07 23:23:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba94c74c28fcab3ffbb4d481a858672bcc84d682","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 03:12:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/675228/1/check/openstack-tox-cover/5b4a439/ : FAILURE in 37m 08s\n- openstack-tox-lower-constraints https://logs.opendev.org/28/675228/1/check/openstack-tox-lower-constraints/0417264/ : FAILURE in 37m 07s\n- openstack-tox-pep8 https://logs.opendev.org/28/675228/1/check/openstack-tox-pep8/0ad81ab/ : SUCCESS in 7m 48s\n- openstack-tox-py27 https://logs.opendev.org/28/675228/1/check/openstack-tox-py27/f8377d1/ : FAILURE in 26m 22s\n- openstack-tox-py36 https://logs.opendev.org/28/675228/1/check/openstack-tox-py36/a404f12/ : FAILURE in 36m 48s\n- openstack-tox-py37 https://logs.opendev.org/28/675228/1/check/openstack-tox-py37/0363cb8/ : FAILURE in 32m 55s\n- openstack-tox-docs https://logs.opendev.org/28/675228/1/check/openstack-tox-docs/4ab4ade/html/ : SUCCESS in 11m 30s\n- tempest-full https://logs.opendev.org/28/675228/1/check/tempest-full/c0611ce/ : FAILURE in 1h 22m 05s\n- neutron-grenade https://logs.opendev.org/28/675228/1/check/neutron-grenade/392d3cc/ : FAILURE in 1h 05m 13s\n- grenade-py3 https://logs.opendev.org/28/675228/1/check/grenade-py3/78f29a0/ : FAILURE in 1h 05m 59s\n- tempest-full-py3 https://logs.opendev.org/28/675228/1/check/tempest-full-py3/5b2ba62/ : FAILURE in 1h 07m 39s\n- keystone-dsvm-functional https://logs.opendev.org/28/675228/1/check/keystone-dsvm-functional/dfa4712/ : SUCCESS in 52m 31s\n- keystone-dsvm-py3-functional https://logs.opendev.org/28/675228/1/check/keystone-dsvm-py3-functional/5ef80e5/ : SUCCESS in 33m 33s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/28/675228/1/check/keystone-dsvm-functional-federation-opensuse15/27ce4e3/ : SUCCESS in 41m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/28/675228/1/check/keystone-dsvm-py3-functional-federation-opensuse15/a5b86c1/ : SUCCESS in 58m 30s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/28/675228/1/check/keystoneclient-devstack-functional/9405c71/ : FAILURE in 16m 09s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/28/675228/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/1d5dd7a/ : FAILURE in 1h 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33215f14e1abb079c8f28883d510d89ef3763fa8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-09 03:26:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5f4ccc58d40e3995a04609741f8c749de259cce8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 05:16:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/675228/2/check/openstack-tox-cover/0e922c5/ : FAILURE in 53m 59s\n- openstack-tox-lower-constraints https://logs.opendev.org/28/675228/2/check/openstack-tox-lower-constraints/3063896/ : TIMED_OUT in 41m 06s\n- openstack-tox-pep8 https://logs.opendev.org/28/675228/2/check/openstack-tox-pep8/9a8a0a1/ : SUCCESS in 5m 13s\n- openstack-tox-py27 https://logs.opendev.org/28/675228/2/check/openstack-tox-py27/4dcebab/ : TIMED_OUT in 40m 52s\n- openstack-tox-py36 https://logs.opendev.org/28/675228/2/check/openstack-tox-py36/9f40016/ : TIMED_OUT in 40m 53s\n- openstack-tox-py37 https://logs.opendev.org/28/675228/2/check/openstack-tox-py37/acb6452/ : TIMED_OUT in 40m 58s\n- openstack-tox-docs https://logs.opendev.org/28/675228/2/check/openstack-tox-docs/f72c638/html/ : SUCCESS in 7m 19s\n- tempest-full https://logs.opendev.org/28/675228/2/check/tempest-full/82a56e0/ : SUCCESS in 1h 41m 33s\n- neutron-grenade https://logs.opendev.org/28/675228/2/check/neutron-grenade/6825482/ : SUCCESS in 54m 57s\n- grenade-py3 https://logs.opendev.org/28/675228/2/check/grenade-py3/24f3396/ : SUCCESS in 55m 36s\n- tempest-full-py3 https://logs.opendev.org/28/675228/2/check/tempest-full-py3/f7d54c6/ : SUCCESS in 1h 40m 55s\n- keystone-dsvm-functional https://logs.opendev.org/28/675228/2/check/keystone-dsvm-functional/31f557f/ : SUCCESS in 30m 35s\n- keystone-dsvm-py3-functional https://logs.opendev.org/28/675228/2/check/keystone-dsvm-py3-functional/3d6c852/ : SUCCESS in 31m 02s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/28/675228/2/check/keystone-dsvm-functional-federation-opensuse15/fdf777f/ : SUCCESS in 49m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/28/675228/2/check/keystone-dsvm-py3-functional-federation-opensuse15/a6fa0f0/ : SUCCESS in 49m 50s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/28/675228/2/check/keystoneclient-devstack-functional/bdff34d/ : SUCCESS in 18m 44s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/28/675228/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/44e90c9/ : SUCCESS in 1h 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6981431cfe568156a0e47e739d7126177947ef3a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-09 18:02:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"16801151ba33421620fedad79337c5dc7f6e97cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:51:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/675228/3/check/openstack-tox-cover/0616e8a/ : FAILURE in 34m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/28/675228/3/check/openstack-tox-lower-constraints/d9a2c4b/ : FAILURE in 31m 13s\n- openstack-tox-pep8 https://logs.opendev.org/28/675228/3/check/openstack-tox-pep8/c8134e9/ : SUCCESS in 5m 38s\n- openstack-tox-py27 https://logs.opendev.org/28/675228/3/check/openstack-tox-py27/2c2a7d4/ : FAILURE in 31m 05s\n- openstack-tox-py36 https://logs.opendev.org/28/675228/3/check/openstack-tox-py36/a870f48/ : FAILURE in 31m 58s\n- openstack-tox-py37 https://logs.opendev.org/28/675228/3/check/openstack-tox-py37/0452ff7/ : FAILURE in 28m 46s\n- openstack-tox-docs https://logs.opendev.org/28/675228/3/check/openstack-tox-docs/573f8b2/html/ : SUCCESS in 8m 06s\n- tempest-full https://logs.opendev.org/28/675228/3/check/tempest-full/67640f4/ : SUCCESS in 1h 38m 13s\n- neutron-grenade https://logs.opendev.org/28/675228/3/check/neutron-grenade/4c42cc9/ : SUCCESS in 1h 01m 32s\n- grenade-py3 https://logs.opendev.org/28/675228/3/check/grenade-py3/11ea148/ : SUCCESS in 1h 14m 34s\n- tempest-full-py3 https://logs.opendev.org/28/675228/3/check/tempest-full-py3/34257a8/ : SUCCESS in 1h 31m 28s\n- keystone-dsvm-functional https://logs.opendev.org/28/675228/3/check/keystone-dsvm-functional/64f761a/ : SUCCESS in 31m 59s\n- keystone-dsvm-py3-functional https://logs.opendev.org/28/675228/3/check/keystone-dsvm-py3-functional/7ba4e27/ : SUCCESS in 32m 01s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/28/675228/3/check/keystone-dsvm-functional-federation-opensuse15/e2ee501/ : SUCCESS in 33m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/28/675228/3/check/keystone-dsvm-py3-functional-federation-opensuse15/77fecec/ : SUCCESS in 38m 05s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/28/675228/3/check/keystoneclient-devstack-functional/6076336/ : SUCCESS in 15m 09s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/28/675228/3/check/legacy-tempest-dsvm-ldap-domain-specific-driver/19dcebb/ : SUCCESS in 1h 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e09068496e9d38c27e71a2855fbb053f6f5319bb","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 01:49:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"da580f74cb1f523c355b3a8dcef5937417eabdb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 02:52:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/675228/4/check/openstack-tox-cover/6230507/ : FAILURE in 30m 26s\n- openstack-tox-lower-constraints https://logs.opendev.org/28/675228/4/check/openstack-tox-lower-constraints/1718a20/ : FAILURE in 30m 27s\n- openstack-tox-pep8 https://logs.opendev.org/28/675228/4/check/openstack-tox-pep8/84bb996/ : SUCCESS in 5m 35s\n- openstack-tox-py27 https://logs.opendev.org/28/675228/4/check/openstack-tox-py27/a72c10a/ : FAILURE in 26m 04s\n- openstack-tox-py36 https://logs.opendev.org/28/675228/4/check/openstack-tox-py36/d52b9b3/ : FAILURE in 25m 02s\n- openstack-tox-py37 https://logs.opendev.org/28/675228/4/check/openstack-tox-py37/2f81473/ : FAILURE in 20m 33s\n- openstack-tox-docs https://logs.opendev.org/28/675228/4/check/openstack-tox-docs/4ece96f/html/ : SUCCESS in 6m 49s\n- tempest-full https://logs.opendev.org/28/675228/4/check/tempest-full/51869c4/ : FAILURE in 21m 24s\n- neutron-grenade https://logs.opendev.org/28/675228/4/check/neutron-grenade/f997e2a/ : FAILURE in 52m 08s\n- grenade-py3 https://logs.opendev.org/28/675228/4/check/grenade-py3/cdbb318/ : FAILURE in 50m 00s\n- tempest-full-py3 https://logs.opendev.org/28/675228/4/check/tempest-full-py3/2d2528a/ : FAILURE in 22m 31s\n- keystone-dsvm-functional https://logs.opendev.org/28/675228/4/check/keystone-dsvm-functional/5015e50/ : FAILURE in 19m 48s\n- keystone-dsvm-py3-functional https://logs.opendev.org/28/675228/4/check/keystone-dsvm-py3-functional/b4ad733/ : FAILURE in 16m 49s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/28/675228/4/check/keystone-dsvm-functional-federation-opensuse15/5361d5b/ : FAILURE in 27m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/28/675228/4/check/keystone-dsvm-py3-functional-federation-opensuse15/5b084ae/ : FAILURE in 21m 22s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/28/675228/4/check/keystoneclient-devstack-functional/54679c9/ : FAILURE in 14m 19s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/28/675228/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/56d63cb/ : FAILURE in 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8328ba66285a78a2fd8a7e43abe1b310100362c0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 04:19:23.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"620417fb7d3160ce4d32c2277983e3e78677d0c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 06:11:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/675228/5/check/openstack-tox-cover/77784a2/cover/ : SUCCESS in 30m 40s\n- openstack-tox-lower-constraints https://logs.opendev.org/28/675228/5/check/openstack-tox-lower-constraints/9bb2fd1/ : SUCCESS in 33m 55s\n- openstack-tox-pep8 https://logs.opendev.org/28/675228/5/check/openstack-tox-pep8/57ad0c0/ : SUCCESS in 6m 04s\n- openstack-tox-py27 https://logs.opendev.org/28/675228/5/check/openstack-tox-py27/1186f01/ : SUCCESS in 31m 29s\n- openstack-tox-py36 https://logs.opendev.org/28/675228/5/check/openstack-tox-py36/890624e/ : TIMED_OUT in 40m 53s\n- openstack-tox-py37 https://logs.opendev.org/28/675228/5/check/openstack-tox-py37/d3c24b7/ : TIMED_OUT in 40m 48s\n- openstack-tox-docs https://logs.opendev.org/28/675228/5/check/openstack-tox-docs/7bebd6c/html/ : SUCCESS in 9m 39s\n- tempest-full https://logs.opendev.org/28/675228/5/check/tempest-full/b5ad380/ : SUCCESS in 1h 37m 40s\n- neutron-grenade https://logs.opendev.org/28/675228/5/check/neutron-grenade/157766b/ : SUCCESS in 57m 26s\n- grenade-py3 https://logs.opendev.org/28/675228/5/check/grenade-py3/bd23a95/ : SUCCESS in 53m 08s\n- tempest-full-py3 https://logs.opendev.org/28/675228/5/check/tempest-full-py3/92ae6b2/ : SUCCESS in 1h 28m 05s\n- keystone-dsvm-functional https://logs.opendev.org/28/675228/5/check/keystone-dsvm-functional/d934a8a/ : SUCCESS in 40m 08s\n- keystone-dsvm-py3-functional https://logs.opendev.org/28/675228/5/check/keystone-dsvm-py3-functional/dc63069/ : SUCCESS in 35m 54s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/28/675228/5/check/keystone-dsvm-functional-federation-opensuse15/f3289e5/ : SUCCESS in 44m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/28/675228/5/check/keystone-dsvm-py3-functional-federation-opensuse15/b805038/ : SUCCESS in 37m 56s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/28/675228/5/check/keystoneclient-devstack-functional/b9ddfdc/ : SUCCESS in 13m 44s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/28/675228/5/check/legacy-tempest-dsvm-ldap-domain-specific-driver/af2407e/ : SUCCESS in 1h 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e4e1dae71e8966cde5f9f53d56b13bccd575cc0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 14:07:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"77b0d07fb1edbcd2bb8a08202f42184b3f2a0118","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 18:10:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0634dd4ef27b4a428fb56525d4ee658f : SUCCESS in 32m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a53fa28fce6a4bcc8f94e4deaf354e34 : SUCCESS in 35m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f4a1e94386469a9c9b77e571fcc838 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a16dfd2f812453385219dfcf9e057b9 : SUCCESS in 33m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfe38fdb1b54870b151c9b65a44a69c : SUCCESS in 23m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf1ce6fe74a545ee9e1e2fa83211046e : SUCCESS in 23m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4646c844a5d443fa4b6570d85aee995 : SUCCESS in 7m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cfa08bec9cca4a2599156ce6146cc70f : SUCCESS in 1h 57m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/adc02e507458462497d0b4f4e0078382 : SUCCESS in 54m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0f54f22efe4140729772396166b508a1 : SUCCESS in 1h 12m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d97cba545f440379dd2f88b5aedeed3 : SUCCESS in 1h 29m 27s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/8821c82b4bc3462ab0a280e163bf6ce4 : SUCCESS in 38m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d9e8e055496c411a8e4b358ed3a47ebe : SUCCESS in 38m 31s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/64a0fe3bd2594695bbfbfdd2c1257d3a : SUCCESS in 38m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d8a0e5c261684dfcb17f8722b50a0713 : SUCCESS in 48m 38s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/089756feaa7c478baf5a770fd271b62a : SUCCESS in 16m 54s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/70529698c12945eabdf99deb8af27134 : SUCCESS in 1h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"59f0c69ced7bc7e2b740a512a705dba9419ada17","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-24 19:28:40.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"037fbc6c0584ff633765d8aee6df01304bd94a11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 21:00:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efba63df447547519835c3d6713a121b : SUCCESS in 39m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db8af1a29ed24d6caf7da4dd1e93072a : TIMED_OUT in 41m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57fc2adc7bf4dcaa3b20a120bf8d425 : FAILURE in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f574c0dd29c445dbb3e55111f2783d4b : SUCCESS in 35m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4e15d7faf44414b977fd9bc9aca37e3 : SUCCESS in 40m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d112a28093d4d7c93e4249d17f915b0 : TIMED_OUT in 41m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb40481765b040e0bb4e761212c2394a : FAILURE in 5m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/41749b9cf6e948519c6fdd586e89b1c6 : FAILURE in 1h 30m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af67b96305e548adb5ba28619ffaf516 : SUCCESS in 1h 01m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1d16adcbfa1b4688a2ddff7623740630 : SUCCESS in 58m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad2461958d794e6ba831e738ce6d8106 : FAILURE in 1h 09m 35s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/4fe44ae4ed6643dd9421ca431f6301ed : SUCCESS in 38m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2345f79217164eee80d36acbe1e783e1 : SUCCESS in 31m 13s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f3b1fff48a2a4dd8b74e21ed695e2bc9 : RETRY_LIMIT in 4m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/120f1c3e1c9f4a8fa2b7617171867563 : RETRY_LIMIT in 4m 20s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4cdf5884d8da42ba8f25f1b16ee753a7 : SUCCESS in 13m 23s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/962a6a325289490fb774ff5380442890 : FAILURE in 1h 20m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72710ab013e4431286cd46d0642ffbdf : SUCCESS in 1h 03m 56s","accounts_in_message":[],"_revision_number":6},{"id":"98969bbbcbe250017ec63441adc98bf871b08346","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-08-29 17:42:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3e0dffce96329ea523cd63273b105ce06a268450","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 22:19:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab26629ff82d4f23ba383ddf8f8a1c9e : POST_FAILURE in 35m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56283480865c4f008ee610742f3405a0 : SUCCESS in 38m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af03ae88ef83439fb5cd8bd98016cedd : SUCCESS in 8m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d99b13938cf4f98b0d84acaa4d1e6dd : SUCCESS in 33m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81120d6090314c4b967fbe8cf27926ec : TIMED_OUT in 41m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/935343a001934ee6a7d9080b0534b67a : SUCCESS in 31m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adfe72bba6264a2dadddf8b6f3d9b004 : SUCCESS in 16m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09ebc0f55ea6485197b10afed12ea26a : FAILURE in 1h 43m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/72de841178c249a99c1ffab067f21683 : SUCCESS in 1h 02m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1d17b5648c4c44a79057bf21383d7ea1 : SUCCESS in 59m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95553ea5ee544a0fa5acc6af5e253dbb : FAILURE in 1h 18m 36s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/e12a66e39e654ca98e8eb3bbc7e7bd3a : SUCCESS in 33m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6a7ff69e57ce4c769823e44966f5cb62 : SUCCESS in 40m 08s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fede6beb8ccc4beaa74e9b7e5a0463ed : SUCCESS in 39m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ad45fe2accf64fbaa19d22e1275f4413 : SUCCESS in 39m 45s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4db06e6fea144208b098e70f8b2de2b4 : SUCCESS in 18m 54s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9a1cff51b4a44a7cbf0f81f842eddfe0 : FAILURE in 1h 16m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102aef266b0a4a6697282f93cfabb90e : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":7},{"id":"12a7d607d91590760893de399dc5eb3a3694d752","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-09 23:36:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"839d54f7630bdcc4183c9081922f1489611156ee","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-09 23:37:22.000000000","message":"Patch Set 8:\n\nRestored patchset 6 and rebased, patchset 7 incorrectly introduced changes to immutable.py","accounts_in_message":[],"_revision_number":8},{"id":"00155a62ed6a690a2c4f5e3ad0e1a6a268df7fae","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-09 23:50:09.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"38712d43e89f09720137186d231062b033b5ed6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:19:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40fac99699a9477f9474280bd6c29c73 : FAILURE in 42m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0db3cd3087a346c5af5f408dc044081a : FAILURE in 36m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/884ea8de44d64cfbb1c7c7370ff4557c : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b4caa2c64284576b165d2ec943ed641 : FAILURE in 34m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d701d18259124ee1a111f241e7809b7d : FAILURE in 34m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/476e8d3ae49c4a02b80208c19fbfd365 : FAILURE in 38m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b340ed28a17a466a9f4ec3aeb1860d8a : SUCCESS in 9m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50e6a0ff66c440d79aa8af1ff734e392 : SUCCESS in 1h 40m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/88fe75e61a3e4d0a80b637db419f5ebb : SUCCESS in 1h 02m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2e8252b46a0b4bdc84252e653fa0e357 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b50399992f3d40ef9dc86a6c3f3b5a52 : SUCCESS in 1h 48m 35s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/e95ba934fde24812b94d8327b45332a9 : SUCCESS in 31m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/425dd83f3e60411e9ef4a9f6375366d7 : SUCCESS in 33m 45s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f0596257d60f42cf9dd7278d4d35e3a1 : SUCCESS in 37m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/8362737189a7484fad99024954c86344 : SUCCESS in 35m 37s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d5d68a822e7c4e36b0be687706d15028 : SUCCESS in 17m 33s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e05de908155043daa2d75602e555046a : SUCCESS in 1h 38m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/475ab1bf00974e4aa0f229a58fca2227 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":9},{"id":"e30eea0f2de10eeaf5cea710b0c0dca5bc579b23","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 06:31:29.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"f282d5b2a91a83bfde23247969d25199b710f0c7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 07:03:40.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"45e30da94bc7ac71b05ae67bab23d15873d47c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:42:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/037560a0a47d43f28c78d4da09c0571d : SUCCESS in 38m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4c84164ada44968ad849a26679dd4ed : TIMED_OUT in 41m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db59f60abb4547e28d857bb9ce720728 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aed64d2a9cc4452aa13bddafb7d7e3f5 : SUCCESS in 36m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdd899faf45c404bb16f4511d637d992 : SUCCESS in 36m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c112476f964b4935825e42dd88e7d6f3 : SUCCESS in 28m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96dd6816d6414c97a3f1b5833965d7e7 : SUCCESS in 10m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1be32f12286a4cbaa71c930eddb142be : SUCCESS in 1h 41m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe67af6eca884551a21f9d3252ad54b1 : SUCCESS in 1h 01m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a3815103e986481f967c65f4c963e08f : SUCCESS in 1h 01m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff52b988e95a4f85a31aedce40c7bc46 : SUCCESS in 1h 39m 38s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5b6f1a3c323432ba41bc295c3bbae54 : SUCCESS in 31m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/503e92e1acdd4bf9ac80667886f23aa0 : SUCCESS in 35m 19s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2ced95cd30f94a589ddbe9a22adba6b0 : SUCCESS in 44m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/14a07de033284ca79cf4d0c7cdd819e5 : SUCCESS in 32m 13s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ee334c294379484db0416c70f8852f37 : SUCCESS in 22m 09s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c2b3ec6781174a55bec96a72468d5b46 : SUCCESS in 1h 29m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6483eb18eec74c819811ce75ccde052b : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":11},{"id":"bfef3f85a3fb6e4efb0cf63922bed4283152bf3a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 15:10:02.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"3e4c332d2cff3f00120303796a4cf1c016ffaee7","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-10 16:18:02.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"2bfbcc6ccae86a862ca69517983c6e9a017c2459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 00:21:45.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ede94d40ba7648c78a062c3107b07423 : SUCCESS in 49m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b4087076457400a8d2132ad9457076d : TIMED_OUT in 40m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065d2309c69b45c2bb40fc4722c2619c : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e89275de4244ac7805f319d6257f1ce : SUCCESS in 45m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/971e059e9020482eb6f2415a5731c1ee : SUCCESS in 30m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14751c174c7348f0a29b78eb6f0840c8 : SUCCESS in 34m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af871670c1a4e7586ccd4b16dfc73d3 : SUCCESS in 11m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/38f943a298da4b93b6f0f4e7c3bdc4cd : TIMED_OUT in 2h 06m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/49f433db835449298af69b824000dc5b : SUCCESS in 1h 02m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/83a097ff869a462395ad3e02317e82ca : SUCCESS in 1h 03m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72ca5fd03bd04935ada863474bc13783 : SUCCESS in 1h 29m 37s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/67354067ef73429eaef953669ff4de60 : SUCCESS in 33m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e73bd30929e4458a9f943b7cf8fd845e : SUCCESS in 37m 54s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c2dd3eb2ea404564aac56cca38a4b66b : SUCCESS in 40m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/b914278da8cf4479970c2c072a813031 : SUCCESS in 39m 46s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d3f9ad339a334a249e96702a5b4b66f5 : SUCCESS in 16m 42s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bfb6a84bc3e54006b1fddd1b1c684ad6 : SUCCESS in 2h 35m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/799cea454f874d69964c2f54eea60a58 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":11},{"id":"7f5cd5f8bd3767a4df7ab9f5004c7527076325d9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 00:44:04.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"3381cccd8e7d818e8388bdadecacfa0a430225a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 02:42:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56f451dd1f7f470faecd20f0ce15865c : SUCCESS in 50m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3847e11e4294470aaae21eed2eea096 : SUCCESS in 32m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715f37ecdc384a358db356d1f932002d : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7b272ccb54d45ac8a570a69de707605 : SUCCESS in 28m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1500991fd2244a2592945af7eef24107 : SUCCESS in 26m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d74e5edc616141d0bb75ff88d45373b5 : SUCCESS in 40m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40509dddcbff4455a05a0e817ccc41d4 : SUCCESS in 12m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ced00b493c84887b0ecb14b89267b91 : SUCCESS in 1h 41m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/51b6ffb7b70c48168b567c04124dcffd : SUCCESS in 1h 12m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fad6c20aad074b4a909aa525ea665d83 : SUCCESS in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcfedb52a01548c38cbee03d401e88da : SUCCESS in 1h 49m 54s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/7306370178ea447a8ca1c622fffa44e9 : SUCCESS in 36m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/adf84bd686b34ae89cdc03deeccfd50d : SUCCESS in 39m 53s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9fff4567f4a94110bad478e64ad050c9 : SUCCESS in 35m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/746a546c70c447bb94913848a5d58495 : SUCCESS in 46m 20s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6f4138d60d5e4e9696adeba806f9260e : SUCCESS in 17m 08s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9b33aacc2c7143578cf0cb9c4f8a9432 : SUCCESS in 1h 56m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9ebe90315dd43ce9ea6cecbd5a3c677 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":11},{"id":"a95fc0534ba2e5e1390201f0cba8ead3478e9c22","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-09-11 17:47:18.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"4891d28503d818c2887b117dad972a26e556c21b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 18:09:12.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"aefb9b4dba3e6e5df81b2faa08b10aecc9ea063b","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-09-11 18:58:04.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6b8c134156825ae94c4eb270ce398f93db651a20","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 19:04:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"37c4ea0533e885e5151637f83311432f32036675","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-09-11 19:07:24.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)\n\nlgtm","accounts_in_message":[],"_revision_number":11},{"id":"67b4d40c92996e9ef1da7a26efc34aec527dd140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 23:14:17.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"a9980cfb220d9e48ea552184b6d9c22b1406c4ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 00:57:40.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"e7efa7afcee83f82fdd285fe3bcf83a317f7fbab","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-12 06:02:07.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"cfdfac7ecec3b55ceeac2141e39d74c9273bb6c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:36:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63ecb22c3bde4b929b5286cdb98c58ce : SUCCESS in 49m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80335141d7c64840adfd46c63ba6dbf3 : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d8bd52b74044bca48516a9a8221c73 : SUCCESS in 8m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53ebae685b784d54875e0adb928cc0fb : SUCCESS in 43m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/315eef7696e14c52b9e5c3859511c166 : TIMED_OUT in 1h 00m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/866d3bc0b2cf41849b42e2167e76a6ff : SUCCESS in 32m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344e135fd5984674a7d3d32e3335f297 : SUCCESS in 16m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/948e6ce857a241248653017637e622cb : SUCCESS in 1h 47m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a33ab01ef8644ea4b0d3487681744324 : SUCCESS in 1h 05m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9a89f3fea6a4062aa2889f73bc6cf1a : SUCCESS in 1h 09m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86eaaf247cb74be5beeaa1ebe7d6a8ed : SUCCESS in 1h 35m 00s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/84d0ff69ec1b4d818da79f10599ae41c : SUCCESS in 40m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/60ac320e938043178a7fb34bed69429d : SUCCESS in 37m 08s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ef745f66e396421eb1bcdfed8977a68d : SUCCESS in 40m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/bf2c52b5b7c04027bbe7d11a78ef8123 : SUCCESS in 40m 40s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6e9adc28abc04512a0331119ee6abdb4 : SUCCESS in 17m 33s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/497b1a4a1b22464e8bd07a9a3d1287ed : SUCCESS in 1h 31m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a1bd3a2db104c71bf6b76c9ee573c04 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":11},{"id":"ba0e8a87a07bc1aeb35c45ec72a2b20ce6f1f065","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-12 12:33:15.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"a3a4178d6642540842d4322add6581e352df3063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 01:38:23.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d459d36ca53a4804afd5a448fdfac9f3 : SUCCESS in 45m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c8020d2914b44a7993dff89d4b9e447 : SUCCESS in 39m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb0966929d849c9848b4a94f02cca0d : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1378b1dcdf56438dbf6ba232f38bf3b9 : SUCCESS in 34m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50765e2354854515b5fe2fc8c4183d82 : SUCCESS in 37m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8661e7d6842247b9b457a6f2d0eb2ebe : SUCCESS in 29m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07fcc8d714b4a2493125322417796a2 : SUCCESS in 12m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2bc7c785834e4feda4372a58da08da4e : TIMED_OUT in 2h 06m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2703a3f1e4b34ab5966290a6e81c126b : SUCCESS in 59m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4c34f68bd7a40fdb64821f6a5c8d8f4 : SUCCESS in 1h 00m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f67f314807049e78019a6a7f448c10d : SUCCESS in 1h 21m 14s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b272a203cdc4857b292928e90c1167b : SUCCESS in 33m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/da49ae4349f6403dbdef66487f60fea3 : SUCCESS in 34m 46s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/90198b74e50448ad9a66dc4044352923 : SUCCESS in 36m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d436a445498d476cbd8a057a7553667b : SUCCESS in 37m 06s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b127f72e2d414db5aa27499485c86f30 : SUCCESS in 18m 31s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/131c338b9fa649c39227dcbaf23f9ab4 : SUCCESS in 1h 29m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/732a683ab4f748a08f1648fe9ce2b3aa : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":11},{"id":"1d4dee7ceda6eb37f116634d192f8f9c302abc02","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-13 01:39:05.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c399c066b6ec6bdd3ed527afcb8de9f357db9c7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:03:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76625219b8c347389a05fa9b8f31eabe : SUCCESS in 43m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d098e4a4936415a8e0d5d8bbc18c2d4 : SUCCESS in 46m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0838e4e050e4250b1ecc6e3bd4cd1a7 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e0c94ac7c9f417089d5c627fb0aee62 : SUCCESS in 32m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0bfe858729f48ee9b609c53e4e5abea : SUCCESS in 38m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97204353217e48efaa2205d782b69a31 : SUCCESS in 32m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709c76c8d2c542aabb69764672fd7b95 : SUCCESS in 13m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2236f9e14c184347b5931a0823a02aee : SUCCESS in 1h 43m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5978473401c943b288e4a2f9db025353 : SUCCESS in 1h 00m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e235493715543c2bb17c6d7603c1f64 : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2f1c02f517348e196e75e415ea38ab5 : SUCCESS in 1h 33m 04s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ac1ad14780947c089b39d8b5ebce44c : SUCCESS in 47m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/12998506e0a34435bd23df8cd68b82b2 : SUCCESS in 37m 21s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d03bd6e51d00447f8584c76c21da428b : SUCCESS in 42m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/a1f5dc02ea7c4abe94faaf7ab3c16ce9 : SUCCESS in 34m 39s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c06ea95c3afa4856a45c503672f47a92 : SUCCESS in 18m 50s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d1b95ef760a34469beebd82833a816db : SUCCESS in 1h 30m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9143a51e9b417688d6e63c135fc139 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":11},{"id":"f5286300db61754f82f876ed7cf919fd4108a533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 23:29:47.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"8c84f9e1470d07c3e7622f0a1a8117a5b7329927","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-13 23:58:20.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"aea748b135262079caa28547a2929f930a47b3c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:10:25.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0723eecc7db64ef485e1d2dfccaa4b47 : SUCCESS in 43m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe8fa9638124b6ea93605d56d104db8 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05a255a9a698449f90d96f3c5f0ea902 : SUCCESS in 43m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c8e3e4a42af4f20acde765a0f771af0 : SUCCESS in 39m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cfe7111b46c41e5bc7dbfa14cb91a27 : SUCCESS in 54m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d4bbe31db64dae92efa0aa225e4a7c : SUCCESS in 15m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b85f8f48c1447aea5c8b3a10f302267 : SUCCESS in 1h 37m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/745f136ff0ce493d90da7eace4b52fee : SUCCESS in 57m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e28889b61b384039ab5f61ddeb124a98 : SUCCESS in 53m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d200a1c736d42538777632a0a075e72 : SUCCESS in 1h 25m 02s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/1525a92721a743398c446a60655c4d22 : SUCCESS in 37m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/47c0223869ad43a68699d8072b44d120 : SUCCESS in 37m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0503562c58b4782abe2d2989ecfade0 : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":11},{"id":"9e1d593ceb98c3ac5f4afac45e87a6cbc1dee336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:10:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"fcf56f3c789b48ff33fc26e56052cff7e8a63f66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:13:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592c48afe44e4febaabc5579682a3bec : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae","revisions":{"20b338ac01c1a02ffa73f98f817dbfcb2a8a4e80":{"kind":"REWORK","_number":1,"created":"2019-08-07 23:23:24.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/1"}}},"commit":{"parents":[{"commit":"4aeae16e140a0b0f05f48f624d61831aaa5d7512","subject":"[WIP] Add immutable resource option for roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4aeae16e140a0b0f05f48f624d61831aaa5d7512"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20b338ac01c1a02ffa73f98f817dbfcb2a8a4e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20b338ac01c1a02ffa73f98f817dbfcb2a8a4e80"}]},"branch":"refs/heads/master"},"a44acf5e31aef681716ee8492d5b0413aed14ec3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-09 03:26:33.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/2"}}},"commit":{"parents":[{"commit":"4c9c57f0e871a2d2edc5e7ce7fa81ca391dc4819","subject":"[WIP] Add immutable resource option for roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4c9c57f0e871a2d2edc5e7ce7fa81ca391dc4819"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:18:23.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a44acf5e31aef681716ee8492d5b0413aed14ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a44acf5e31aef681716ee8492d5b0413aed14ec3"}]},"branch":"refs/heads/master"},"a40c1c090637bd05309e103a77a26a06ff12228c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-09 18:02:57.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/3"}}},"commit":{"parents":[{"commit":"ce7c927f0a0576d7951100815f1a8cce6c3e0395","subject":"[WIP] Add immutable resource option for roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ce7c927f0a0576d7951100815f1a8cce6c3e0395"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 18:02:52.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a40c1c090637bd05309e103a77a26a06ff12228c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a40c1c090637bd05309e103a77a26a06ff12228c"}]},"branch":"refs/heads/master"},"3b3f6ca25d8d335997dd0e63dfdee23a68991c20":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-12 01:49:37.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/4"}}},"commit":{"parents":[{"commit":"53d92358f1ec96c4f5e97b188d9c89b83882f19b","subject":"[WIP] Add immutable resource option for roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/53d92358f1ec96c4f5e97b188d9c89b83882f19b"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 01:49:32.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3b3f6ca25d8d335997dd0e63dfdee23a68991c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3b3f6ca25d8d335997dd0e63dfdee23a68991c20"}]},"branch":"refs/heads/master"},"4f94b369d34e460b499e702329c432790b95b2ad":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-12 04:19:23.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/5"}}},"commit":{"parents":[{"commit":"61fee6931b1370dd2f8b67557dbbcf25fc3ca16a","subject":"[WIP] Add immutable resource option for roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/61fee6931b1370dd2f8b67557dbbcf25fc3ca16a"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 04:19:02.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f94b369d34e460b499e702329c432790b95b2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f94b369d34e460b499e702329c432790b95b2ad"}]},"branch":"refs/heads/master"},"75ba9da5464bae226e3414eba4de46d0fbf4ea00":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-24 19:28:40.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/28/675228/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/6"}}},"commit":{"parents":[{"commit":"d1281377307067273e8f5ee138d6f81f172c788e","subject":"Add immutable option for roles and projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d1281377307067273e8f5ee138d6f81f172c788e"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2019-08-24 19:28:40.000000000","tz":0},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/75ba9da5464bae226e3414eba4de46d0fbf4ea00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/75ba9da5464bae226e3414eba4de46d0fbf4ea00"}]},"branch":"refs/heads/master"},"5025de43f3fff3b285dc12bf25b6381183fc7c62":{"kind":"REWORK","_number":7,"created":"2019-08-29 17:42:55.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/28/675228/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/7"}}},"commit":{"parents":[{"commit":"d1281377307067273e8f5ee138d6f81f172c788e","subject":"Add immutable option for roles and projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d1281377307067273e8f5ee138d6f81f172c788e"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-08-29 17:41:54.000000000","tz":330},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5025de43f3fff3b285dc12bf25b6381183fc7c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5025de43f3fff3b285dc12bf25b6381183fc7c62"}]},"branch":"refs/heads/master"},"1ecdb401a2741ae6574aaa6f4f4997f08cedad15":{"kind":"REWORK","_number":8,"created":"2019-09-09 23:36:33.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/8"}}},"commit":{"parents":[{"commit":"aadb1cdfe37ea20e6dfd912700359c2fcc7d89e4","subject":"Add immutable option for roles and projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aadb1cdfe37ea20e6dfd912700359c2fcc7d89e4"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-09 23:29:51.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ecdb401a2741ae6574aaa6f4f4997f08cedad15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ecdb401a2741ae6574aaa6f4f4997f08cedad15"}]},"branch":"refs/heads/master"},"1340e230053502a4e1765219ba1ba6b09900bb22":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-09 23:50:09.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/9"}}},"commit":{"parents":[{"commit":"67a875336ca896a718365f7481d032bb212da08f","subject":"Add immutable option for roles and projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67a875336ca896a718365f7481d032bb212da08f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-09 23:48:29.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1340e230053502a4e1765219ba1ba6b09900bb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1340e230053502a4e1765219ba1ba6b09900bb22"}]},"branch":"refs/heads/master"},"d316058642231c993c321b4fea4d155f42b3e6fc":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-09-10 06:31:29.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/10"}}},"commit":{"parents":[{"commit":"90be7f5c0658b108424f0a0010b25eea091e62e2","subject":"Add immutable option for roles and projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/90be7f5c0658b108424f0a0010b25eea091e62e2"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-09-10 06:31:29.000000000","tz":0},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d316058642231c993c321b4fea4d155f42b3e6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d316058642231c993c321b4fea4d155f42b3e6fc"}]},"branch":"refs/heads/master"},"a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-10 07:03:40.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/28/675228/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/675228/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/675228/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/675228/11"}}},"commit":{"parents":[{"commit":"a38d4a650f3573065848a372254c63c3a6598979","subject":"Add immutable option for roles and projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a38d4a650f3573065848a372254c63c3a6598979"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-07 23:22:05.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-10 07:03:20.000000000","tz":-420},"subject":"Add --immutable-roles flag to bootstrap command","message":"Add --immutable-roles flag to bootstrap command\n\nThis implements Step 2 of the Proposed Change for Immutable\nResources[1].\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: I4d99f630cb16e1d58261012e59d3a92c7035734c\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
