)]}'
{"id":"openstack%2Fkeystone~675509","triplet_id":"openstack%2Fkeystone~master~Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1","project":"openstack/keystone","branch":"master","topic":"immutable-resources","hashtags":[],"change_id":"Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1","subject":"Add immutable roles status check","status":"MERGED","created":"2019-08-09 03:26:33.000000000","updated":"2019-09-14 16:30:23.000000000","submitted":"2019-09-14 16:18:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675509-1568477932185-f6217199","meta_rev_id":"174fe10d43e63e7809c585d455478d93f69086ea","_number":675509,"virtual_id_number":675509,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-14 01:24:38.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-14 16:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-09-11 19:13:25.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2019-09-14 11:52:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-14 08:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-11 18:52:17.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":1,"date":"2019-09-13 23:58:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-09-11 19:13:25.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 18:52:17.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-09-11 19:13:25.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-09-13 23:58:00.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 08:58:24.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 11:52:29.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-14 16:18:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f62ca879abb46de343ff3fd66a886394e3ad4ca2","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-09 03:26:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c38c1b82b1b2dc9af36291969a4350b09add70b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 05:11:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/09/675509/1/check/openstack-tox-cover/720c408/ : FAILURE in 36m 40s\n- openstack-tox-lower-constraints https://logs.opendev.org/09/675509/1/check/openstack-tox-lower-constraints/d70ce95/ : FAILURE in 34m 18s\n- openstack-tox-pep8 https://logs.opendev.org/09/675509/1/check/openstack-tox-pep8/94e5bf3/ : FAILURE in 5m 17s\n- openstack-tox-py27 https://logs.opendev.org/09/675509/1/check/openstack-tox-py27/9bd9145/ : FAILURE in 23m 28s\n- openstack-tox-py36 https://logs.opendev.org/09/675509/1/check/openstack-tox-py36/0508c0c/ : TIMED_OUT in 40m 57s\n- openstack-tox-py37 https://logs.opendev.org/09/675509/1/check/openstack-tox-py37/f4dfc82/ : TIMED_OUT in 41m 23s\n- openstack-tox-docs https://logs.opendev.org/09/675509/1/check/openstack-tox-docs/30a2e66/html/ : SUCCESS in 7m 11s\n- tempest-full https://logs.opendev.org/09/675509/1/check/tempest-full/3d688f9/ : SUCCESS in 1h 38m 39s\n- neutron-grenade https://logs.opendev.org/09/675509/1/check/neutron-grenade/aa7030b/ : SUCCESS in 1h 00m 59s\n- grenade-py3 https://logs.opendev.org/09/675509/1/check/grenade-py3/c6d15a2/ : SUCCESS in 55m 22s\n- tempest-full-py3 https://logs.opendev.org/09/675509/1/check/tempest-full-py3/3f3b51e/ : SUCCESS in 1h 39m 52s\n- build-openstack-releasenotes https://logs.opendev.org/09/675509/1/check/build-openstack-releasenotes/b330405/html/ : SUCCESS in 11m 31s\n- keystone-dsvm-functional https://logs.opendev.org/09/675509/1/check/keystone-dsvm-functional/c711f25/ : SUCCESS in 43m 17s\n- keystone-dsvm-py3-functional https://logs.opendev.org/09/675509/1/check/keystone-dsvm-py3-functional/1c65a33/ : SUCCESS in 28m 36s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/09/675509/1/check/keystone-dsvm-functional-federation-opensuse15/db44c24/ : SUCCESS in 34m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/09/675509/1/check/keystone-dsvm-py3-functional-federation-opensuse15/d15095e/ : SUCCESS in 40m 32s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/09/675509/1/check/keystoneclient-devstack-functional/c901a5a/ : SUCCESS in 12m 56s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/09/675509/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/fca65c5/ : SUCCESS in 1h 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f3c591cad1f8de3e1c70b7e91508cffb8355ae9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-09 18:02:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"97f3cba0a63815dcbf5dd305431c602dd3944f65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:50:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/09/675509/2/check/openstack-tox-cover/70a2450/ : FAILURE in 30m 41s\n- openstack-tox-lower-constraints https://logs.opendev.org/09/675509/2/check/openstack-tox-lower-constraints/5a550c7/ : FAILURE in 35m 26s\n- openstack-tox-pep8 https://logs.opendev.org/09/675509/2/check/openstack-tox-pep8/cb84846/ : FAILURE in 5m 45s\n- openstack-tox-py27 https://logs.opendev.org/09/675509/2/check/openstack-tox-py27/43bdb97/ : FAILURE in 26m 38s\n- openstack-tox-py36 https://logs.opendev.org/09/675509/2/check/openstack-tox-py36/1d731bb/ : FAILURE in 27m 17s\n- openstack-tox-py37 https://logs.opendev.org/09/675509/2/check/openstack-tox-py37/d082157/ : FAILURE in 27m 26s\n- openstack-tox-docs https://logs.opendev.org/09/675509/2/check/openstack-tox-docs/56a92e0/html/ : SUCCESS in 8m 54s\n- tempest-full https://logs.opendev.org/09/675509/2/check/tempest-full/234cbe5/ : SUCCESS in 1h 43m 16s\n- neutron-grenade https://logs.opendev.org/09/675509/2/check/neutron-grenade/a33dfc4/ : SUCCESS in 1h 01m 51s\n- grenade-py3 https://logs.opendev.org/09/675509/2/check/grenade-py3/feee20a/ : SUCCESS in 58m 49s\n- tempest-full-py3 https://logs.opendev.org/09/675509/2/check/tempest-full-py3/449ffbc/ : SUCCESS in 1h 23m 17s\n- build-openstack-releasenotes https://logs.opendev.org/09/675509/2/check/build-openstack-releasenotes/25f4854/html/ : SUCCESS in 9m 46s\n- keystone-dsvm-functional https://logs.opendev.org/09/675509/2/check/keystone-dsvm-functional/14d927b/ : SUCCESS in 34m 18s\n- keystone-dsvm-py3-functional https://logs.opendev.org/09/675509/2/check/keystone-dsvm-py3-functional/bde70ce/ : SUCCESS in 35m 05s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/09/675509/2/check/keystone-dsvm-functional-federation-opensuse15/8c7d6de/ : SUCCESS in 43m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/09/675509/2/check/keystone-dsvm-py3-functional-federation-opensuse15/f526afd/ : SUCCESS in 34m 51s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/09/675509/2/check/keystoneclient-devstack-functional/f190788/ : SUCCESS in 16m 13s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/09/675509/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/3a93859/ : FAILURE in 1h 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af1a13abca23012cbde8c764cd1bb84a4f8ba55b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 01:49:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ddb40f255724a3bd3eb4635b296f7876bae7591b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 02:43:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/09/675509/3/check/openstack-tox-cover/1d56134/ : FAILURE in 29m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/09/675509/3/check/openstack-tox-lower-constraints/8e42a29/ : FAILURE in 29m 20s\n- openstack-tox-pep8 https://logs.opendev.org/09/675509/3/check/openstack-tox-pep8/5426ea4/ : FAILURE in 5m 30s\n- openstack-tox-py27 https://logs.opendev.org/09/675509/3/check/openstack-tox-py27/97ced1b/ : FAILURE in 28m 29s\n- openstack-tox-py36 https://logs.opendev.org/09/675509/3/check/openstack-tox-py36/40cc384/ : FAILURE in 22m 45s\n- openstack-tox-py37 https://logs.opendev.org/09/675509/3/check/openstack-tox-py37/34d006c/ : FAILURE in 23m 14s\n- openstack-tox-docs https://logs.opendev.org/09/675509/3/check/openstack-tox-docs/faf98f5/html/ : SUCCESS in 8m 08s\n- tempest-full https://logs.opendev.org/09/675509/3/check/tempest-full/a3b6435/ : FAILURE in 18m 40s\n- neutron-grenade https://logs.opendev.org/09/675509/3/check/neutron-grenade/0c3fb64/ : FAILURE in 47m 53s\n- grenade-py3 https://logs.opendev.org/09/675509/3/check/grenade-py3/116cbd3/ : FAILURE in 44m 37s\n- tempest-full-py3 https://logs.opendev.org/09/675509/3/check/tempest-full-py3/17c38e3/ : FAILURE in 19m 31s\n- build-openstack-releasenotes https://logs.opendev.org/09/675509/3/check/build-openstack-releasenotes/099dd58/html/ : SUCCESS in 11m 54s\n- keystone-dsvm-functional https://logs.opendev.org/09/675509/3/check/keystone-dsvm-functional/cffe46b/ : FAILURE in 16m 13s\n- keystone-dsvm-py3-functional https://logs.opendev.org/09/675509/3/check/keystone-dsvm-py3-functional/e1b461f/ : FAILURE in 17m 31s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/09/675509/3/check/keystone-dsvm-functional-federation-opensuse15/bbf12f8/ : FAILURE in 15m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/09/675509/3/check/keystone-dsvm-py3-functional-federation-opensuse15/798985a/ : FAILURE in 27m 23s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/09/675509/3/check/keystoneclient-devstack-functional/364579c/ : FAILURE in 10m 35s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/09/675509/3/check/legacy-tempest-dsvm-ldap-domain-specific-driver/97e1571/ : FAILURE in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb579ce2b70d78ae3bc81cdcc0334acc11d0569d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 04:19:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"dab3c1d6a5844311407862eda9c3ad550b7b9326","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 06:47:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/09/675509/4/check/openstack-tox-cover/86f0975/cover/ : SUCCESS in 38m 29s\n- openstack-tox-lower-constraints https://logs.opendev.org/09/675509/4/check/openstack-tox-lower-constraints/45f32f8/ : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://logs.opendev.org/09/675509/4/check/openstack-tox-pep8/3da4d1a/ : FAILURE in 5m 28s\n- openstack-tox-py27 https://logs.opendev.org/09/675509/4/check/openstack-tox-py27/1599d0e/ : SUCCESS in 37m 52s\n- openstack-tox-py36 https://logs.opendev.org/09/675509/4/check/openstack-tox-py36/0e97de5/ : SUCCESS in 29m 35s\n- openstack-tox-py37 https://logs.opendev.org/09/675509/4/check/openstack-tox-py37/c89bf71/ : SUCCESS in 34m 12s\n- openstack-tox-docs https://logs.opendev.org/09/675509/4/check/openstack-tox-docs/fd26d98/html/ : SUCCESS in 7m 32s\n- tempest-full https://logs.opendev.org/09/675509/4/check/tempest-full/e5b15a3/ : SUCCESS in 1h 33m 13s\n- neutron-grenade https://logs.opendev.org/09/675509/4/check/neutron-grenade/2cbd985/ : SUCCESS in 1h 22m 07s\n- grenade-py3 https://logs.opendev.org/09/675509/4/check/grenade-py3/1d6dc0c/ : SUCCESS in 55m 17s\n- tempest-full-py3 https://logs.opendev.org/09/675509/4/check/tempest-full-py3/b3a8373/ : SUCCESS in 1h 24m 44s\n- build-openstack-releasenotes https://logs.opendev.org/09/675509/4/check/build-openstack-releasenotes/445d5bf/html/ : SUCCESS in 8m 55s\n- keystone-dsvm-functional https://logs.opendev.org/09/675509/4/check/keystone-dsvm-functional/6151b4d/ : SUCCESS in 28m 50s\n- keystone-dsvm-py3-functional https://logs.opendev.org/09/675509/4/check/keystone-dsvm-py3-functional/3bab387/ : SUCCESS in 39m 31s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/09/675509/4/check/keystone-dsvm-functional-federation-opensuse15/3b9e8cc/ : SUCCESS in 33m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/09/675509/4/check/keystone-dsvm-py3-functional-federation-opensuse15/9db9ff3/ : SUCCESS in 34m 52s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/09/675509/4/check/keystoneclient-devstack-functional/16b5a1b/ : SUCCESS in 14m 41s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/09/675509/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/24c1cb4/ : SUCCESS in 2h 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"60cfe8d9ab373692daf838c99b990e8f020fb04b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 14:18:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"68332fa1af4ee6cd6d59b9e02f8ae4740abb95bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 17:39:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5301bc3bcff4e6097f52b165cfdf05c : SUCCESS in 36m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f93f9bfe784d6e8fe04d697d5528bf : SUCCESS in 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909ccb77c50e43a393c5d99e0558602a : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/427748e39ec8490f933b14a1158a8cc0 : SUCCESS in 30m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06f039f69b874a3abf59ef6ce78a0e3c : SUCCESS in 25m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad82c34c4760415e9f7484e6d925731c : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b4c5c2720b44e39a0299249a8bb92e : SUCCESS in 7m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4963e11a89cf47b2a1e4465171c587bb : SUCCESS in 1h 43m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c14209c15914bc8b617d1387b66ef2f : SUCCESS in 58m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0eece52b300441488c84bb4de04bdf2c : SUCCESS in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc5d20a1b3ca44aabee033acd2e469aa : SUCCESS in 1h 28m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3a252fdd1324cb8971bac682a65673d : SUCCESS in 11m 42s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/45c67349967e4fa287078606a7664896 : SUCCESS in 31m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9c9b3fe6d89d43ce870019841739bfbb : SUCCESS in 34m 56s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5947d044b83e43da94235d50da0191cd : SUCCESS in 40m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0c9234713a054b04a919b542f30034b7 : SUCCESS in 40m 36s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d68fa36921d14b209eae33ae0a231813 : SUCCESS in 19m 36s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c6fae9b3c1d54fa9a01699a26a9e1f14 : FAILURE in 1h 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"50a91232fef7cb1eb5c7ff44005a735d8bd5185b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-13 20:46:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6566d7466f1e3d2fd9cc4771843ceaf420477c60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 22:58:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6881fe07664620aecd005c6698fa46 : SUCCESS in 35m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4213445723b445e29fd04e108d7ba646 : SUCCESS in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778d4943afbd40189cd0e1fa06244261 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5df1a6c2f77d4ff389583ccf704075fa : SUCCESS in 24m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bd8ec9a52f24ce9844863dba3de3dad : SUCCESS in 28m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ef8f8f5b3d84187a86bfa833f15c5f2 : SUCCESS in 29m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4401bf485cfa4d949f1f06729ccc656f : SUCCESS in 9m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d1080a0b97d24af18924d550c18aae9e : SUCCESS in 1h 32m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58ee202f6f774665b850c91f62a87d26 : SUCCESS in 53m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/008e3ded26864e4d89c43d3d7b4232d7 : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbc56d5ddfea4b1eac6848ab08989b60 : SUCCESS in 1h 20m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f217b4e7f204459b5b77d11096917d1 : SUCCESS in 9m 56s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/b035d70c07c347a58ca29ee67de79c09 : SUCCESS in 33m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d1d6d74ad82b43e6a28f9107f28b02c1 : SUCCESS in 31m 28s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9306f29af02c40acbf28b0320b2cd929 : SUCCESS in 35m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cbe45f0a7691428dbb7d6417ea1c0472 : SUCCESS in 44m 39s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/18981faa23fc4f6eb08e698068d60103 : SUCCESS in 14m 50s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e2efb952ca7742bb8d60bbfc21d73405 : FAILURE in 2h 03m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f481e1e0581a405589b6f615ab20fce1 : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":6},{"id":"3a7fb318f51f9bea2605ea71b3c8473a08c10671","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-24 19:28:47.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"7abefaf595167fd6269fc89e2ec4c6b9fa90e6b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 21:08:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed539929b5648249dd8eca29e899d82 : SUCCESS in 40m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6837840ecbe646899f65c75688d5f506 : SUCCESS in 37m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60478e3b03d74902a3e203fa5fc1fec1 : FAILURE in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66dad030741f4d09983d2393fe75df1c : SUCCESS in 36m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e9b6688f5bc430b946032e012b1154d : SUCCESS in 35m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/863051206fb548a59fe47173962479e7 : SUCCESS in 32m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7233061bad4a20aa32b0f14f3d7d39 : FAILURE in 6m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/66c664ff9ff34b4d84c4b72ba33ed686 : FAILURE in 1h 23m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5aea09eb65ea44c8ae688bc58a9a7aeb : SUCCESS in 55m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0b43a6e9ee2a4150a7d4f3a7c8ca9ee4 : SUCCESS in 1h 01m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/225d4cc8096a441eab4085b77860006f : FAILURE in 1h 14m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba7ba05b238f4f5fa53b775838deb749 : SUCCESS in 12m 57s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef5abfde1e284efaa1ac22f7ec94805a : SUCCESS in 36m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4c7c3a50c70241699d2126254429cb0b : SUCCESS in 39m 56s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f1d4900238a94d66ba82c3a1e8d40806 : RETRY_LIMIT in 4m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/021893d8733240f98ae260c639941d0f : RETRY_LIMIT in 4m 13s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/69ecaa7acdfd4bd59c5c90f482b7858b : SUCCESS in 17m 01s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8719c912ba0b4f7f994c5291f665b099 : FAILURE in 1h 21m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac05b068d11b4a30913abafd3f132bdb : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":7},{"id":"a02e4c38a05ea474cd93a40b0ff4a6a005dba65c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-08-29 17:44:35.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"2fc71d4cdc9ee08e2cc1fc837d0ef146655c18c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 22:10:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50da035c489a489eb7dc00a42e3f72e8 : SUCCESS in 38m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1ec05b7acd046e29561ee317bce7066 : SUCCESS in 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab9efba37f8f476db7ba2adff7442100 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/259ac69658184b2e8c3c08619d7cf872 : SUCCESS in 33m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a5c5d04d22845d3aa142ab51ae87e3f : SUCCESS in 26m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83a2a777324e4e08a7d38343ff3bff59 : SUCCESS in 29m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748b8e40256a46b8af0475932c166544 : SUCCESS in 16m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7aa3b3044d024cc69972f65407e85594 : FAILURE in 1h 34m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7275591818aa419ea59f078bddcaddbe : SUCCESS in 1h 08m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/130dbdff32cb41068436a3f2e779d1a6 : SUCCESS in 1h 09m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f40ff8c17cca4b5f9d603f9f9f2c4da2 : FAILURE in 1h 22m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b47aa8b263014f68bde21eda80f18562 : SUCCESS in 14m 06s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/44c4ce6282c245de99ec7ab39221450d : SUCCESS in 33m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b371014c338b45e58604c11652939e60 : SUCCESS in 37m 34s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cea964e4bc094010a1560eeddea573f9 : SUCCESS in 48m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ff931d9b8e494964b99f415ce1cbb3b4 : SUCCESS in 41m 26s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f785ff71557e4da4ab413d4adb55612f : SUCCESS in 20m 25s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a9e57c664d1641daaf868d569916122c : FAILURE in 1h 15m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f90713fb15a741e3a83d8a32330b270c : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":8},{"id":"2165bb00a33bad9ef5e1acf48cc41018e6b2a2dd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-09 23:44:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"86a180555beafcd1768acc8412668fad227da054","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-09 23:50:09.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"e8d90b83652e9fd0b034e2d9ac4c1191e4e41b59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:26:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03b214077bcb454eb525b2ef6034c9fc : FAILURE in 37m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45d81724ca7040f29240659cc197301a : TIMED_OUT in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e294fdeeaa87469884a706311843838a : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ed6e71e45d14bb9b85b1df554563a0b : FAILURE in 40m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e2f9d81c7242adb922dfe4aa6e90cd : FAILURE in 49m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a11afc547edc4ce982c36ca3510e118b : FAILURE in 38m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb526ecebf04e5b956343f87877d4a5 : SUCCESS in 10m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b867b83a4f541189dee94a22c99179b : SUCCESS in 1h 54m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/670ae7b2e63d4888980e65dbbc8baeb1 : SUCCESS in 57m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1467e9ae0392456a9701657b85d44ec6 : FAILURE in 1h 00m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8446b2997ea44d9f885e9f5301978fc8 : SUCCESS in 1h 35m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e0fa86c1a354abaa1629b6c87a07822 : SUCCESS in 12m 57s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5ad0a0d5c5143e0b11f81530be73d38 : SUCCESS in 32m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2d00fec8afc04bb7953f749f332373ec : SUCCESS in 35m 54s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/a5d32eb1f46b4589bbc0556a1ab80f8a : SUCCESS in 37m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/05e184f447db47caae88f1269ffffd3b : SUCCESS in 41m 31s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7af909741a3e4d19a0d530f10e526d74 : SUCCESS in 17m 29s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3c32dcd5cfc44a0d8bf4e85d8ae89ac2 : SUCCESS in 1h 30m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fb0e44ae0bf4882a762a53c8533979d : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":10},{"id":"8d76ec5b68d94a752d076103fd807af1dfa046e6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 06:31:34.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"bce7ceef0332c91eb457046f55fec42084c8994c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 07:03:40.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"0c693b61aa6769a0da82715d99421d13e0c6658b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:55:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc2a9d96c104bd3bed3545fd29a7533 : SUCCESS in 49m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e843ed310ddc492982916e141d3bb23e : TIMED_OUT in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca22711f3131405196705dd22a63eb4a : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/260d3cdb20c7461484bb2bab0b1d5d8a : SUCCESS in 38m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aa90625a6b74ad980c1cf0d365db820 : TIMED_OUT in 1h 00m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdbc16ce261b48ddbcc7508a80113445 : SUCCESS in 35m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9272dce546454adebcd3b4df49c45b9f : SUCCESS in 10m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4a8024a72dd84b7f87c8b7188cc85251 : SUCCESS in 1h 51m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/546ecba298b047fe942bbba99890ad1e : SUCCESS in 58m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/adaca531831c46ca90a9dc3f3a6db12a : SUCCESS in 58m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb0dfd64e34148f78fb6d3ee9c868803 : SUCCESS in 1h 32m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2081c4acef464d6fa24cfc58685f6260 : SUCCESS in 11m 06s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce2a1c12be0b4d729db92f26cfd3d1d0 : SUCCESS in 35m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f6b9ddda8c1648e392f2983d2a5f7c0b : SUCCESS in 42m 14s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/a687db977f474035b69ec0cdd759b49e : SUCCESS in 41m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/87714f16c8eb429291b0499e7dae6003 : SUCCESS in 40m 11s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2c76929045004f798a651352fa8dd095 : SUCCESS in 18m 14s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/08ca7a0c005b40cba4f03c604d96b6e5 : SUCCESS in 1h 37m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9d0f9fdfd2f47ef9c6d5bb0324077e0 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":12},{"id":"70d856ff7cbb29535f84cbed11c0aeb19594f724","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 15:09:53.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"06822f250b7704e49b006b21fcbdb1dc073da034","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-10 16:18:22.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"19c774c4614a54007f7d53b9a918b68534f97872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:35:45.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65e5228d773d49d19ddf91bf325113a4 : SUCCESS in 53m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8d3e9ecae594a849af71c6e227a8f40 : TIMED_OUT in 41m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13899919bc464f81b9978e77637d424f : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c6760ee69ec4b618a355bec9c453482 : SUCCESS in 42m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30ef2a79abf0478ebe828cd71535b5eb : SUCCESS in 27m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b560b052876e4a83bd4ec1b5bd89a5eb : SUCCESS in 31m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c970ab0cbe4fadb228b6b0afaaa7da : SUCCESS in 10m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/35a79f4c748945d4bd624acf43af4357 : SUCCESS in 2h 03m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ae961dfd1c446fb80b1970edd3679d1 : SUCCESS in 1h 07m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ec1c8456e791488ca1631920b3aeb6fd : SUCCESS in 51m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63a8f70c44f64e298b4ef02f6f4ed199 : SUCCESS in 1h 37m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41afacb1491a43c991443c8531191ae5 : SUCCESS in 12m 23s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9f311268d244485b85c0b579ec50ed4 : SUCCESS in 35m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/411f03f4b2fe42ddacbd31f540817b0a : SUCCESS in 36m 53s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5dbbe7d42c2f4e83886496208e52be58 : SUCCESS in 40m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e4a803d7dae746d98d1df04ee37177ee : SUCCESS in 44m 56s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/70ff99bcc7f64ad2a34ceda50a36eceb : SUCCESS in 21m 40s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/adf042f6739e4777be4f3ab4b28509c3 : SUCCESS in 1h 56m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57ba055cb5f84e8c86e6bb8e04397d85 : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":12},{"id":"6240dfe83d23de41a0eb1a176156e752eb448107","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 23:39:35.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"7fb157dba282ae22dfafd02360ea45628831b459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 02:11:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f98e0e33b094154b880e7ad6579e0c8 : SUCCESS in 50m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bac929fdc65247a5ab95976f1a4381fe : SUCCESS in 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6244a7e71854c22831193d31a709b56 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35b5264e13424f4faba035b586f99b59 : SUCCESS in 31m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa88f7f9c26f4f0db0f8d09e95f0eaf3 : SUCCESS in 33m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f59dc795d7b4de49559bf53ad5ba668 : SUCCESS in 30m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e93c31554af4052be490d58cfdd410b : SUCCESS in 10m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b62a9c622404243ad6013e1a41ad7e2 : SUCCESS in 1h 57m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8c56c5577baf4995847a4a4d0c98480c : SUCCESS in 53m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4001ec127e464c489ffa0266ab9bcd47 : SUCCESS in 1h 03m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/013861d3e9914525a2c633718abd5f43 : SUCCESS in 1h 33m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abad6a19d5594dd9bf85ad2c2eaccd2d : SUCCESS in 14m 06s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/61564b7a286e4b808d1ffa7cda1136b8 : SUCCESS in 31m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f1636ef00d6a42c386f11279fd4feb41 : SUCCESS in 34m 55s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/977f1a5c1cd24c7f9075643c8f52da74 : SUCCESS in 38m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/43e15767feb34c38a4b4b6c48305064b : SUCCESS in 41m 33s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/431e8e154fd94150921414ec674e5742 : SUCCESS in 20m 15s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4e9399b210984fd0ab88f0f7cf4ae226 : SUCCESS in 1h 46m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dc82775991548d9a0a5ff46e0fb1701 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":12},{"id":"40bbd4f5e18db538c726437ab48eb400a5736f5e","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-09-11 17:33:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bb36465b87f6fe4829635aa7d696ea7c9608079b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 17:43:17.000000000","message":"Patch Set 12: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d345cff2cfd3426e6eea389e36220fa141ee7e4d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 18:04:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3c37094e11ce22dd57b9960abbe9cda1554490db","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-09-11 18:52:17.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"eea88afac11d140daf199815b36933aeafd7d3e7","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-11 19:13:25.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"de644528ee75feffc11b42020874bf8cc83cfc65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 01:26:54.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/506d528b62ad4c3ca0aabf70f9dbeb4e : SUCCESS in 49m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b742a735257d447d94ce042bee87a97e : TIMED_OUT in 1h 00m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adf1abcee8641f19ecb75acd4d29573 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3935c5d914946dba1ee91569877e53a : FAILURE in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/306725ce67254f97b2281c0b99fa0de7 : SUCCESS in 44m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d85f5275a3644358fd6f09cd1236aa7 : SUCCESS in 31m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d1cab9e8f943ccaf5b44a237518f31 : SUCCESS in 12m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b2da96754dc647fe84341e098f6e2c82 : SUCCESS in 1h 47m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/63000f5ffb9b48808fb567878e76f519 : FAILURE in 41m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac74d10ec24545ee8f8f8964308d9d67 : SUCCESS in 1h 01m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7914f10c7ec4d1c9480cda02ce9c010 : SUCCESS in 1h 30m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5188974ccd4647949e58f931af920815 : SUCCESS in 12m 17s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/94540f9e47e347c2b9eadd13835b8028 : SUCCESS in 30m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a84860e4d5464422bb28091666e060be : SUCCESS in 34m 06s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c20e80095576407a940211ea6020ff5d : SUCCESS in 34m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3d60b6cfd83a4c898d9597d4820920a0 : SUCCESS in 38m 25s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/58925a90a2c1435687f834a2541b6f4f : SUCCESS in 15m 17s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/650c5222fd30401ab1038764ac57c388 : SUCCESS in 1h 42m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0dd58aa8aa140a0a24183715056ee3d : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":13},{"id":"ed1eb0bd9fa1ea4ba46ba66090ca8e919cce73f8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-12 06:02:16.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"36da4d3344414b9c32eb5a609dc64638abe72ca8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:36:17.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c4911e7006c4cd5bb4bfab82a562fb1 : SUCCESS in 55m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abb7b693bc2a4a059a6265d660e0e415 : SUCCESS in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8937f1beb74edd9f20c112450b0771 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/192641775a9248eaa2392cd2ed7d1ea1 : SUCCESS in 38m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/751ebb6fe51d4f30b35dcf67b01d9bbb : SUCCESS in 34m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e74f129325e4aebaff42d99420f3a18 : SUCCESS in 31m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1873e9d83154ac890db2db49e01f9fc : SUCCESS in 13m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f936fa55e0884a8dadbcb476099543e7 : SUCCESS in 1h 45m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91834bd7074049818abb99eb285b8890 : SUCCESS in 1h 05m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3699a377ac4e4d50848c9159963c5366 : SUCCESS in 1h 05m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbf6b74585214f499b40995604909a2d : SUCCESS in 1h 22m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/358636b851254362bafffd5b43a1ac0a : SUCCESS in 13m 33s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fa66b3cccfc4d29a63f236b32a3238a : POST_FAILURE in 59m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/17a6a5fd3db54ba7abec5e16b8e268ad : SUCCESS in 30m 14s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c0e19be7281a41a99eb62b5b715c7a09 : POST_FAILURE in 1h 02m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5f694b61e01d4c2b9314e729323d9a2b : FAILURE in 54m 42s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/19488497288e425182c6ebf01236522f : SUCCESS in 23m 29s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/28b9ac11c6904cc8a5ab47ee5d58febe : SUCCESS in 1h 32m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc0ec41290dc4f19937436f1acb64056 : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":13},{"id":"56d60c723c417b459abba1ff67d767f410ca64df","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-12 12:33:52.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"f88494843c52af3e680f89463220e907bfd33618","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 01:58:08.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97e3d7f62c6c4c3ca82c197b744899b5 : SUCCESS in 38m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ad437762bdc43588cb715eeaf98a880 : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e02542011b1484a976c8dd1d05da0d2 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd0c4db306584111b2f0bbd96103a4ee : SUCCESS in 37m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/838c67ca45f84dd09d8ab06157a652fa : SUCCESS in 37m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55795e0c82c14c5e9b981b3c9ac5fb97 : SUCCESS in 27m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c26a3ca7ffb413d95071c2efd7159e5 : SUCCESS in 13m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f36b89190f884481a7cc81c0f8c9e5cf : FAILURE in 1h 26m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f04f10c093b14c679f73f828993a2ca2 : SUCCESS in 1h 02m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5dcfc10c4dee42be9590cfabb92209f4 : SUCCESS in 1h 01m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4f832b86c9c4e66a6552d33337b5d30 : SUCCESS in 1h 46m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2bfdc9dcc474ebdb6d68b50a1545a28 : SUCCESS in 12m 52s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c639b2ed9c74af1adcf8a4c99502e7d : SUCCESS in 34m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1e515e899dbc41d2aff6699bffe53264 : SUCCESS in 34m 00s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/da7b4f1776b74913a9e5635e866562b1 : SUCCESS in 43m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d507f65f236a446cb1e15a178970f1de : SUCCESS in 38m 16s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/65ef7d410f62462b8d466740b3b3e975 : SUCCESS in 14m 45s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/78660bc158c841f9a5fc35b212819c61 : SUCCESS in 2h 08m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e158a8e6f344e0ea1e513312abfc12f : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":13},{"id":"e1757feaf268296b0884344d48e6c0539d3abd87","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-13 01:59:31.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"b270929d067308ff72f0b18e4cc6a7d4bd7049b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:21:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3be2e56e198491d91b11dc94c2959ad : SUCCESS in 43m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3a10403d5564fc4bfe2abb7fd2b5986 : SUCCESS in 39m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3744491f61f948e6a063592dfbed7495 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6040bec76dd0466883fd6646c4dc57db : SUCCESS in 35m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/188169c41d7c4237af5028c1ccf68fef : SUCCESS in 29m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc9a501724d34c44a96150e86620ed11 : SUCCESS in 39m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff14ceabbebb457d868fd74f8fec9652 : SUCCESS in 13m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04f4b84f08ea408e999450134e9e0b63 : SUCCESS in 1h 49m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e22342c93dc445a89ec0c56c32073a35 : SUCCESS in 1h 00m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c0621305d27d469bb0998f14ce8f1148 : SUCCESS in 1h 00m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10b5af9afede448baf258101e6ce7165 : SUCCESS in 1h 28m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e96fa87747aa4df79bbeeac40c0f41f6 : SUCCESS in 12m 48s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/46f363e7210a44fb8120ca08ad32cd86 : SUCCESS in 35m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fed716f4c828490c8b97f2053f56e3d2 : SUCCESS in 39m 05s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ca5deca75806418a946d230ca13f862f : SUCCESS in 38m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1a29820d10cc48d4b008f92fca0b3918 : SUCCESS in 39m 07s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3d3aeefc3ef6451ba05c096c8973d56f : SUCCESS in 20m 31s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/718e30a340d84c8ba1b867cebe4aeabb : SUCCESS in 1h 48m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d768474e65f4806a2d7a8e32cc9b4bb : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":13},{"id":"0b1eb676ee9cbbd807f69c2e0f857e79deb589d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 23:29:48.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"02af1e03f6dc7c9169c9e3de63a8cd0adae244c3","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-13 23:58:00.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"bd1feac16a233978ddcaa77b8202c3e4b7edf456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:23:08.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83786420ae8444898821226ece71dbab : SUCCESS in 1h 01m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bdd83ef8ebb4a50ba6d996314ffe11c : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a3038d8a1284447bb2ff81ac4c0e497 : SUCCESS in 1h 01m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74d4275340de40f092595e744d1b82d4 : SUCCESS in 1h 28m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e765f56aae1644f19c5e14f1a4d05fa8 : TIMED_OUT in 1h 30m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65426050a0034861a7ca97eb96e5abc1 : SUCCESS in 13m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b9124776da942d792806df1b8885fad : SUCCESS in 1h 45m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/01b121da6a4c4d33a30a83f03962659d : SUCCESS in 57m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9211685f68cb4498947b997801c4a7e4 : SUCCESS in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e95279535c924fabbfb299354dc9feb9 : SUCCESS in 1h 37m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c318bc0b204343c0b7149f413fbe8ec2 : SUCCESS in 12m 48s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/43ba4e66acac4c97a325ffd2cc05a0ee : SUCCESS in 39m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/11246586df1b468095644cf16cb786c4 : SUCCESS in 41m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e79446a6d5e4c2786dadd111a451235 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":13},{"id":"dcf283b9ecac7e279aca1307c344f2095bca6ce6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-14 01:24:38.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"05a3d53cb1993fe3541445c23f28f0b46c10df5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:46:40.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7325cc8052f74fa1bca895d4b281fa8e : SUCCESS in 53m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e0f7a321a0c48d897eb5188ac118db2 : SUCCESS in 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68281ed97a64cc48fe515bd805eaae5 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2417ce8752149c88bab811aa684bec4 : SUCCESS in 41m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff3304cacd474519932d13c217cddff2 : SUCCESS in 39m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/377486d5944348ca8861c6e8d6e04c7a : SUCCESS in 31m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42ed312b17b4e5883666add704f87bb : SUCCESS in 12m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db9722fccc1d4d5fa2cdd87d09e75398 : SUCCESS in 1h 55m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ade4fe8f1f249f6b95ef035738f277e : SUCCESS in 59m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c8f08c5981a413c924f66d0358bd1f7 : SUCCESS in 52m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c9f52fd0de84e6889cbca4927507cac : SUCCESS in 1h 19m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9715e84b40004d1cb46d71adcfcf9417 : SUCCESS in 12m 23s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/993c02183c7f4389b0c961b0d6cb9432 : SUCCESS in 33m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/41718b0c9bb14c4893edb47531df41c5 : SUCCESS in 33m 20s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/de325b9f2f0041d680f5e2af35f8e33f : SUCCESS in 39m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/37ecd892060e43ea8aecde5063d5fd33 : SUCCESS in 40m 49s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1348bf6744964841aaaa578d3c7a3d37 : SUCCESS in 19m 11s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9257c334350e444da8e92b11427ea9fe : SUCCESS in 1h 52m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78b3a21f64b24d7e8db5bc74ad9603c4 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":13},{"id":"57f6f6c8780ac9e49b2780037316322a75e35c7f","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 08:58:24.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"92ecb31ad1a76666251ff20c2ba7a006cb7fef4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 09:47:14.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"bbf2ee86e33d7cb82e8cb0196d293e1b6d081b86","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 11:52:29.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"ad7bcf7f79a1a0ec47111f27eb928e23aeb51129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:18:51.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/281e96f19791465bb0032ea8d3304b31 : SUCCESS in 45m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18897ff35e0443d1838d8cfdd2d6c999 : SUCCESS in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efffe85bc08d43628e459235bfc672e6 : SUCCESS in 42m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/430980dd3fc642bab79eaf4257c184e0 : SUCCESS in 47m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3c16e572f1147de96ada18056dadb20 : SUCCESS in 58m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42522c2404fc4bc68212e75fb9308cc0 : SUCCESS in 14m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e36406342ca842fea7c24624d022101a : SUCCESS in 1h 31m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4f106df30ea4837a38d67acb3d46b02 : SUCCESS in 57m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ff9826d8c72469c9770fd2e8abe8c5d : SUCCESS in 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d39047975ad4f1a945393f0c7abb2b5 : SUCCESS in 1h 15m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83051b3aa3b64eaa8c2e66a0a90b0b17 : SUCCESS in 11m 09s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a63681c8e9740139feb731e33878e01 : SUCCESS in 31m 37s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/14ff30f156404ebf9747cdff77ebdd88 : SUCCESS in 34m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/195584c3723c4fd7a6720707862a7acb : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":13},{"id":"cb6bd760c74d34c7292ae55785e6fa3e24081c8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:18:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"d12189ff60fca9c3bac763eea9a8f320f1582b33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:30:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58e9d964e2544958a7f63da428491de : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ac659ca0412416bab40649c75f0ca62 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"5e06ec8163249a7581b12db321036979b7d0604c","revisions":{"bf6c6a002c5faf24961b04eab3ea9d95e35e5065":{"kind":"REWORK","_number":1,"created":"2019-08-09 03:26:33.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/1"}}},"commit":{"parents":[{"commit":"a44acf5e31aef681716ee8492d5b0413aed14ec3","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a44acf5e31aef681716ee8492d5b0413aed14ec3"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:25:35.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bf6c6a002c5faf24961b04eab3ea9d95e35e5065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bf6c6a002c5faf24961b04eab3ea9d95e35e5065"}]},"branch":"refs/heads/master"},"01870209b372dde9c3e06f19ada3a4bbe2574a7e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-09 18:02:57.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/2"}}},"commit":{"parents":[{"commit":"a40c1c090637bd05309e103a77a26a06ff12228c","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a40c1c090637bd05309e103a77a26a06ff12228c"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 18:02:52.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01870209b372dde9c3e06f19ada3a4bbe2574a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01870209b372dde9c3e06f19ada3a4bbe2574a7e"}]},"branch":"refs/heads/master"},"64d0b8c32768882575c5f3ea4a55a744b575911a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-12 01:49:37.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/3"}}},"commit":{"parents":[{"commit":"3b3f6ca25d8d335997dd0e63dfdee23a68991c20","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3b3f6ca25d8d335997dd0e63dfdee23a68991c20"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 01:49:32.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/64d0b8c32768882575c5f3ea4a55a744b575911a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/64d0b8c32768882575c5f3ea4a55a744b575911a"}]},"branch":"refs/heads/master"},"3fc20f8fba0f9a5fe451af2cca86b76f0d1fdbb2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-12 04:19:23.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/4"}}},"commit":{"parents":[{"commit":"4f94b369d34e460b499e702329c432790b95b2ad","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f94b369d34e460b499e702329c432790b95b2ad"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 04:19:02.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fc20f8fba0f9a5fe451af2cca86b76f0d1fdbb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fc20f8fba0f9a5fe451af2cca86b76f0d1fdbb2"}]},"branch":"refs/heads/master"},"b0f2ddc334264444db288cd902b930319a09c612":{"kind":"REWORK","_number":5,"created":"2019-08-12 14:18:25.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/5"}}},"commit":{"parents":[{"commit":"4f94b369d34e460b499e702329c432790b95b2ad","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f94b369d34e460b499e702329c432790b95b2ad"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 14:18:18.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b0f2ddc334264444db288cd902b930319a09c612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b0f2ddc334264444db288cd902b930319a09c612"}]},"branch":"refs/heads/master"},"b78d200c9ad4985deb3c5c195ff5a2a7c83cbb56":{"kind":"REWORK","_number":6,"created":"2019-08-13 20:46:46.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/6"}}},"commit":{"parents":[{"commit":"4f94b369d34e460b499e702329c432790b95b2ad","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f94b369d34e460b499e702329c432790b95b2ad"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-13 20:46:39.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b78d200c9ad4985deb3c5c195ff5a2a7c83cbb56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b78d200c9ad4985deb3c5c195ff5a2a7c83cbb56"}]},"branch":"refs/heads/master"},"53766306f822b29d56c99b80fb7a6dfdd841f80c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-24 19:28:47.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/09/675509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/7"}}},"commit":{"parents":[{"commit":"75ba9da5464bae226e3414eba4de46d0fbf4ea00","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/75ba9da5464bae226e3414eba4de46d0fbf4ea00"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2019-08-24 19:28:47.000000000","tz":0},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/53766306f822b29d56c99b80fb7a6dfdd841f80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/53766306f822b29d56c99b80fb7a6dfdd841f80c"}]},"branch":"refs/heads/master"},"e8a938811d4910310e206a11516f718b27f2cea2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-29 17:44:35.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/09/675509/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/8"}}},"commit":{"parents":[{"commit":"5025de43f3fff3b285dc12bf25b6381183fc7c62","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5025de43f3fff3b285dc12bf25b6381183fc7c62"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-08-29 17:44:35.000000000","tz":0},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e8a938811d4910310e206a11516f718b27f2cea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e8a938811d4910310e206a11516f718b27f2cea2"}]},"branch":"refs/heads/master"},"7f986abc4c9302c8ea5b220dd35bf28e9cc0b8d2":{"kind":"REWORK","_number":9,"created":"2019-09-09 23:44:56.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/9"}}},"commit":{"parents":[{"commit":"1ecdb401a2741ae6574aaa6f4f4997f08cedad15","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ecdb401a2741ae6574aaa6f4f4997f08cedad15"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-09 23:44:45.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7f986abc4c9302c8ea5b220dd35bf28e9cc0b8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7f986abc4c9302c8ea5b220dd35bf28e9cc0b8d2"}]},"branch":"refs/heads/master"},"f331df18bb3da717d0d172067e869a3c71d787e7":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-09-09 23:50:09.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/10"}}},"commit":{"parents":[{"commit":"1340e230053502a4e1765219ba1ba6b09900bb22","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1340e230053502a4e1765219ba1ba6b09900bb22"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-09 23:48:29.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f331df18bb3da717d0d172067e869a3c71d787e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f331df18bb3da717d0d172067e869a3c71d787e7"}]},"branch":"refs/heads/master"},"39cc47d09b2272f8fdd19ef1ae750af846389733":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-10 06:31:34.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/11"}}},"commit":{"parents":[{"commit":"d316058642231c993c321b4fea4d155f42b3e6fc","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d316058642231c993c321b4fea4d155f42b3e6fc"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-09-10 06:31:34.000000000","tz":0},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39cc47d09b2272f8fdd19ef1ae750af846389733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39cc47d09b2272f8fdd19ef1ae750af846389733"}]},"branch":"refs/heads/master"},"01c7c8035e0d10533e19a661704a5c92c557ac90":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-09-10 07:03:40.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/12"}}},"commit":{"parents":[{"commit":"a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-10 07:03:20.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01c7c8035e0d10533e19a661704a5c92c557ac90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01c7c8035e0d10533e19a661704a5c92c557ac90"}]},"branch":"refs/heads/master"},"5e06ec8163249a7581b12db321036979b7d0604c":{"kind":"REWORK","_number":13,"created":"2019-09-11 18:04:36.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/09/675509/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/675509/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/675509/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/675509/13"}}},"commit":{"parents":[{"commit":"a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae","subject":"Add --immutable-roles flag to bootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a80d83e76dfb1114b8cd1b31bd2f175d36ae18ae"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-09 03:00:28.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-11 18:04:29.000000000","tz":-420},"subject":"Add immutable roles status check","message":"Add immutable roles status check\n\nThis implements part 3 of the proposed change for immutable roles[1], as\nwell as adds a release note.\n\nPart 4 (changing the default behavior of ``keystone-manage bootstrap``\nwill have to come in the next cycle.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/immutable-resources.html#proposed-change\n\nChange-Id: Ie9d658deb1fa69e9007f3c50535b5c48a7a292d1\nPartial-bug: #1823258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e06ec8163249a7581b12db321036979b7d0604c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e06ec8163249a7581b12db321036979b7d0604c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
