)]}'
{"id":"openstack%2Fkeystone~675667","triplet_id":"openstack%2Fkeystone~master~I3d07637b0046cc88a66bcb51a0a4fe7c146c1549","project":"openstack/keystone","branch":"master","topic":"broken-links","hashtags":[],"change_id":"I3d07637b0046cc88a66bcb51a0a4fe7c146c1549","subject":"Docs: Make robust with using real links","status":"MERGED","created":"2019-08-09 16:49:37.000000000","updated":"2019-08-10 04:26:55.000000000","submitted":"2019-08-10 04:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675667-1565411215031-e4a4d76b","meta_rev_id":"8dc646b18337487bb048de478a4080344a168595","_number":675667,"virtual_id_number":675667,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-08-10 04:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-08-09 19:02:19.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-09 20:01:18.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-09 20:01:18.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 19:02:19.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-09 20:01:18.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-08-10 04:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c44c88a51968eb80462ce48f9de42cf251b693d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 16:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ad78be00a1fe18aa75259882aa0a45f96cd46bf","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-09 17:17:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"34980fccadaea9cc4967c906b8b87ebcf074c46b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 17:32:31.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7199c59f0fa5c941273a1482d6603d9f7f612e3e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 17:44:15.000000000","message":"Patch Set 1:\n\nNext version:\n* readds the support matrix\n* go back to include files - but rename files as .inc so that they only get published once","accounts_in_message":[],"_revision_number":1},{"id":"7c6d00c21af041fcc9ff1954fdd1e6e4e57cd2d8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-09 18:15:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e09cbdedb50e0da7b36d0a73051b9688ef1bd26","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-09 19:02:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14e8c224bfc12547b9ee73bac06fc4a23a6217b7","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-09 20:01:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56362743eb30331fe1084732816d54dc7e042a19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:00:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/67/675667/2/check/build-openstack-api-ref/fdac60b/html/ : SUCCESS in 5m 28s\n- openstack-tox-cover https://logs.opendev.org/67/675667/2/check/openstack-tox-cover/eff3a09/cover/ : SUCCESS in 33m 53s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/675667/2/check/openstack-tox-lower-constraints/e0b3550/ : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://logs.opendev.org/67/675667/2/check/openstack-tox-pep8/924df1a/ : SUCCESS in 5m 43s\n- openstack-tox-py27 https://logs.opendev.org/67/675667/2/check/openstack-tox-py27/1643f37/ : SUCCESS in 24m 55s\n- openstack-tox-py36 https://logs.opendev.org/67/675667/2/check/openstack-tox-py36/33e63bd/ : SUCCESS in 29m 19s\n- openstack-tox-py37 https://logs.opendev.org/67/675667/2/check/openstack-tox-py37/0d75d3e/ : SUCCESS in 29m 25s\n- openstack-tox-docs https://logs.opendev.org/67/675667/2/check/openstack-tox-docs/fdba206/html/ : SUCCESS in 7m 10s\n- tempest-full https://logs.opendev.org/67/675667/2/check/tempest-full/4217714/ : SUCCESS in 1h 29m 24s\n- neutron-grenade https://logs.opendev.org/67/675667/2/check/neutron-grenade/6f86afc/ : SUCCESS in 1h 04m 55s\n- grenade-py3 https://logs.opendev.org/67/675667/2/check/grenade-py3/57d4ed4/ : SUCCESS in 1h 10m 48s\n- tempest-full-py3 https://logs.opendev.org/67/675667/2/check/tempest-full-py3/a82e158/ : SUCCESS in 1h 23m 27s\n- build-openstack-releasenotes https://logs.opendev.org/67/675667/2/check/build-openstack-releasenotes/6d80f92/html/ : SUCCESS in 11m 23s\n- keystone-dsvm-functional https://logs.opendev.org/67/675667/2/check/keystone-dsvm-functional/6f2bf71/ : SUCCESS in 31m 10s\n- keystone-dsvm-py3-functional https://logs.opendev.org/67/675667/2/check/keystone-dsvm-py3-functional/02127f6/ : SUCCESS in 31m 20s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/67/675667/2/check/keystone-dsvm-functional-federation-opensuse15/0470b74/ : SUCCESS in 33m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/67/675667/2/check/keystone-dsvm-py3-functional-federation-opensuse15/191adfb/ : SUCCESS in 32m 48s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/67/675667/2/check/keystoneclient-devstack-functional/c8e64ae/ : SUCCESS in 17m 34s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/67/675667/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/5c50a2d/ : SUCCESS in 2h 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69ead500dbe10ec4d906fedf42ddc4f74be30117","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:00:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8d8760aeef60e507bdb0a72e47927adb5a2da29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 04:26:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/67/675667/2/gate/build-openstack-api-ref/336c37d/html/ : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/675667/2/gate/openstack-tox-lower-constraints/54c0ee2/ : SUCCESS in 41m 02s\n- openstack-tox-pep8 https://logs.opendev.org/67/675667/2/gate/openstack-tox-pep8/3464ba7/ : SUCCESS in 7m 07s\n- openstack-tox-py27 https://logs.opendev.org/67/675667/2/gate/openstack-tox-py27/241248d/ : SUCCESS in 30m 46s\n- openstack-tox-py36 https://logs.opendev.org/67/675667/2/gate/openstack-tox-py36/369bc74/ : SUCCESS in 28m 55s\n- openstack-tox-py37 https://logs.opendev.org/67/675667/2/gate/openstack-tox-py37/ea5a7dd/ : SUCCESS in 30m 07s\n- openstack-tox-docs https://logs.opendev.org/67/675667/2/gate/openstack-tox-docs/943bf18/html/ : SUCCESS in 8m 44s\n- tempest-full https://logs.opendev.org/67/675667/2/gate/tempest-full/8cfcbc9/ : SUCCESS in 1h 31m 42s\n- neutron-grenade https://logs.opendev.org/67/675667/2/gate/neutron-grenade/b1e97a6/ : SUCCESS in 1h 05m 53s\n- grenade-py3 https://logs.opendev.org/67/675667/2/gate/grenade-py3/5379ed3/ : SUCCESS in 1h 03m 40s\n- tempest-full-py3 https://logs.opendev.org/67/675667/2/gate/tempest-full-py3/e6c49f4/ : SUCCESS in 1h 18m 31s\n- build-openstack-releasenotes https://logs.opendev.org/67/675667/2/gate/build-openstack-releasenotes/fd2fb0b/html/ : SUCCESS in 10m 50s\n- keystone-dsvm-functional https://logs.opendev.org/67/675667/2/gate/keystone-dsvm-functional/ac555f7/ : SUCCESS in 32m 13s\n- keystone-dsvm-py3-functional https://logs.opendev.org/67/675667/2/gate/keystone-dsvm-py3-functional/f272ece/ : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":2},{"id":"afe1644f97150164710878dc839dac933b0bd448","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 04:26:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f10f95b455806ce28b112827a794618957e1f829","revisions":{"3005ecdc0c9e0b4344fa96c54a363e81c6052126":{"kind":"REWORK","_number":1,"created":"2019-08-09 16:49:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/675667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/67/675667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/67/675667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/67/675667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/67/675667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/67/675667/1"}}},"commit":{"parents":[{"commit":"f850095710071d1998bc7dabeeb7135dd75080bd","subject":"Merge \"Fix missing print format and missing ws between words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f850095710071d1998bc7dabeeb7135dd75080bd"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-09 14:18:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-09 16:46:01.000000000","tz":120},"subject":"Docs: Make robust with using real links","message":"Docs: Make robust with using real links\n\nOur tools noticed that keystone links to\nhttps://docs.openstack.org/keystone/latest/admin/identity-domain-specific-config.html\nwhich does not exist anymore.\n\nThe page was removed but the link to it was not changed. Replace this\nand similar links with internal links that will work even if files are\nmoved - and can be verified, thus sphinx will error in case of broken\ntargets.\n\nThese changes include a few other fixes for broken keystone links, e.g.\nto renamed anchors.\n\nDo not use includes in admin/configuration.rst and\nadmin/federation/configure_federation.rst - the files were\npublished twice (as separate files and on this page) and thus\nreferencing failed. Use a normal toctree instead.\n\nChange-Id: I3d07637b0046cc88a66bcb51a0a4fe7c146c1549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3005ecdc0c9e0b4344fa96c54a363e81c6052126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3005ecdc0c9e0b4344fa96c54a363e81c6052126"}]},"branch":"refs/heads/master"},"f10f95b455806ce28b112827a794618957e1f829":{"kind":"REWORK","_number":2,"created":"2019-08-09 18:15:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/675667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/67/675667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/67/675667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/67/675667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/67/675667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/67/675667/2"}}},"commit":{"parents":[{"commit":"f850095710071d1998bc7dabeeb7135dd75080bd","subject":"Merge \"Fix missing print format and missing ws between words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f850095710071d1998bc7dabeeb7135dd75080bd"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-09 14:18:51.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-09 18:15:14.000000000","tz":120},"subject":"Docs: Make robust with using real links","message":"Docs: Make robust with using real links\n\nOur tools noticed that keystone links to\nhttps://docs.openstack.org/keystone/latest/admin/identity-domain-specific-config.html\nwhich does not exist anymore.\n\nThe page was removed but the link to it was not changed. Replace this\nand similar links with internal links that will work even if files are\nmoved - and can be verified, thus sphinx will error in case of broken\ntargets.\n\nThese changes include a few other fixes for broken keystone links, e.g.\nto renamed anchors.\n\nFor the include files in admin/configuration.rst and\nadmin/federation/configure_federation.rst: Rename them to *inc.\nThe files were\npublished twice (as separate files and on this page) and thus\nreferencing failed. Renaming avoids this.\n\nAlso, put doctree outside of html tree so that it does not get\npublished.\n\nChange-Id: I3d07637b0046cc88a66bcb51a0a4fe7c146c1549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f10f95b455806ce28b112827a794618957e1f829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f10f95b455806ce28b112827a794618957e1f829"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
