)]}'
{"id":"openstack%2Fkeystone~675720","triplet_id":"openstack%2Fkeystone~master~I6a5eca8240aa905e02fbf9bec335996c3a4f1c79","project":"openstack/keystone","branch":"master","topic":"trust-policies","hashtags":[],"change_id":"I6a5eca8240aa905e02fbf9bec335996c3a4f1c79","subject":"Add protection tests for trusts API","status":"MERGED","created":"2019-08-10 04:42:36.000000000","updated":"2019-08-19 04:45:10.000000000","submitted":"2019-08-19 04:45:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675720-1566189910720-ed945153","meta_rev_id":"01833f8d0055c12a812917c02293576be4a4d71d","_number":675720,"virtual_id_number":675720,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-17 06:56:22.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-08-19 04:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-08-17 13:05:47.000000000","_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-19 02:53:45.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-19 02:53:45.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 17:05:41.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-08-17 13:05:47.000000000","updated_by":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"reviewer":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"state":"REVIEWER"},{"updated":"2019-08-19 02:53:45.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-08-19 04:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"532bf6ec8eb05c518ffcaf0d25d1689312eb745f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-10 04:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b3518d983b4a693e0af4bf90c00a2ac517e475e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 06:26:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/20/675720/1/check/openstack-tox-cover/d56e507/cover/ : SUCCESS in 34m 25s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/675720/1/check/openstack-tox-lower-constraints/9cb0ce7/ : SUCCESS in 35m 37s\n- openstack-tox-pep8 https://logs.opendev.org/20/675720/1/check/openstack-tox-pep8/12ec790/ : SUCCESS in 5m 10s\n- openstack-tox-py27 https://logs.opendev.org/20/675720/1/check/openstack-tox-py27/8b93241/ : SUCCESS in 27m 29s\n- openstack-tox-py36 https://logs.opendev.org/20/675720/1/check/openstack-tox-py36/9738ca8/ : SUCCESS in 29m 26s\n- openstack-tox-py37 https://logs.opendev.org/20/675720/1/check/openstack-tox-py37/c1e82d3/ : SUCCESS in 30m 39s\n- openstack-tox-docs https://logs.opendev.org/20/675720/1/check/openstack-tox-docs/6280991/html/ : SUCCESS in 6m 56s\n- tempest-full https://logs.opendev.org/20/675720/1/check/tempest-full/cc1e08e/ : SUCCESS in 1h 38m 00s\n- neutron-grenade https://logs.opendev.org/20/675720/1/check/neutron-grenade/0637907/ : SUCCESS in 56m 18s\n- grenade-py3 https://logs.opendev.org/20/675720/1/check/grenade-py3/68bd864/ : SUCCESS in 1h 06m 08s\n- tempest-full-py3 https://logs.opendev.org/20/675720/1/check/tempest-full-py3/2d55dd3/ : SUCCESS in 1h 27m 13s\n- keystone-dsvm-functional https://logs.opendev.org/20/675720/1/check/keystone-dsvm-functional/6a5ac27/ : SUCCESS in 32m 12s\n- keystone-dsvm-py3-functional https://logs.opendev.org/20/675720/1/check/keystone-dsvm-py3-functional/5cb05d6/ : SUCCESS in 30m 36s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/20/675720/1/check/keystone-dsvm-functional-federation-opensuse15/3d3fb03/ : SUCCESS in 34m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/20/675720/1/check/keystone-dsvm-py3-functional-federation-opensuse15/a8796d6/ : SUCCESS in 43m 55s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/20/675720/1/check/keystoneclient-devstack-functional/3131525/ : SUCCESS in 15m 42s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/20/675720/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/5513e42/ : SUCCESS in 1h 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cf4ac7aec588527e474b3b1b8b95a07c1e55809","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 02:31:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f4de2fecae18ffac37332b0911915686e05a6422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 04:17:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/20/675720/2/check/openstack-tox-cover/d23a33b/cover/ : SUCCESS in 34m 45s\n- openstack-tox-lower-constraints https://logs.opendev.org/20/675720/2/check/openstack-tox-lower-constraints/5c209ed/ : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://logs.opendev.org/20/675720/2/check/openstack-tox-pep8/fde6963/ : SUCCESS in 5m 27s\n- openstack-tox-py27 https://logs.opendev.org/20/675720/2/check/openstack-tox-py27/f8d6bec/ : SUCCESS in 34m 27s\n- openstack-tox-py36 https://logs.opendev.org/20/675720/2/check/openstack-tox-py36/3d18846/ : SUCCESS in 30m 27s\n- openstack-tox-py37 https://logs.opendev.org/20/675720/2/check/openstack-tox-py37/2184c1e/ : SUCCESS in 30m 08s\n- openstack-tox-docs https://logs.opendev.org/20/675720/2/check/openstack-tox-docs/942a1fe/html/ : SUCCESS in 7m 09s\n- tempest-full https://logs.opendev.org/20/675720/2/check/tempest-full/cb011d4/ : SUCCESS in 1h 40m 03s\n- neutron-grenade https://logs.opendev.org/20/675720/2/check/neutron-grenade/379d28b/ : SUCCESS in 46m 46s\n- grenade-py3 https://logs.opendev.org/20/675720/2/check/grenade-py3/f52acde/ : SUCCESS in 56m 43s\n- tempest-full-py3 https://logs.opendev.org/20/675720/2/check/tempest-full-py3/11d959b/ : SUCCESS in 1h 20m 09s\n- keystone-dsvm-functional https://logs.opendev.org/20/675720/2/check/keystone-dsvm-functional/941d80e/ : SUCCESS in 29m 14s\n- keystone-dsvm-py3-functional https://logs.opendev.org/20/675720/2/check/keystone-dsvm-py3-functional/c8b5acc/ : SUCCESS in 28m 56s\n- keystone-dsvm-functional-federation-opensuse15 https://logs.opendev.org/20/675720/2/check/keystone-dsvm-functional-federation-opensuse15/a30be56/ : SUCCESS in 31m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://logs.opendev.org/20/675720/2/check/keystone-dsvm-py3-functional-federation-opensuse15/ecaf6d6/ : SUCCESS in 33m 47s (non-voting)\n- keystoneclient-devstack-functional https://logs.opendev.org/20/675720/2/check/keystoneclient-devstack-functional/404a484/ : SUCCESS in 13m 45s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://logs.opendev.org/20/675720/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/59bff8e/ : SUCCESS in 1h 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfafeca3e3106ef45881e127066231ceff269e03","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-12 22:58:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"423c2d9c8707118b6aaeadd5f978b03da8f18f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:41:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71b96b9a673846b5a0a2f189425f994c : SUCCESS in 32m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ff040590c2423e8ae2313895cc0616 : SUCCESS in 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bc803a2273f4deb87dfb64bcdf46630 : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cd0334e20b04a818a2490996d85e9b5 : SUCCESS in 30m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18e6756ec6a846f086d5a4b6ba72d092 : SUCCESS in 28m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbcd55bc4ab84106b9be327c1034cdbc : SUCCESS in 23m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6dc56c889e94d39b47a84bf910a3f47 : SUCCESS in 7m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eb6b90b5b11c402a9978bd12d62169b6 : SUCCESS in 1h 36m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5c11cb42db564007a185695d8ff618bf : SUCCESS in 1h 02m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/314cc2f7771b4c5a9c4b951e227beed9 : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0de53293f5ca4afca0a0f92b63fcd357 : SUCCESS in 1h 26m 29s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d04db16ed4b4413fba16eb30036c3ac0 : SUCCESS in 31m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/230a0881e2d34ce2b63e2c57469c61e8 : SUCCESS in 36m 18s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0dee4494f9c840f3b131df6dcf247ce8 : SUCCESS in 30m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/45a4310db64d415e91fabc6ecd8c76bc : SUCCESS in 38m 24s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3f070ba2820d4139989808b9b24b1bdd : SUCCESS in 18m 48s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/21514e69a18d4f74ab530deb0ef06861 : SUCCESS in 1h 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a4c8c01cc1fff20b42abb4b487c83db46c5d78da","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-13 23:13:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d0532a7e6f6ba4c63af695c6fa81843eac19b7d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 00:45:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e8ede6e03a4ce195f40dab31c167cc : SUCCESS in 32m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52ba139d96324bf8813d72cd7358cc9a : SUCCESS in 32m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc205abd1dd4b4ca73e4b709dbcbf2a : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f361990dc474ee1a5941c12279548ba : SUCCESS in 24m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0662718bd64b7aa175fe2bebd82b89 : SUCCESS in 23m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55bde28af98b4a0b98c8e1ea636d25ff : SUCCESS in 22m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fea29ca8248471fb3bae5b0138f0a79 : SUCCESS in 7m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/525e089ceaee4b8a945f867c0c0270d6 : SUCCESS in 1h 25m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/42470b6b2e1d451aa697a5b43434c08f : SUCCESS in 54m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e8a328aa09cc4a43b68b3a10fe83f5dd : SUCCESS in 56m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c08cf64258974a63b44df002dec6f583 : SUCCESS in 1h 26m 58s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/76c1666f759d49e991897b76a694d56b : SUCCESS in 34m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b8df06e0029a40fa8a749fe913a3b902 : SUCCESS in 32m 26s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/4617825a0adb4fe88b760db989372eee : SUCCESS in 35m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/693f2191954942f9b738f5bc2e5b7e82 : SUCCESS in 36m 32s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d82fff3d22ee44ca8e3712ee7bbbd70c : SUCCESS in 17m 08s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9c27c261d84d494480c90b586a87c819 : SUCCESS in 1h 16m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9cf04d1fe94b019c447e5b19798a7c : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":4},{"id":"87190181666dbbb36b76492b51a320c9e8407986","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-14 22:43:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"14d253ae11ac0cf06512d8b244507b021382809c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:22:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcf6efd0ae6342e9ad1aeba60743a791 : SUCCESS in 31m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b5099bd286f4d11a0ce4fd9a7910728 : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf6af5f22b841b7bdcff0252fe2fe74 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3ac1c0eb6524fb08033e2ce2a28121e : SUCCESS in 37m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3f39df379394b7a93bbd7013e6f45c8 : SUCCESS in 30m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09a4e0e415684590bc50fe7033afde97 : SUCCESS in 27m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a613e79a3b4d94b64e577db5638da8 : SUCCESS in 7m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/599682e09c9a471da2855f4d46df54f1 : FAILURE in 1h 21m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/319a4779a6ba4fc38334513e2f5bca8d : FAILURE in 55m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1adf6634e0e64d488dd609d97eecf440 : FAILURE in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdd458c60ef04c88870c0d1701bed5e4 : FAILURE in 1h 03m 28s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4e2585751384b84b5aefccbfe919496 : SUCCESS in 31m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a7e3195a6d7f43e8a54b45489e52534e : SUCCESS in 29m 09s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/b6e8efb7b46c481c882a94cf5dcd6e20 : SUCCESS in 37m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5a3f8d53f8094635af7c1ad7d9f2a604 : SUCCESS in 39m 29s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/72bb3f93299249d5b037a8a6fcd59724 : SUCCESS in 17m 05s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a8fadcc758094b90b5d71b99187b46dd : FAILURE in 1h 09m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d89bea3879a1431b9b31b40b50dd3351 : FAILURE in 1h 04m 42s","accounts_in_message":[],"_revision_number":5},{"id":"2c6af6f55d3ed07ffbae5c6984f095ed9c19e610","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-16 00:29:56.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"75b63957a20a9fa4c2c21679f086e1859c6c6793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 02:10:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b77ae17daf2a444fae2aff83c4a9c2a3 : SUCCESS in 37m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dcbd388b51043d89e5671179b56bf5f : SUCCESS in 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626dd9ffd7494db9a6ede42c48d6419e : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba8814201d604ae3b02c6be6a5c05a69 : SUCCESS in 34m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e663beef6b648349c80e783d1792f30 : SUCCESS in 28m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61341c9be24d4eafb9e70f714e2e1b6c : SUCCESS in 27m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a12628e7e94e059e1a46a104f87be8 : SUCCESS in 10m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e94d7c148924a9cb6cba3b14cc63186 : SUCCESS in 1h 31m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b849642c40364c9c834801c51473e8d8 : SUCCESS in 1h 04m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/816e3509c26849b481b4a88d535db8fb : SUCCESS in 1h 07m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67f96d8fd83f4defbf1efc996a9e56c1 : SUCCESS in 1h 21m 34s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc8eb184a8214bfe8ed256b98571d485 : SUCCESS in 33m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8112be98d3bc4f169cf1cebe1a0e8d13 : SUCCESS in 37m 42s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/4ec8eb49d3b54d378c85a9342133af16 : SUCCESS in 40m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d907caad7237437d9f2e590345eb92d0 : SUCCESS in 42m 46s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ed63e29dc91c4bde8d6fab82006712a5 : SUCCESS in 16m 47s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/84dbbe419f8147dbad0fe82993470b69 : SUCCESS in 1h 30m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/155ae77c4f564bd39e37774e2b4767c9 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":6},{"id":"c58817599b95802bad5bd9b704005fad6e3ef559","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-16 17:05:41.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"367e630cde144556b0159e49f4dd43d9ec5aba78","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-16 22:22:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"294036b54a2a10fe689d270308b194b56165bf41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 00:24:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b9a89e04d643ab98eee52967d0b16f : SUCCESS in 42m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19032214a47c45b2a23d8c4b44acbee8 : TIMED_OUT in 41m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bcffc6e43be4ac7af11613eb0b6ecc1 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89b24f140a064eeaaa7907dd3d22d281 : TIMED_OUT in 41m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb9b162558d349419ac45eb09f4f32b9 : TIMED_OUT in 42m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e0a2687200647bb88c6bc51d68c67b4 : TIMED_OUT in 41m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2938706809c14e18bc1ee01ed95ca79e : SUCCESS in 10m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3bca3e0b72274a6fb0ee6bb6739c45cb : SUCCESS in 1h 22m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1eb70bb46c2743ed8267bb4fdcad3148 : SUCCESS in 57m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b2f335d4c2fc43c481cfba0690060117 : SUCCESS in 1h 07m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a233a6f19b024d9aa1a6d78b83217d0e : SUCCESS in 1h 29m 20s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9af7a59e074490ab0f03b51e9ea6f32 : SUCCESS in 34m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d2b67a2463ef420c8f959e3684ec21c6 : SUCCESS in 35m 03s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/391a306f85a141b2acc80851f7f3691b : SUCCESS in 38m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6053f0d276b94d7894f3fce4a9310d70 : SUCCESS in 40m 17s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/68259f6f1c4a440b854dc522252c5d94 : SUCCESS in 13m 29s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c6278b98155b430989b99d88c2c828bb : SUCCESS in 1h 36m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc8cc4c3c7c54363ac745d73c14fb321 : SUCCESS in 1h 13m 03s","accounts_in_message":[],"_revision_number":7},{"id":"1b5c4b0587fa2feb2a41412e8ef152bd2e230a81","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-17 02:40:40.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"010cd0237b7a0c2c544d2f080f4eff96b52e6747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 04:14:27.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad8d96b521814f83b9245f4b6a2e11ce : SUCCESS in 34m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03884f3dbe874060a1a65f8d838087b8 : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e567de3f63041ffbcc2b3b6ba6d38be : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b185aef4f9e435ba03b34e5124cd0af : SUCCESS in 38m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca48e39312a24603a706149a71c57f77 : TIMED_OUT in 42m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6660b10fed84734a97b6dac2d6a5040 : SUCCESS in 38m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda72e4664d945e2bfc5bd43278615a4 : SUCCESS in 10m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f063101ff9a44183af0169112108e001 : SUCCESS in 1h 30m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fd6114bfbe244a758b6855bb938e96f3 : SUCCESS in 56m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a3ffcbd57e5b48948a40002221f4c894 : SUCCESS in 58m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0f2f17afabe4f828a54e567dd1380f3 : SUCCESS in 1h 17m 59s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a942e07e18142df90a71f160c631a22 : SUCCESS in 29m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8f98656dd9294602bf6419af1e9140d6 : SUCCESS in 29m 39s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2364c8b0aae144ef8b6b9ffa7964887d : SUCCESS in 30m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/a6facd17796b40f8b57a1656360773c7 : SUCCESS in 41m 25s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9c7f247f7cc14762a3652ac176a553b7 : SUCCESS in 16m 53s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/20230a5bf4b141dd94b2317fc033df68 : FAILURE in 1h 26m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d4c70d23534146b727950a8842ac2d : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":7},{"id":"ef6970d88f047bd5af79d731c772c56551c387c8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-17 06:56:22.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"af329ffa9bb176d405de2085413e1e051ab71829","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 08:37:26.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a68b3f7999404804b448a1f25e7b2d90 : SUCCESS in 39m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f1cdf7254cb4d9586b78154fc5486d1 : TIMED_OUT in 41m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d165c3efb2640138a1f5d0e8c1c6d40 : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f60126dbf6ed4575b4c97750a2d2c6b8 : TIMED_OUT in 41m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d09295b2f3f482da93a02a9b09a38e1 : SUCCESS in 37m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac94412528134c8291bffabd07c3c569 : SUCCESS in 30m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ccd0b14dad44ec94354e582ec75405 : SUCCESS in 10m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ddb2c4070edb492899270feb22c02320 : SUCCESS in 1h 37m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83534527ffb947b49f9f4f5173df63e6 : SUCCESS in 58m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f8681a13ce6f49edaa4bc6c42494cf56 : SUCCESS in 1h 11m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ac0725cdf164a2692b1f5ab922e604f : SUCCESS in 1h 29m 00s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/41c3e54551bd4d82b59ceec34b14f5a7 : SUCCESS in 34m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a0d651f8e63b4ef4ac85eaa79f8ccf5d : SUCCESS in 44m 53s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d58830ff974b4a808c64cbe556f461a5 : SUCCESS in 37m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fa29406ac8034b719a0fdeefd5105b48 : SUCCESS in 42m 03s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fb5af78ac1164deaba4d771125e2421d : SUCCESS in 20m 05s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/44dffd0712f94168ae3574653de210bf : SUCCESS in 1h 39m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06070c16e8614e929a4ef6ffa6160309 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":7},{"id":"c601955619ce8bd00764201c1a1c2300aa548272","author":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"date":"2019-08-17 13:05:47.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"5949aabc2b3cecdb07e738992864c23b6d679184","author":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"date":"2019-08-17 13:06:06.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"26ced28e4b9119ad7f6eee92a757693cf57e201d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 14:55:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21f73f09993448c98aeb4ad606fea0dd : SUCCESS in 36m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83c46ec9e136437594393b1e5bb87783 : SUCCESS in 32m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d587b8364f6140b7bed1ae5a58460c93 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f15d6c18f44476bb74726720fa9607b : SUCCESS in 25m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbad3c7e63d642968780142a137438b9 : SUCCESS in 26m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f40f6537c1914612a2ca7b2263b50a9c : SUCCESS in 26m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92da9e1ae9c49fb9a4dbfa48f0c4da0 : SUCCESS in 11m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2b86a9d99e77414ba5abf416ab2d4f4d : SUCCESS in 1h 43m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/213d92634cf84a18bb3ebe1d31648eae : SUCCESS in 1h 00m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c334c4ade694121a1aa24dce9d5d551 : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11b73ae2a1ff4f70a23f51408cf4c07d : SUCCESS in 1h 28m 47s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/97f03139404a46d9adbfec560866a561 : SUCCESS in 32m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/92242da907484b7b8e5832fa866b65db : SUCCESS in 35m 02s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2795be3bac314f059bae9763476849f1 : SUCCESS in 36m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ca23754605544ce2afd5b303f2fd4cfd : SUCCESS in 38m 58s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e87b5c8295c147559961fa062b174761 : SUCCESS in 16m 26s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0d8e3653dae14d1d8edd3113ac1b8c55 : SUCCESS in 1h 42m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c024f736b824cef9fe4a88880fb40e7 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":7},{"id":"311916a63f340cd91050450b4b20d7b3c7e5c9b5","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-08-19 02:53:45.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"be60536745ad949575c0347ef5bc63194b009ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 02:54:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a15ffdd707cf647539fe826594769d426b5bd7be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 04:45:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af14fe995df04f17960db0866c73a6b2 : SUCCESS in 31m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6088356c17cd4c9aab8972d3f945f265 : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c25fcd03272041e3b5fa684f454fe328 : SUCCESS in 31m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5775daaf7706436281bf2284294bc074 : SUCCESS in 29m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d839f17d4c14bcabaf42450f163fcac : SUCCESS in 30m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac065e8ec02415f8007bf8565389579 : SUCCESS in 10m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/583bbee825e84b29ada6bb0ea3012893 : SUCCESS in 1h 43m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bc5d291277464e6c970b925693d8b2bc : SUCCESS in 1h 04m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7c3f96eb779644cba050700b6213f55a : SUCCESS in 54m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bb4efc18a9e4c988adb475c2d42734a : SUCCESS in 1h 26m 49s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcbe2c65f5b64e3ea56fd2a4b217e34b : SUCCESS in 38m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1f014058d3c6471f8c7cb0a6c0291d9b : SUCCESS in 33m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55613c31e758495b84143608b4ae5d71 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":7},{"id":"882f194d9fb5afabcf734ad53298c2eb780dc109","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 04:45:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5086709ae2d673716653cd8812247ea5a1cb5e69","revisions":{"22609d85d5f732e49d3563e27468b58ee0c3c7a0":{"kind":"REWORK","_number":1,"created":"2019-08-10 04:42:36.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/1"}}},"commit":{"parents":[{"commit":"f850095710071d1998bc7dabeeb7135dd75080bd","subject":"Merge \"Fix missing print format and missing ws between words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f850095710071d1998bc7dabeeb7135dd75080bd"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22609d85d5f732e49d3563e27468b58ee0c3c7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22609d85d5f732e49d3563e27468b58ee0c3c7a0"}]},"branch":"refs/heads/master"},"b43f91b6e6b0052dfa8aeaccd39b3ba94e789d8c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-12 02:31:33.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/2"}}},"commit":{"parents":[{"commit":"6dc9b685959d390cecf087fda4ab09a46807927f","subject":"Merge \"Suppress policy deprecation warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6dc9b685959d390cecf087fda4ab09a46807927f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 02:31:25.000000000","tz":-420},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b43f91b6e6b0052dfa8aeaccd39b3ba94e789d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b43f91b6e6b0052dfa8aeaccd39b3ba94e789d8c"}]},"branch":"refs/heads/master"},"95f669796b19dc0f0229c5b2bae09f9188433e4f":{"kind":"REWORK","_number":3,"created":"2019-08-12 22:58:48.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/3"}}},"commit":{"parents":[{"commit":"6dc9b685959d390cecf087fda4ab09a46807927f","subject":"Merge \"Suppress policy deprecation warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6dc9b685959d390cecf087fda4ab09a46807927f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-12 22:56:55.000000000","tz":-420},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/95f669796b19dc0f0229c5b2bae09f9188433e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/95f669796b19dc0f0229c5b2bae09f9188433e4f"}]},"branch":"refs/heads/master"},"24959ad8e0382a91321c370d24d31ed6400f3e75":{"kind":"REWORK","_number":4,"created":"2019-08-13 23:13:03.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/4"}}},"commit":{"parents":[{"commit":"6dc9b685959d390cecf087fda4ab09a46807927f","subject":"Merge \"Suppress policy deprecation warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6dc9b685959d390cecf087fda4ab09a46807927f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-13 22:57:47.000000000","tz":-420},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/24959ad8e0382a91321c370d24d31ed6400f3e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/24959ad8e0382a91321c370d24d31ed6400f3e75"}]},"branch":"refs/heads/master"},"4f1056ad4a366d371dc1129e332ee9e1349ef641":{"kind":"REWORK","_number":5,"created":"2019-08-14 22:43:49.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/5"}}},"commit":{"parents":[{"commit":"1d89cde11aabeb6b90005264f50ead7629b2411a","subject":"Don\u0027t reveal trust existence to unauthorized users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d89cde11aabeb6b90005264f50ead7629b2411a"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-14 22:43:05.000000000","tz":-420},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f1056ad4a366d371dc1129e332ee9e1349ef641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f1056ad4a366d371dc1129e332ee9e1349ef641"}]},"branch":"refs/heads/master"},"6a3648e782138bbbcabd0f28bdab41d1ab0ac2f1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-16 00:29:56.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/6"}}},"commit":{"parents":[{"commit":"fcab1cdd2af88c32851bcaec346117e5ca9c4ad6","subject":"Don\u0027t reveal trust existence to unauthorized users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fcab1cdd2af88c32851bcaec346117e5ca9c4ad6"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-08-16 00:29:56.000000000","tz":0},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6a3648e782138bbbcabd0f28bdab41d1ab0ac2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6a3648e782138bbbcabd0f28bdab41d1ab0ac2f1"}]},"branch":"refs/heads/master"},"5086709ae2d673716653cd8812247ea5a1cb5e69":{"kind":"REWORK","_number":7,"created":"2019-08-16 22:22:16.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/675720/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/675720/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/675720/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/675720/7"}}},"commit":{"parents":[{"commit":"6dc9b685959d390cecf087fda4ab09a46807927f","subject":"Merge \"Suppress policy deprecation warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6dc9b685959d390cecf087fda4ab09a46807927f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-10 04:37:23.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-08-16 22:10:44.000000000","tz":-420},"subject":"Add protection tests for trusts API","message":"Add protection tests for trusts API\n\nCurrently, the majority of access control enforcement for the trusts API\nis not done in policy, but hardcoded in the controller logic. The\ndefault policy check strings for these routes are empty. Before we can\nenable system scope and default roles through the trusts policy, we need\nto replicate the existing access control in policy. To do that, we\nshould test how it currently works. This patch adds those tests.\n\nThe trusts API is mostly only useable by the trustor or trustee. Mostly,\nadmins can\u0027t perform trust actions on behalf of the trustor or trustee.\nThe exception is for the delete action, but only when the is_admin\ncontext is set. This change also fixes a minor regression where the\nis_admin admin could not perform this action due to the auth_context not\nbeing populated.\n\nChange-Id: I6a5eca8240aa905e02fbf9bec335996c3a4f1c79\nPartial-bug: #1818846\nPartial-bug: #1818850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5086709ae2d673716653cd8812247ea5a1cb5e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5086709ae2d673716653cd8812247ea5a1cb5e69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
