)]}'
{"id":"openstack%2Fkeystone~677896","triplet_id":"openstack%2Fkeystone~stable%2Fqueens~Idd2b6254bb4fcf4b168cd35a1a27a9da2ca2280d","project":"openstack/keystone","branch":"stable/queens","hashtags":[],"change_id":"Idd2b6254bb4fcf4b168cd35a1a27a9da2ca2280d","subject":"Cap bandit","status":"MERGED","created":"2019-08-22 06:31:54.000000000","updated":"2019-08-22 10:44:46.000000000","submitted":"2019-08-22 10:44:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677896-1566470686514-66c16398","meta_rev_id":"36ab130da145fbf3cc57e38d78566e8ee26ad358","_number":677896,"virtual_id_number":677896,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-08-22 10:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":2,"date":"2019-08-22 07:18:20.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":1,"date":"2019-08-22 07:18:20.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 07:18:20.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-08-22 10:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bd9d5ac29c87f473b51a6d0a339e6bfe1b1b47d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-22 06:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"407f5d1d6f01e6f6d6249ffd4a5252c9f71f6ea0","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-08-22 07:18:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8dcbcac2953f3d124e6d03c664553f182cec2204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:33:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b20c75e612346d3ab9f792af8358d4e : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e3256de00e419d885d6a56db79568e : SUCCESS in 11m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86400949a99742d293559123bab6cc31 : SUCCESS in 14m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f964b84af7774f849afd7b8bfb0d1d70 : SUCCESS in 14m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3a92c219d64d486aa05f172a8c8a554e : SUCCESS in 9m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/829683f8f0d2429d84df5286862390e8 : SUCCESS in 1h 27m 45s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/1eec145bb5794e9da00de25290c70384 : SUCCESS in 42m 52s\n- keystone-dsvm-functional-v3-only https://zuul.opendev.org/t/openstack/build/92995a7c994e4f2bb7d2d3a5dd55b46e : FAILURE in 30m 27s (non-voting)\n- keystone-dsvm-py35-functional-v3-only https://zuul.opendev.org/t/openstack/build/73c2da8cc7f1422ea1571e8eed464dcd : FAILURE in 37m 05s (non-voting)\n- openstack-ansible-keystone-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a71e187ee8a471fb90913146769622a : FAILURE in 6m 43s (non-voting)\n- legacy-keystoneclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/aef19fdc8ad449a1978dfd0ac7bc4f2c : SUCCESS in 31m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fde17cf819143afa1e4400bceb76bb3 : SUCCESS in 19m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2c2b7de273fd2540499b1c0d2bf176e6a425b8ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:33:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de1b870e2484a1853873a6ac9c00d94edbc7fa3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:44:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36ab130da145fbf3cc57e38d78566e8ee26ad358","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:44:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fca8804be4ac486f9b30579423fbebd8 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d669146a01a743579ce72a80f99453ca : SUCCESS in 9m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8f5457a51bf4bb0841f0bc4d0dfe4f2 : SUCCESS in 15m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d6585dba73984e669a5a753db7f84994 : SUCCESS in 15m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d6f9a13e1f9949cd868b20a534b2b63e : SUCCESS in 12m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c4e68d62c534ebd93230957e9c3664c : SUCCESS in 1h 35m 10s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a3ce5a84c7544c2bc955a60831a8a47 : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3a2d75eacac179f3f94c66a184ad51fde16bef0","revisions":{"a3a2d75eacac179f3f94c66a184ad51fde16bef0":{"kind":"REWORK","_number":1,"created":"2019-08-22 06:31:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/96/677896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/96/677896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/677896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/677896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/677896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/96/677896/1"}}},"commit":{"parents":[{"commit":"c2f619b14c1468f3d4d89e1b07d3e85815889df4","subject":"Blacklist bandit 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c2f619b14c1468f3d4d89e1b07d3e85815889df4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-22 06:31:02.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-22 06:31:02.000000000","tz":120},"subject":"Cap bandit","message":"Cap bandit\n\nBandit is a linter, let\u0027s cap the version in a stable branch\nso that new versions of bandit do not introduce pepe8 failures.\n\nChange-Id: Idd2b6254bb4fcf4b168cd35a1a27a9da2ca2280d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a3a2d75eacac179f3f94c66a184ad51fde16bef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a3a2d75eacac179f3f94c66a184ad51fde16bef0"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
