)]}'
{"id":"openstack%2Fkeystone~680570","triplet_id":"openstack%2Fkeystone~stable%2Frocky~Ib9e161411f0985785eec46c51d721ef7421ee090","project":"openstack/keystone","branch":"stable/rocky","hashtags":[],"change_id":"Ib9e161411f0985785eec46c51d721ef7421ee090","subject":"Add retry for DBDeadlock in credential delete","status":"MERGED","created":"2019-09-06 04:35:48.000000000","updated":"2019-10-18 22:10:48.000000000","submitted":"2019-10-18 22:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680570-1571436520009-9e4bfd77","meta_rev_id":"75d2d8da3f4e5e7470b41f46322ebb99171ad041","_number":680570,"virtual_id_number":680570,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-09-09 16:31:27.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2019-09-10 06:05:18.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 22:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-18 18:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2019-10-17 21:28:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-10-18 18:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-10-18 18:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 06:05:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-10-17 21:28:17.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-10-18 18:30:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-18 18:52:03.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-10-18 22:08:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26ee0b6b24e71c891814366118cf3e10fd3e1ac6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-09-06 04:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c302f2463c72b3b651833ca60e203bf54ed1acde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 19:00:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2572889e78d4a8793a7fbda711bbb47 : FAILURE in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d2ef01526d646b0b824cee8032e1182 : FAILURE in 20m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e514cda8cb4421bacf8653bbe918bc8 : FAILURE in 11m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f467b76802954eb9a576fc78a81077dc : SUCCESS in 10m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bc0384d7dd1b4becb2c1e81c1791ccbd : SUCCESS in 1h 28m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/085eea4eb8c84707a206f52c0a8364c1 : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/963fe37dcd8a4261ba96d24a3321b8d0 : SUCCESS in 1h 25m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c16e2c4080fb41848f1d6014f483011f : SUCCESS in 13m 43s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/eef6ea66aabe488fbd5f66a2580d94e6 : SUCCESS in 36m 04s\n- keystone-dsvm-functional-v3-only https://zuul.opendev.org/t/openstack/build/509676b4faa04a1d988e0938cc71c5a4 : SUCCESS in 36m 44s (non-voting)\n- keystone-dsvm-py35-functional-v3-only https://zuul.opendev.org/t/openstack/build/e5092110bc3142c4a02ee04a28c82c15 : SUCCESS in 38m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbf501bd014047b7be580e9bf984b682 : FAILURE in 19m 26s\n- legacy-keystoneclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/f337d89921844354a7c73999be8ea6e4 : SUCCESS in 36m 19s (non-voting)\n- openstack-tox-cover-keystone-xenial https://zuul.opendev.org/t/openstack/build/2294af16a0d44e319c5b72a4a6c47202 : FAILURE in 18m 07s\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b6a4d171329a4c9d89586db0360cc722 : SUCCESS in 1h 40m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc403b78828ea250adbcc4f3ddfadb7d629e4bf2","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-09-09 03:23:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8309f6e38ab7df3c041f7e6830230313eec050f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 05:05:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecddae6d9e6d4996a2747126e451bb9a : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba495097f2fa41fd92b7fe16801187ef : SUCCESS in 17m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/acfa9d8a057549589aed7a178f78fb85 : SUCCESS in 19m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a98b1df1ce5040ae93f46a68338f3d54 : SUCCESS in 9m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/098822797ec6459b84d8ebc2b4ea469c : SUCCESS in 1h 25m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6d41327b5a5e4d54b274bded3752e84e : SUCCESS in 55m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/006d4f0ce319498bb40be038d5f03681 : SUCCESS in 1h 31m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b669bf20a47b458f9f8628e11aec76b1 : SUCCESS in 10m 49s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/89fb174006bc467881ad8acfef92d34c : SUCCESS in 31m 32s\n- keystone-dsvm-functional-v3-only https://zuul.opendev.org/t/openstack/build/75efb377947c47fca7a0768382613ab6 : SUCCESS in 31m 01s (non-voting)\n- keystone-dsvm-py35-functional-v3-only https://zuul.opendev.org/t/openstack/build/01907a4d95e2471688ced5a496b9cae2 : SUCCESS in 32m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d7a98723756477b9cb38d9dd6515a8d : SUCCESS in 13m 31s\n- legacy-keystoneclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/43a08e2187234739a62954a927c47f73 : SUCCESS in 43m 38s (non-voting)\n- openstack-tox-cover-keystone-xenial https://zuul.opendev.org/t/openstack/build/f5b7a8db5c6543829e09a8724675ca5e : SUCCESS in 19m 41s\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4611a01e31694609aaf29cba0f519ae5 : SUCCESS in 1h 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2a2f7554e82a4251a05c238867d5a71c9ccff33","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-09-09 16:21:17.000000000","message":"Patch Set 2:\n\nFile conflict should be mentioned in commit message:\n\nConflicts:\n       keystone/tests/unit/test_v3_credential.py\n\nThe conflict is due to stable/rocky only patch: Ic7dddc4d2fe7b6c6ae3bf6aed6c4c048743b3eed\n\n(Also it\u0027s easier to follow the backport changes if the patch is cherry-picked from the stein to rocky and not directly from master, which would add one extra cherry-pick line in commit message:\n\"(cherry picked from commit 52ef61868930fcccbabe91f27a805211abebde84)\")\n\nOtherwise the code looks good.","accounts_in_message":[],"_revision_number":2},{"id":"967e64a5e440ca9d1267fcbe1d889cbebb465705","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-09-09 16:31:27.000000000","message":"Patch Set 2:\n\n\u003e File conflict should be mentioned in commit message:\n\nI don\u0027t think there is any real conflict in code, as this adds a new test and 2 related imports like other branches.","accounts_in_message":[],"_revision_number":2},{"id":"f1aa991f7a97dcb587f9965de6ebddab9d69ff41","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-09-10 06:05:18.000000000","message":"Patch Set 2:\n\n\u003e I don\u0027t think there is any real conflict in code, as this adds a new test and 2 related imports like other branches.\n\nAccording to git there is. Anyway, it\u0027s just a good practice to indicate that.","accounts_in_message":[],"_revision_number":2},{"id":"639f9d513f6ec7937282d9fabb7b337b42c155f7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-17 21:28:17.000000000","message":"Patch Set 2: Code-Review+1\n\npatch exists on master and stein","accounts_in_message":[],"_revision_number":2},{"id":"0fd333c00bcb38f2d29a83ceeb9bd0e097106fba","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-18 18:30:32.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e File conflict should be mentioned in commit message:\n \u003e \n \u003e Conflicts:\n \u003e keystone/tests/unit/test_v3_credential.py\n \u003e \n \u003e The conflict is due to stable/rocky only patch: Ic7dddc4d2fe7b6c6ae3bf6aed6c4c048743b3eed\n \u003e \n \u003e (Also it\u0027s easier to follow the backport changes if the patch is\n \u003e cherry-picked from the stein to rocky and not directly from master,\n \u003e which would add one extra cherry-pick line in commit message:\n \u003e \"(cherry picked from commit 52ef61868930fcccbabe91f27a805211abebde84)\")\n \u003e \n \u003e Otherwise the code looks good.\n\nAgree, it would be better to go branch by branch. It does help when tracking things.\n\nWe are trying to get this back to queens, and the deadline for queens going to EM is coming up quick. This has been merged to later branches and looks good to backport.","accounts_in_message":[],"_revision_number":2},{"id":"d263d9a3ced8ebd55840a5d35ed6a10b536074a5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-10-18 18:52:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ece7403365018372e3bae88bbd8adafa745f0f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:52:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fc107edc9e6d8b08558c6e00ca47b15e77904e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 22:08:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d751065b4042f4a208524aaab495f3 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00df57e59ed048329ddd8b7ccd73ca4e : SUCCESS in 19m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e2231dfbf1924500ad65f2cd9707567e : SUCCESS in 19m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0671cb509afa44718205f0702e9a3b74 : SUCCESS in 9m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c4447f6ee15343349d569c36f842d0c0 : SUCCESS in 1h 38m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4dcc63195f0a486cb9292c51ff433a79 : SUCCESS in 57m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a34089eb47da4a4aadfe1223503e75de : SUCCESS in 1h 30m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8d74fdf47f54488aa5f46ae9d27b78d : SUCCESS in 15m 01s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/2596e262807d4d5cb2f70beb5ee4a3bc : SUCCESS in 36m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2619d97f5261430ab2d0de4accc7960b : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":2},{"id":"419b58558aecc4582dbb721057c33acd2f559f65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 22:08:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"75d2d8da3f4e5e7470b41f46322ebb99171ad041","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 22:10:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a181b289abb44ac580247bf6155cb10a : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09547696f2bb952f29b5dc95cbb572b9d1d0b72b","revisions":{"fd305074757065b4a12de1c31d019c958c7ed69c":{"kind":"REWORK","_number":1,"created":"2019-09-06 04:35:48.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/70/680570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/70/680570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/680570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/680570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/680570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/70/680570/1"}}},"commit":{"parents":[{"commit":"c220cc450cabb652c4b0afab265b93c7a7aa2afd","subject":"Revert \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c220cc450cabb652c4b0afab265b93c7a7aa2afd"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-15 15:53:46.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-09-06 04:33:39.000000000","tz":330},"subject":"Add retry for DBDeadlock in credential delete","message":"Add retry for DBDeadlock in credential delete\n\nAdds oslo.db retry wrapper to delete_credential_for_user method.\n\nChange-Id: Ib9e161411f0985785eec46c51d721ef7421ee090\nCloses-Bug: #1840291\n(cherry picked from commit e989bd0637629b0fb15976186b0941fec0f13b25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fd305074757065b4a12de1c31d019c958c7ed69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fd305074757065b4a12de1c31d019c958c7ed69c"}]},"branch":"refs/heads/stable/rocky"},"09547696f2bb952f29b5dc95cbb572b9d1d0b72b":{"kind":"REWORK","_number":2,"created":"2019-09-09 03:23:04.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/70/680570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/70/680570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/680570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/680570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/680570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/70/680570/2"}}},"commit":{"parents":[{"commit":"c220cc450cabb652c4b0afab265b93c7a7aa2afd","subject":"Revert \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c220cc450cabb652c4b0afab265b93c7a7aa2afd"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-15 15:53:46.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-09-09 03:22:45.000000000","tz":330},"subject":"Add retry for DBDeadlock in credential delete","message":"Add retry for DBDeadlock in credential delete\n\nAdds oslo.db retry wrapper to delete_credential_for_user method.\n\nChange-Id: Ib9e161411f0985785eec46c51d721ef7421ee090\nCloses-Bug: #1840291\n(cherry picked from commit e989bd0637629b0fb15976186b0941fec0f13b25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/09547696f2bb952f29b5dc95cbb572b9d1d0b72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/09547696f2bb952f29b5dc95cbb572b9d1d0b72b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
