)]}'
{"id":"openstack%2Fkeystone~681161","triplet_id":"openstack%2Fkeystone~master~I213d6f91a815ba0dad5f5a04a5f0309722c91f62","project":"openstack/keystone","branch":"master","topic":"lower-constraints-timeout","hashtags":[],"change_id":"I213d6f91a815ba0dad5f5a04a5f0309722c91f62","subject":"Bump timeout for lower-constraints job","status":"MERGED","created":"2019-09-10 06:25:00.000000000","updated":"2019-09-11 05:56:55.000000000","submitted":"2019-09-10 23:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681161-1568156902436-7d3e54bf","meta_rev_id":"31b79f322037288a5322f8cd15eebf9a8a851f05","_number":681161,"virtual_id_number":681161,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-10 15:02:13.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-09-10 23:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-09-10 07:16:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-09-10 15:02:45.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-09-10 15:03:50.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 07:16:19.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-10 15:03:50.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-10 23:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b7324a8bc5a19c49d76e851d4a0a442bb3ef8da","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 06:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0fa04caf6dfe8775073e9aa6dad4664bcd9e85e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-10 07:16:19.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Colleen. This will really help for gates to get rid of timeouts failure.","accounts_in_message":[],"_revision_number":1},{"id":"cd2d1366267125e1078f602ffae38ef8b5cc8d20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:52:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd1b09070584b5a85fb2d6037d3cbc3 : SUCCESS in 52m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1414d517b59541e18ec4e9a8cd5c0748 : SUCCESS in 51m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcdd8632c32d4292a82b48134db063d4 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1061d8f9b6d346e7a661bceab2120b2c : SUCCESS in 34m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/922e3ff33bec460cbba3c9e26a3dc5c3 : FAILURE in 32m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca64e2f93c3947e884ff835a376f6115 : SUCCESS in 34m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1b6379af32405fa6a776bcf071cd37 : SUCCESS in 10m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9453ae837c924dbca326b9863053494b : SUCCESS in 1h 42m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9d2f97571e7a47d7ad8af8ec8afc8367 : SUCCESS in 58m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c98c79b4eb49447682cad4d1acbf64f5 : SUCCESS in 1h 12m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a34b4fa901741d6b2f724e60a857079 : SUCCESS in 1h 25m 31s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/751c8eceb8274f97ac11beb158f45b1e : SUCCESS in 33m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1f36a84253bd4ee8a249323703e3e09d : SUCCESS in 33m 12s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/8cfde3aedb064982b9850e6d954baae7 : SUCCESS in 39m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/73befbd3b5df4576927577c6223cfe19 : SUCCESS in 33m 13s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/065672da81be46d5a05caa1e7b2b5fb0 : SUCCESS in 16m 19s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/651cd99b029442d69ba286b4b9314fc5 : SUCCESS in 1h 39m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c34bcb0f23744dc89628c4ac42daa9c : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"084d05d6f506f2e200e94baa632e89cf39b062fd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 15:02:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d0c704e8b2f2c8212f76adcb815034ef4a6cd4d4","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-10 15:02:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0cc9b05cca22ad96cca3cfa606e2775f81342af6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-10 15:03:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69336e328dbfb69b8495fb13970c87ed8707cf71","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-10 15:04:26.000000000","message":"Patch Set 1:\n\nGoing to kick this through since it is trivial and relates to something we\u0027re already merged. It\u0027s also imperative to get things done this week before feature freeze.","accounts_in_message":[],"_revision_number":1},{"id":"1dea467a38500e5f2106cc16d19b89797fac9758","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 20:53:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"551247c83c16dd878efb0c899664ba294d24da57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:02:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f419b477434cadade95c55b0c5da0a : SUCCESS in 47m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a21681ca62a4a4c90b788a0afb08140 : SUCCESS in 38m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5359d296016a48d39953c5ef2321d811 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/581f843873304e2cbc564c87b772d6d4 : SUCCESS in 40m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7d2593db7d3414782ace0039a8f1a36 : SUCCESS in 32m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a36fa153382c4d1d9604da715cde70cf : SUCCESS in 35m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c534d3ef2d4dd4834d4e7a66169276 : SUCCESS in 9m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/65e5eb26b5f546f4bbe8fb2107ae3457 : SUCCESS in 1h 28m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/210aa52b9d5843a7a49cd5cecf00bba0 : SUCCESS in 54m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e3cde4f6fefd49aba3d7fe347645d4d7 : SUCCESS in 1h 27m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d3ce2f130db42f3af1ffeb992050f9d : SUCCESS in 1h 41m 03s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c2fa79c8a0e46d79e683801dc2abd3d : SUCCESS in 38m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b0b582d08de547859e6d32ae6565d09a : SUCCESS in 35m 24s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fb42c2d87b1541829d6d4fc120521a9a : SUCCESS in 34m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fb8918c43fb74d8c9b02ffa167eefe6c : SUCCESS in 46m 38s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/aef48f9439b242e5b5231c4ebd785d39 : SUCCESS in 19m 32s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cb8c9004afb74f8d8972f9c1f56e36da : SUCCESS in 1h 34m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23ddbfb16f574dc195525de741ca0d75 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"942749a2db6460cf188165f3328cd43e6cbcd700","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:08:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ba5ea03d08c42aeb7a6dd5c4ee715b0 : SUCCESS in 40m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ba11ec05e5454fa632c3d8d05d49db : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40fe65cf4455473d9e7831dda2ab1a91 : SUCCESS in 40m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4ca6b36747b4c78b9f6810415e27545 : SUCCESS in 34m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/461e580f4e984b918e6cc4d87b6d3171 : SUCCESS in 33m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/048db708c69c4ee59e2eb057b507464c : SUCCESS in 12m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a47ffd79557450fa11310b76eb17f45 : SUCCESS in 1h 41m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/78c56dca33c542e39e4ec6ce5691dc14 : SUCCESS in 1h 02m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a6a7025b647044b29a1a4644caa2f731 : SUCCESS in 1h 12m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e13dedd19f24bf986765a276ed5c4f1 : SUCCESS in 1h 23m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a4c70e31ac94c33ade7ff411e1e2301 : SUCCESS in 41m 02s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/a26338a606f34a75a37a2e49b57930ef : SUCCESS in 36m 50s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d67218af94644133a3d49969a143a5da : SUCCESS in 33m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91f949bf84f74ab99a282113f0d3bcb9 : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"25d6193e2fd62e8738badfcbfc987f6b5e2f19a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:08:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aaac5d34ad602e96ee2c47956243205870bd46f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:11:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3baf28f5a04aa287682f96f87f336a : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1},{"id":"93fc9202c505a125e3e278441b29d1699ed30b80","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-11 05:52:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2fc989d2ab82af833c8def1957fc5e0dfba13d7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 05:56:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a766085abc1f52775019f9d7da515e06e31dbc3b","revisions":{"a766085abc1f52775019f9d7da515e06e31dbc3b":{"kind":"REWORK","_number":1,"created":"2019-09-10 06:25:00.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/61/681161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/61/681161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/681161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/681161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/681161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/61/681161/1"}}},"commit":{"parents":[{"commit":"96de25c6715e29270204855b696363596266d5de","subject":"Merge \"Add tests for domain users for trusts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/96de25c6715e29270204855b696363596266d5de"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-10 06:24:16.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-10 06:24:16.000000000","tz":-420},"subject":"Bump timeout for lower-constraints job","message":"Bump timeout for lower-constraints job\n\nIn ba0dbdf we raised the timeouts for most of the tox jobs but neglected\nthe lower-constraints job which suffers the same timeout issues, so add\nthat one as well.\n\nChange-Id: I213d6f91a815ba0dad5f5a04a5f0309722c91f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a766085abc1f52775019f9d7da515e06e31dbc3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a766085abc1f52775019f9d7da515e06e31dbc3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
