)]}'
{"id":"openstack%2Fkeystone~682266","triplet_id":"openstack%2Fkeystone~master~Ie45955f5cc54563cc9704d7cb2b656b5544ae030","project":"openstack/keystone","branch":"master","topic":"train-rc1-policy-fixes","hashtags":[],"change_id":"Ie45955f5cc54563cc9704d7cb2b656b5544ae030","subject":"Remove policy.v3cloudsample.json","status":"MERGED","created":"2019-09-16 02:56:12.000000000","updated":"2019-10-04 21:48:40.000000000","submitted":"2019-10-04 21:46:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682266-1570225610955-e8c316f0","meta_rev_id":"0dd5e2e4f3f0746a69708ed504f11093009b8015","_number":682266,"virtual_id_number":682266,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"date":"2019-10-04 15:43:12.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 21:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-10-02 21:42:31.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-10-04 09:52:15.000000000","_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-10-04 09:52:15.000000000","_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 04:04:59.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2019-09-27 08:54:02.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-09-27 10:10:28.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-10-02 21:42:31.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-10-04 09:52:15.000000000","updated_by":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"reviewer":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"state":"REVIEWER"},{"updated":"2019-10-04 21:46:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"154b9580b6885c0e94b26a21f0da0c8a730849c6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-16 02:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88ae37e787d282886550d72c1b6a818923d0f812","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 04:17:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dd77a17c033409e9b8b73eec9889902 : FAILURE in 1h 07m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab1502d14e8b492daabd085f9b321abf : FAILURE in 1h 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1017eb8d1c49d4b8f043d0d925424e : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/535ab57b16c54376afaaf3904b988332 : FAILURE in 1h 20m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e44fb6992414fa6ac92951b3353d3c7 : FAILURE in 1h 16m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c7c86c24eb04e3895ef2df56d609bf7 : FAILURE in 1h 02m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c08b0517d04f6d840c3c73f0a221f5 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c0bb53c68c0e01ac20b5468adef98f25f70c52c9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-16 20:37:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"293d54518e391207974eb0272621539024edf5c0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-16 21:29:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ff41968d1ce8b8f69c063c1644d39fe79005458","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 01:15:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3cc7b6c2c94432983a988e2c4858871 : FAILURE in 1h 07m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee9ac2dfb5284536822f8510d8ac255a : FAILURE in 1h 08m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6815295179844a42a0391c7097e62922 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef09846cfd7543dc9e9d5757964bfe94 : FAILURE in 49m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2438f01219bd4cecb3d861b9f46935d3 : FAILURE in 43m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d023b08e078646d08e5e7fcbb69b56b9 : FAILURE in 38m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d103994908ee4214ab74828e5098b2a4 : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"53b8f237132d29df786429752c4ed8b2c7e52fc9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-18 14:16:31.000000000","message":"Topic set to train-rc1-policy-fixes","accounts_in_message":[],"_revision_number":3},{"id":"1b6aa1d5ee08890e0af023f19d959632830dac13","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-26 13:27:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"39ed5c89fdb1bce43533106f919b4dd79a7d6ce2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-26 13:36:41.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"440dce471b250d941d55d714be5ca668964db61c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:11:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c48363708945649cabfccd26737838 : SUCCESS in 27m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd83f82dd4db4ad9992ca33a33fc05c8 : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985da82e0de4469ea54e9281be9089b4 : FAILURE in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb9f1e3a717640fa88fc9de3855aba1f : SUCCESS in 15m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9892e7cc024481ea17c1d8c7aec7a09 : SUCCESS in 13m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cfcf27e04864c268dbe4fa4ec357a6e : SUCCESS in 16m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d4aaaab86348deaae059b4539deeae : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/811aac01c07c4505b0c830ccee674e59 : SUCCESS in 15m 35s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7193118c18294feb8de4d1af26f64f26 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":4},{"id":"78b083a1df35a43346608cc3f1cc65864a3e0233","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-27 04:04:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"225e649d82a8284ecf6786c5b1e6e570ad5186d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:49:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e432fd4066244219bbdf54a732c4b80 : SUCCESS in 16m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c57ee5322d904c158d8203ff911104c7 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf21526a31c6454b9e3d85a9c9942a1c : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf6475c0972c4260a8d38319efe1b400 : SUCCESS in 24m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb4e065dc51d48f2a466f33be2437a8e : SUCCESS in 22m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3594bc4765b34b9ab1dec804b3a3172e : SUCCESS in 20m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a142ffe1eb4c81b6de330d612305f0 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e018500ce39c473691f4bd762f1d346e : SUCCESS in 14m 21s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a405e58dcb234528885b8602f60cc369 : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":5},{"id":"5e8843fdfd0370b8ca42e0787ff51fd6e44a3e0a","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-09-27 08:54:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c02ac1b73e40cca438f97ed2c3a367b850a243fb","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-27 10:10:28.000000000","message":"Patch Set 5: Code-Review+1\n\nSince the deprecated policies are getting removed, we have tests lined up for cleaning.[1]\n\n[1]https://github.com/openstack/keystone/blob/master/keystone/tests/protection/v3/test_credentials.py#L1124","accounts_in_message":[],"_revision_number":5},{"id":"688e2cdf973a92cb75e752928843cddf52df9807","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-27 13:10:14.000000000","message":"Patch Set 5:\n\n\u003e Since the deprecated policies are getting removed, we have tests\n \u003e lined up for cleaning.[1]\n \u003e \n \u003e [1]https://github.com/openstack/keystone/blob/master/keystone/tests/protection/v3/test_credentials.py#L1124\n\nOh - good point. I suppose we can remove test_v3_protection.py with the patches that clean up the policy overrides.","accounts_in_message":[],"_revision_number":5},{"id":"9e0a26e071bd60407df748ecff40d527adf72a07","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-30 18:22:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8c89c8106e2fdd152fc26304546e9f62987d4fb8","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-10-02 20:26:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"547ee19585dce9ae108f7aa45b4eb2dac29145b3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-10-02 20:26:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"65508ee8698b0b8464ea42fbb667a0a86cc53e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 21:22:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12182d61f2b046a1b6a163ea5ed40be4 : SUCCESS in 32m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80889880c7cf47c9bdf64d60900757fd : SUCCESS in 25m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135c80da93bf4171a7bc0daff21d4eb0 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdb9370fde354798b6aa414f3777b630 : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/117fd270e07941a5a009cbb52c48bf2c : SUCCESS in 15m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8999ee9fb3a24e749fddb7b2914f6b2d : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8de9769c9b043cbafcef9bffb59e19e : SUCCESS in 15m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/895f15ec824a4a6699dcae6963881aa0 : SUCCESS in 16m 25s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/2a82e53352e54049b15714414e2b4277 : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":6},{"id":"79b6a84754bce17e20b8ae698c180486402dc646","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-02 21:42:31.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4b4a519e24b874d165a2fc8e87365032e9e757f1","author":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"date":"2019-10-04 09:52:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nlgtm!","accounts_in_message":[],"_revision_number":6},{"id":"dbdedc8f15110d62663ca8118eabf8d26990a53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 09:52:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d1af8ff05575cc9fda346fc691e2e4881ebbb9fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:38:28.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8030c0c1512647d386f06b3eb864c618 : TIMED_OUT in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf792fa50a1c439d9b03a3b595a9a935 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac7d0317ac794cf28994048b8c5af856 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ba485c02e14a53993d02d4d33bbc5b : SUCCESS in 13m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da7266d85a9c4a1e84a46aebe13dffd9 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09abe858aa9d4567aca37501d40790c1 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b833f8d07a1e4279ba8ce9f96c371158 : SUCCESS in 15m 41s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/95d4cea631c3404087510c74f7a7d2e5 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":6},{"id":"5f8cf029384b3b3348070efd8043a26e1ad41ed2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-10-04 13:29:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1ef2410885b282779586144a204aa02bdb830070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:17:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0716f7ddf1ba477b9de416d74637083f : SUCCESS in 19m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c8ccd7c7604ccfb8710b66709eaeb0 : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9bc8e231e54733a39d2118ba60610c : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f955c2c54ee1441a93bae8e0994774a0 : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8286d96d84444322b623e4fc0e5a6825 : SUCCESS in 12m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b2fc924b059442a9b9cf18b8ecaded3 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852725ee14f1441c8dcb4959929ca3f5 : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2484fb02786c472da9b7cd83249f1c0c : SUCCESS in 16m 35s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/180c5f0deec743adb03c944efdee94dc : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":6},{"id":"5dc49d80e25d13822c9f9785831c0d2c890d0426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:17:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"613d0fbd2a33e6014bca1c30246c8c33497ac4c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:39:43.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c707d9d9b44be8a79d8cad36b07fe1 : TIMED_OUT in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d872f5838f4bfebfc515eacccd2c56 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d9bfc3676fe4f739f8cd37d5a3dede4 : TIMED_OUT in 40m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80748d7c2bc24278b39ac726e7ae6722 : SUCCESS in 13m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4787652beed4fafb4f2db324111d95a : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ac717b6259f4694a1af000a94a646ac : SUCCESS in 16m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da03b78f04054e36af02447ef0257db2 : SUCCESS in 17m 05s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d4d305823aa7402f97432570d257537f : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":6},{"id":"ce8eb29caff3871f538572b43db83b23e733e0a7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-10-04 15:43:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"aa88a7179830d67d88997964baf72a53fabd89c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:41:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ac3ebf23bf44469c200df1e8724864 : SUCCESS in 14m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74715a2c9f9d4db18876998384cfb8f7 : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347a11e87086463f878f4c7ab373ff66 : SUCCESS in 7m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/833de29f36344963830c956ea37af01d : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d289e1e850834a14bdac6ae8297b7717 : SUCCESS in 15m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5aae23f22250483a850f7e26d06c8c5b : SUCCESS in 18m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad25c0209f59497da470330815dfc0a4 : SUCCESS in 23m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f2ed06991474df8a3f82ddd033ff2cc : SUCCESS in 16m 27s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/686a31624d144a9e84fb8fb299627c4d : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":6},{"id":"9be8fa0c5cf5b2bdb13773c14838af380979c251","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:42:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a3dc5098269b3b6c122685ca7c035b7cc1ac3301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 20:08:55.000000000","message":"Patch Set 6:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a380156356f55950438d364a89d4d5fb4c83cc03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 21:46:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb18e526bac42d7b366bfa53163c14e : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed379cba85245caab75773b626ac363 : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d17cf0d7d416429dabf6fd14f98e84a1 : SUCCESS in 15m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b1e1cafd9484ee084eaf3f5a428ac71 : SUCCESS in 15m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abcc78e6cfd64c8b9955892d68bca90b : SUCCESS in 17m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a10895cffd74a96839e8e1db9602296 : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5239bb1101d049d6b54590e3c908ecff : SUCCESS in 14m 22s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/21c007994f08407e8815862f9cd55c2d : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":6},{"id":"77b83246c2e023a3070666ecc9eb7c01998d40bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 21:46:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"0dd5e2e4f3f0746a69708ed504f11093009b8015","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 21:48:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62cbd36ad2a49118153be3a2d9b81fd : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbc8ff3f1f8747f18aaaf1ef9f25fce3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d4a6023de5bdfe5a6e9214579a35e083a45c1151","revisions":{"220d371798b48694d24c955b5eefac99ef19f671":{"kind":"REWORK","_number":1,"created":"2019-09-16 02:56:12.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/682266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/682266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/682266/1"}}},"commit":{"parents":[{"commit":"18e0080af3dcc0a96ff5d98aeb5f517080a35fb2","subject":"Merge \"Remove system Domain Config from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e0080af3dcc0a96ff5d98aeb5f517080a35fb2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"subject":"WIP: Remove unused and unnecessary policies","message":"WIP: Remove unused and unnecessary policies\n\nThese policies have either been made defaults or are no longer used.\nWe can remove them.\n\nThis patch should do the following before removing WIP:\n\n  - remove the remaining policies in policy.v3cloudsample.json (thus\n    removing the entire file)\n  - remove the policy test in test_policy.py that ensures both policy\n    files match\n  - remove any remaining tests that formerly exercised\n    policy.v3cloudsample.json\n\nChange-Id: Ie45955f5cc54563cc9704d7cb2b656b5544ae030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/220d371798b48694d24c955b5eefac99ef19f671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/220d371798b48694d24c955b5eefac99ef19f671"}]},"branch":"refs/heads/master"},"1282271bcde5ca9e4b55e1578e324a4ad4130047":{"kind":"REWORK","_number":2,"created":"2019-09-16 20:37:12.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/682266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/682266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/682266/2"}}},"commit":{"parents":[{"commit":"18e0080af3dcc0a96ff5d98aeb5f517080a35fb2","subject":"Merge \"Remove system Domain Config from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e0080af3dcc0a96ff5d98aeb5f517080a35fb2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 20:36:59.000000000","tz":0},"subject":"WIP: Remove unused and unnecessary policies","message":"WIP: Remove unused and unnecessary policies\n\nThese policies have either been made defaults or are no longer used.\nWe can remove them.\n\nThis patch should do the following before removing WIP:\n\n  - remove the remaining policies in policy.v3cloudsample.json (thus\n    removing the entire file)\n  - remove the policy test in test_policy.py that ensures both policy\n    files match\n  - remove any remaining tests that formerly exercised\n    policy.v3cloudsample.json\n\nChange-Id: Ie45955f5cc54563cc9704d7cb2b656b5544ae030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1282271bcde5ca9e4b55e1578e324a4ad4130047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1282271bcde5ca9e4b55e1578e324a4ad4130047"}]},"branch":"refs/heads/master"},"85e3d6952752cdfbe9490dded93d2bf782991f57":{"kind":"REWORK","_number":3,"created":"2019-09-16 21:29:55.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/682266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/682266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/682266/3"}}},"commit":{"parents":[{"commit":"18e0080af3dcc0a96ff5d98aeb5f517080a35fb2","subject":"Merge \"Remove system Domain Config from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e0080af3dcc0a96ff5d98aeb5f517080a35fb2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 21:29:47.000000000","tz":0},"subject":"WIP: Remove unused and unnecessary policies","message":"WIP: Remove unused and unnecessary policies\n\nThese policies have either been made defaults or are no longer used.\nWe can remove them.\n\nThis patch should do the following before removing WIP:\n\n  - remove the remaining policies in policy.v3cloudsample.json (thus\n    removing the entire file)\n  - remove the policy test in test_policy.py that ensures both policy\n    files match\n  - remove any remaining tests that formerly exercised\n    policy.v3cloudsample.json\n\nChange-Id: Ie45955f5cc54563cc9704d7cb2b656b5544ae030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/85e3d6952752cdfbe9490dded93d2bf782991f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/85e3d6952752cdfbe9490dded93d2bf782991f57"}]},"branch":"refs/heads/master"},"6ff63fd457e3ba7ff3a8e98ba3ea0fa5c0ae6de0":{"kind":"REWORK","_number":4,"created":"2019-09-26 13:27:32.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/682266/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/682266/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/682266/4"}}},"commit":{"parents":[{"commit":"ea5d02e8e8db72f189b2ab47b5d9cc877161a460","subject":"Merge \"Allow domain users to access the limit API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea5d02e8e8db72f189b2ab47b5d9cc877161a460"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-26 13:27:24.000000000","tz":0},"subject":"Remove policy.v3cloudsample.json","message":"Remove policy.v3cloudsample.json\n\nWe\u0027ve make all the default policies keystone supports better by\nincorporating default roles and scope types. These changes have made\nthe ``policy.v3cloudsample.json`` file obsolete.\n\nLet\u0027s simply things for users, operators, and develpers by removing\nit.\n\nA follow-on patch will remove the test_v3_protection.py file since\nthose behaviors are passing all the protection tests with the default\npolicies in code.\n\nRelated-Bug: 1805880\nCloses-Bug: 1630434\nCloses-Bug: 1806762\nChange-Id: Ie45955f5cc54563cc9704d7cb2b656b5544ae030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6ff63fd457e3ba7ff3a8e98ba3ea0fa5c0ae6de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6ff63fd457e3ba7ff3a8e98ba3ea0fa5c0ae6de0"}]},"branch":"refs/heads/master"},"fc7aedc1c142c36d31487f1656739bfd4a553c2c":{"kind":"REWORK","_number":5,"created":"2019-09-27 04:04:59.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/682266/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/682266/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/682266/5"}}},"commit":{"parents":[{"commit":"e860c69831289a800a1d7bb52e8621fc460f260b","subject":"Merge \"Fix validation of role assignment subtree list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e860c69831289a800a1d7bb52e8621fc460f260b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-27 04:04:39.000000000","tz":0},"subject":"Remove policy.v3cloudsample.json","message":"Remove policy.v3cloudsample.json\n\nWe\u0027ve make all the default policies keystone supports better by\nincorporating default roles and scope types. These changes have made\nthe ``policy.v3cloudsample.json`` file obsolete.\n\nLet\u0027s simply things for users, operators, and develpers by removing\nit.\n\nA follow-on patch will remove the test_v3_protection.py file since\nthose behaviors are passing all the protection tests with the default\npolicies in code.\n\nRelated-Bug: 1805880\nCloses-Bug: 1630434\nCloses-Bug: 1806762\nChange-Id: Ie45955f5cc54563cc9704d7cb2b656b5544ae030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fc7aedc1c142c36d31487f1656739bfd4a553c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fc7aedc1c142c36d31487f1656739bfd4a553c2c"}]},"branch":"refs/heads/master"},"d4a6023de5bdfe5a6e9214579a35e083a45c1151":{"kind":"REWORK","_number":6,"created":"2019-10-02 20:26:14.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/682266/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/682266/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/682266/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/682266/6"}}},"commit":{"parents":[{"commit":"e860c69831289a800a1d7bb52e8621fc460f260b","subject":"Merge \"Fix validation of role assignment subtree list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e860c69831289a800a1d7bb52e8621fc460f260b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-16 02:52:12.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-10-02 20:26:05.000000000","tz":0},"subject":"Remove policy.v3cloudsample.json","message":"Remove policy.v3cloudsample.json\n\nWe\u0027ve make all the default policies keystone supports better by\nincorporating default roles and scope types. These changes have made\nthe ``policy.v3cloudsample.json`` file obsolete.\n\nLet\u0027s simply things for users, operators, and develpers by removing\nit.\n\nA follow-on patch will remove the test_v3_protection.py file since\nthose behaviors are passing all the protection tests with the default\npolicies in code.\n\nRelated-Bug: 1805880\nCloses-Bug: 1630434\nCloses-Bug: 1806762\nChange-Id: Ie45955f5cc54563cc9704d7cb2b656b5544ae030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d4a6023de5bdfe5a6e9214579a35e083a45c1151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d4a6023de5bdfe5a6e9214579a35e083a45c1151"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
