)]}'
{"id":"openstack%2Fkeystone~682447","triplet_id":"openstack%2Fkeystone~master~Id5e6f74a5ed96fdf17e2c241466897974aa5a218","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"Id5e6f74a5ed96fdf17e2c241466897974aa5a218","subject":"Use correct repo for initial version check","status":"MERGED","created":"2019-09-16 18:19:13.000000000","updated":"2019-09-18 10:48:23.000000000","submitted":"2019-09-18 10:46:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682447-1568803571793-fa3c3f25","meta_rev_id":"bb853911dd20b45404d746283f4c256dc360835c","_number":682447,"virtual_id_number":682447,"owner":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-18 10:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-09-16 19:38:16.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-09-17 09:15:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-09-17 07:28:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":2,"date":"2019-09-17 16:59:03.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-09-16 21:07:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2019-09-16 21:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2019-09-16 21:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},{"value":1,"date":"2019-09-16 20:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2019-09-16 20:34:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-17 09:04:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2019-09-17 16:59:03.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 19:38:16.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-09-16 20:34:54.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-16 20:50:44.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-09-16 21:07:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2019-09-16 21:29:48.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2019-09-16 21:39:10.000000000","updated_by":{"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},"reviewer":{"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},"state":"REVIEWER"},{"updated":"2019-09-17 07:28:18.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2019-09-17 09:04:42.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2019-09-17 09:15:20.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-09-17 16:59:03.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-18 10:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e41a5b494bf899ca0d5a4b0ffbd0e5a321400efb","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-16 18:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df2f036383eb06b89898cc28c527f4ac0a84d829","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-16 18:26:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79cdd19b5ab27fcd801dbd6a589e56264ac5dd4a","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-09-16 18:27:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec473eefd613ce33045b4cabd5779453c9e2c3e6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-16 19:38:16.000000000","message":"Patch Set 3: Code-Review+2\n\nWe should add tests but I\u0027m fine with adding them later. Manually verified the fix and OSA team reports https://review.opendev.org/682460 indicates this fixes the issue.","accounts_in_message":[],"_revision_number":3},{"id":"383d8d4f9edddaf15f98f831d82082d269eccdc1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-16 20:34:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"191cc93ab96a10be6228f8141690620da0686df7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-16 20:50:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ece78533bea422cb01e22e9f2ac429e2547e3b07","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-16 21:07:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"64db3c59393769fa9477e4ac8fcde062823c53cb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-16 21:25:38.000000000","message":"Patch Set 3:\n\n\u003e We should add tests but I\u0027m fine with adding them later. Manually\n \u003e verified the fix and OSA team reports https://review.opendev.org/682460\n \u003e indicates this fixes the issue.\n\nagree with Colleen Murphy, we need to fix the issue first to unblock the gates","accounts_in_message":[],"_revision_number":3},{"id":"11168170a434f09fb144c1c0c48be0bc80c9209b","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-09-16 21:29:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"53c4534a0c16a0bfef7a9e933dce6967534b6bef","author":{"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},"date":"2019-09-16 21:35:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bdf8724e401c552cc5ea4f00c869e9b915c15b76","author":{"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},"date":"2019-09-16 21:39:00.000000000","message":"Removed reviewer Maxim Ardasov with the following votes:\n\n* Code-Review+1 by Maxim Ardasov \u003cdok2d@yandex.ru\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"18be86f8c2070ba38b80c0b51990b55fef287cdf","author":{"_account_id":30984,"name":"Maxim Ardasov","email":"dok2d@yandex.ru","username":"Dok2d"},"date":"2019-09-16 21:39:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d4582fbfedbf5aec3bc495b6fffb72b0afd4a7fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:47:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8be7d7eb6dc4062aca0c4756d15729b : SUCCESS in 1h 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f558d29c76a480e9ac344a9f18655d5 : SUCCESS in 1h 01m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb2834cc0719455583322e4f5b11552b : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eac4fdcab6264b5e8443e5b11c8b8c40 : SUCCESS in 52m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8182f61c58594bddbc059aa508d617d2 : SUCCESS in 43m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17f713fb2a034d2b8db8a1d819612775 : SUCCESS in 44m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53854475c5ef4fe2bc1ab8a28928b20a : SUCCESS in 13m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c8c09a3aaf064348aa2c6a737599410a : SUCCESS in 1h 43m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3f1adf9bb8246ee8f7c70225c6e597f : SUCCESS in 59m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e1e299d669dc4403b0125e810b1debe0 : SUCCESS in 1h 04m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55804d4eea214b0c9e12ff8396871a31 : SUCCESS in 1h 23m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb99dd8d6ae74e568735f62cc6a0d8e0 : SUCCESS in 12m 39s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/616411f7c5e848bbb30e945c2486a9ee : SUCCESS in 33m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/93adb1a7ce7f4b4081375b1f883f8d1b : SUCCESS in 32m 35s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c84675fbef3d4a3a8f8db66bdd9cba93 : RETRY_LIMIT in 4m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9d2af582d55f4defa09c08033f2f9b3f : RETRY_LIMIT in 4m 17s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/aeb4ef769511441aad533701f2689c07 : SUCCESS in 13m 18s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/742e8055ca5645898623c81e04406839 : SUCCESS in 1h 35m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cea6ef785ea54881bce91ec92d93b304 : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":3},{"id":"18f7c945efd5e4403d35434eafcaeecdf7f4d759","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2019-09-17 07:28:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"37f1285df7b3777f3abbc2ad82201dd27859f398","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-09-17 09:04:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c38642ce0e0830b029f96463525b0ece1f9ccb65","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-17 09:15:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"97e87c68f888cc8501f09d790b6cea7b7dd4f901","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-17 16:59:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db880e8d79cbf3cb9d0354644fd4532b7b3b8bf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:59:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a45cb38401810e8fbb7ead8ede17ef0b08728ea3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 02:28:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b54e6846c43a40c9965487c23a749004 : SUCCESS in 1h 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4110fb506b204a93b748ec760998b030 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/165d905985a44192ad3809fca161a0b9 : SUCCESS in 1h 08m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8329f06f86542f2b00cb79e23233504 : SUCCESS in 56m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64e8f84ea203463eb124ca35758485b6 : SUCCESS in 58m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e27acfb42264c20859911fbd3ad925c : SUCCESS in 16m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a28c67e34d994df6bffcf0ae3fbddaf9 : TIMED_OUT in 2h 05m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c58e0b0b0e034f919c87f7e64f971e4d : SUCCESS in 1h 18m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7d3eada7e5084b0a9f447c32c7fa632c : SUCCESS in 1h 01m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01e2600f31524a739db8e33a59ea386a : SUCCESS in 1h 35m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21e5140e8e6b46279d612f10ac23c9bc : SUCCESS in 12m 38s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/35e211ade3584b49843a0d4f133211ae : SUCCESS in 31m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2559460383ea41a9be601811bb6e395f : SUCCESS in 33m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1516c0f5a57040118d09556d6d1edbd7 : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":3},{"id":"57a05b657d5c6179a704458fef6a216cecbd9cf7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-18 02:30:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c798d6e5c5502544813958602de9e46da0096325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:14:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff691c8aed8141849e2347d25de30594 : SUCCESS in 58m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56d4942f85f54067941c54f8c88effd6 : SUCCESS in 1h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f3a3d767da435cb607db4f1d94988e : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc6c7734e0f24b8ebc64e4d8d72185f7 : SUCCESS in 1h 02m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76e6cb6ef4994ed3a7c2ac6801313375 : SUCCESS in 1h 02m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c36aff81e324f0386fc1fb8eb11d402 : SUCCESS in 41m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a125dab1a348b0a57d08a5f0331288 : SUCCESS in 13m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7db447bed8be40219e509e5626a133a2 : SUCCESS in 1h 48m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/21e9356458194fcfa0a31a4d6a427152 : SUCCESS in 56m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/054e91a1a12944d79100bbd85ce40c7c : SUCCESS in 1h 01m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3670f4a5d3b41f1b08cce5f7acfc7b4 : SUCCESS in 1h 34m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4e5d83b0e474f3f9f8bb17952bc856d : SUCCESS in 11m 01s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/97de8d65965b4ac694da333f260b76d7 : SUCCESS in 39m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ee09b98863ff42d4a4343033909c84fb : SUCCESS in 27m 48s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ccee4ef2ece0482db75c5f776acd54c5 : SUCCESS in 40m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e4dc6dd3ee7347ad8a3f3603864f4090 : SUCCESS in 41m 46s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/70b22ae232a9472ea932a576605fcef1 : SUCCESS in 19m 35s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1ce0760e30aa478a80dbc303e3bef71e : SUCCESS in 1h 40m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89f50b2f6f68499f8bb87fb3050be598 : SUCCESS in 1h 05m 48s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7ae55c222d594b45a255b726fa5bb5d3 : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"df6f94d1aef021d9f287f8a1f713d15aa0b38670","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-18 05:51:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"30585b5d920f88d31074a58c2d0ee0fcec4f6157","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:46:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72523f9735f44ae5b8e78ca4f72211c3 : SUCCESS in 17m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a2a9ee5498f49c88c83abb176eb7185 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01403c2b6de45b29b6be24506bd853f : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d68b1b5cc68640d893fc7830a98327bf : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e275fbdcf264483f82b3d833b215a02a : SUCCESS in 37m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/935522eb712943ac88fa9b12b5b1847d : SUCCESS in 20m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1dad6bf2959465296f61b979d2ed44e : SUCCESS in 14m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7503c3351c014db4a070028b48dc435a : SUCCESS in 1h 35m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/632f362bed3047b8bfa7aa01f0e71931 : SUCCESS in 59m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e012f88d9d0e48c8a0560a2d4e27246f : SUCCESS in 1h 01m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f308f487c70745a0aa936fa0bdb6a4a4 : SUCCESS in 1h 46m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fca86a64d1ce45a4bc30d43ebe625d90 : SUCCESS in 11m 37s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d2a4c7e89d648838c50e3d101d55d0e : SUCCESS in 31m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/287f8d7f3bf44df7a2bbe06cdab8cf26 : SUCCESS in 37m 46s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c9bb87b0fd0e49698ac69820686e581a : FAILURE in 12m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/98d5e325f4b14de59a6fd5247a11b75a : SUCCESS in 43m 53s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0837db97491f4d69880033f93b7b7bc3 : SUCCESS in 17m 39s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e0c72a2442a74438a5e7ec3136e4fb06 : SUCCESS in 1h 47m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ac57767e30f4428b4462aad2f23b5fc : SUCCESS in 1h 05m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7406aff34f1940658383f2246e0a888f : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d408ddae4e258db4203766818da35a54ad2ce750","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:36:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d0abccae4dd0dd2e714c3a71d53bdb5894aba3b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:46:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e3d6ca3b68f469191ad33fc330809fe : SUCCESS in 43m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c2895ef65224c93846643a2b841e308 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e651787c72d4ac3b2120bf1ab786bb0 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87736865cd914f3abc99f35183dc89e3 : SUCCESS in 34m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01eed51bce3640fc937eec5c1f839748 : SUCCESS in 22m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1431a74222348c680d02e4faf993df0 : SUCCESS in 14m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8051dba23e224a39b58abaa7498f111a : SUCCESS in 1h 35m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/01fcdcc73a8547948c94a448480be5ec : SUCCESS in 58m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d0c6077238e947e0ad0ad3a047400103 : SUCCESS in 53m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/725e812863d34ca5b7a60dcd2d50e5da : SUCCESS in 1h 17m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb31102e8a24d4e9ac0b15cabfc063c : SUCCESS in 14m 23s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/de51033ffdf84c8983ce39122cd5a88e : SUCCESS in 41m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ac1d212981a647be8d6e556d67074827 : SUCCESS in 34m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83375b6fda514f648b5916162c0791b0 : SUCCESS in 1h 14m 33s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/14149cf237794818a3b99ece0e06b5d6 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":3},{"id":"30acc8a3906eb64b27e1086880bae0e2c9bc0762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:46:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bb853911dd20b45404d746283f4c256dc360835c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:48:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df06cc5ab8e4404a50022f8f8e869c8 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b6e6096323d4048b12d7891b4fa4bea : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6bb14c0ff6c95bbce26ce8023ef621e408c111e6","revisions":{"51a013fe1bb05c2cbaad944442245662e50aa20f":{"kind":"REWORK","_number":1,"created":"2019-09-16 18:19:13.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/47/682447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/682447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/682447/1"}}},"commit":{"parents":[{"commit":"18e0080af3dcc0a96ff5d98aeb5f517080a35fb2","subject":"Merge \"Remove system Domain Config from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e0080af3dcc0a96ff5d98aeb5f517080a35fb2"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-09-16 18:15:25.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-09-16 18:15:25.000000000","tz":-420},"subject":"Use correct repo for initial version check","message":"Use correct repo for initial version check\n\nUse the actual repo for the initial version check of the db version when\nperforming `db_sync --check`. Previously we always used the Legacy Repo\nas the repo path was not populated in the `get_init_version` function.\n\nThis allowed the case where the new expand/contract/migrate repos\nmatched the legacy repo version number, the `db_sync --check` command\nwould assume that everything was properly up to date regardless of the\nversion for the expand/contract/migrate repositories. This was most\nnoticable in the case of a db not under version control.\n\nChange-Id: Id5e6f74a5ed96fdf17e2c241466897974aa5a218\ncloses-bug: #1844157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51a013fe1bb05c2cbaad944442245662e50aa20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51a013fe1bb05c2cbaad944442245662e50aa20f"}]},"branch":"refs/heads/master"},"17f84fcfb25121b705fefe062378fdfa70f6af64":{"kind":"REWORK","_number":2,"created":"2019-09-16 18:26:03.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/47/682447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/682447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/682447/2"}}},"commit":{"parents":[{"commit":"18e0080af3dcc0a96ff5d98aeb5f517080a35fb2","subject":"Merge \"Remove system Domain Config from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e0080af3dcc0a96ff5d98aeb5f517080a35fb2"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-09-16 18:15:25.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-09-16 18:26:00.000000000","tz":-420},"subject":"Use correct repo for initial version check","message":"Use correct repo for initial version check\n\nUse the actual repo for the initial version check of the db version when\nperforming `db_sync --check`. Previously we always used the Legacy Repo\nas the repo path was not populated in the `get_init_version` function.\n\nThis allowed the case where the new expand/contract/migrate repos\nmatched the legacy repo version number, the `db_sync --check` command\nwould assume that everything was properly up to date regardless of the\nversion for the expand/contract/migrate repositories. This was most\nnoticable in the case of a db not under version control.\n\nChange-Id: Id5e6f74a5ed96fdf17e2c241466897974aa5a218\ncloses-bug: #1844157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/17f84fcfb25121b705fefe062378fdfa70f6af64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/17f84fcfb25121b705fefe062378fdfa70f6af64"}]},"branch":"refs/heads/master"},"6bb14c0ff6c95bbce26ce8023ef621e408c111e6":{"kind":"REWORK","_number":3,"created":"2019-09-16 18:27:59.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/47/682447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/682447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/682447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/682447/3"}}},"commit":{"parents":[{"commit":"18e0080af3dcc0a96ff5d98aeb5f517080a35fb2","subject":"Merge \"Remove system Domain Config from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e0080af3dcc0a96ff5d98aeb5f517080a35fb2"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-09-16 18:15:25.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-09-16 18:27:57.000000000","tz":-420},"subject":"Use correct repo for initial version check","message":"Use correct repo for initial version check\n\nUse the actual repo for the initial version check of the db version when\nperforming `db_sync --check`. Previously we always used the Legacy Repo\nas the repo path was not populated in the `get_init_version` function.\n\nThis allowed the case where the new expand/contract/migrate repos\nmatched the legacy repo version number, the `db_sync --check` command\nwould assume that everything was properly up to date regardless of the\nversion for the expand/contract/migrate repositories. This was most\nnoticable in the case of a db not under version control.\n\nChange-Id: Id5e6f74a5ed96fdf17e2c241466897974aa5a218\ncloses-bug: #1844157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6bb14c0ff6c95bbce26ce8023ef621e408c111e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6bb14c0ff6c95bbce26ce8023ef621e408c111e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
