)]}'
{"id":"openstack%2Fkeystone~682931","triplet_id":"openstack%2Fkeystone~master~I3542c655522cc7e48fd53ea804bfd7198ab44e48","project":"openstack/keystone","branch":"master","topic":"ec2-fixes","hashtags":[],"change_id":"I3542c655522cc7e48fd53ea804bfd7198ab44e48","subject":"Move remaining protection tests","status":"MERGED","created":"2019-09-18 15:42:54.000000000","updated":"2019-09-20 13:33:06.000000000","submitted":"2019-09-20 13:31:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682931-1568986262613-f233d7c7","meta_rev_id":"efa71ef71d2ca637a79521fe373ad2d16bcac9ce","_number":682931,"virtual_id_number":682931,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-20 01:10:33.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-09-20 13:31:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-09-19 02:51:37.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-19 08:43:36.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-19 08:43:36.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 02:51:37.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-19 08:43:36.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-09-20 13:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5470f9bea8d631b293021766a38daefa158e4dfd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-18 15:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57d2c6eb97a7a43f2e84f8f3b9fdae51a6662261","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 01:50:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be6028f218d14efbb23edd3c5c4dc382 : SUCCESS in 15m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e39232c53a944d6a6aa7e851bdbbedc : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d95cedbe894f5b93f0663da5f3dde3 : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bda89cd110e24afcb74874bd4a277422 : SUCCESS in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22e79f0f541147b0bd703297608687d8 : FAILURE in 14m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6aa8289004dd486bab7ab9549eec538a : SUCCESS in 25m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec7151f2dea480fb80eda435dd14756 : SUCCESS in 14m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/508828fb26034414968b106920ba1a25 : SUCCESS in 1h 27m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/78d272381992426eaa6d513c73e5819a : SUCCESS in 59m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2b983f61e319436a9345abedd249824b : SUCCESS in 1h 01m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e5c62e5596c44a89ab0d99e7bbf0eb3 : SUCCESS in 1h 33m 01s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/57ca7f58dd374b20a98d0caa137880e3 : SUCCESS in 31m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0cd0c60f84964593b8908149543d34ac : SUCCESS in 32m 45s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/92c42221400c44d589fc61cb2246bd89 : SUCCESS in 39m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ca5eb057d7ef4099b8e6141a782bd089 : SUCCESS in 39m 55s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c8a42a6f9de2466ab81e26e146e64c40 : SUCCESS in 18m 44s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0d57f705997a4298a24df27dee90d588 : SUCCESS in 1h 51m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95cd5c8da99f4bc2b9a4f299a016bbd6 : SUCCESS in 1h 09m 52s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/09378a984fc04278aab6f052e15046c0 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3f5d41d73c7454cfa96e6e14942b689aab5dd35e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-19 02:20:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ea9f4018f4473af19f8293141ce7238fa9317f3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-19 02:51:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc8fb68fcb4467b5cbad936f80352b3892e24ce7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 05:44:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f706d560a01e4130847ff5285b96b341 : SUCCESS in 15m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a4759a480824710b8bfb605174c616e : SUCCESS in 24m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c452fba65ecf418c9b69f1ebdfaf02cf : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01582de238cd47e78b8ac8511056f732 : SUCCESS in 1h 18m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c69fdba6f434ca882c43e9a71356818 : SUCCESS in 18m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e71183c125764aabb468c8e5d207bc49 : SUCCESS in 20m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363b9fb5a41e49d9b499c014f951e16f : SUCCESS in 16m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7f8d1bedb4074e19b3f6e0f192a797cf : SUCCESS in 1h 48m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91560e84ed084cc79819020718a2d16f : SUCCESS in 1h 03m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/34b6f8aa94d44ffb9a08e124d48f71a8 : SUCCESS in 1h 09m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce551dacb6343818528ae7336c5253b : SUCCESS in 1h 34m 08s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/196fa9643bc5424a8bdd23eb5ba6c210 : SUCCESS in 1h 03m 47s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2e62af1e741e45e49910bd19977265b2 : SUCCESS in 52m 17s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0e66e8f2aff845208c1a2a5f6f6f7e95 : SUCCESS in 38m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c15813846c5641a28710b53be65ac584 : SUCCESS in 50m 24s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a41b9645ef344b3baa423c331fa299cb : SUCCESS in 18m 06s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/41d7e24f1adb4ea68202f8ee22b7b653 : SUCCESS in 1h 37m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7647b0d07042415595d4767024fe4e11 : SUCCESS in 1h 02m 32s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/cb2308179fa74469aa368881c2c6b913 : TIMED_OUT in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"16805d8a596953e1ea387dc63fea6c65961404b3","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-09-19 08:43:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1051c1f554ae3061c2aa1194721698f4b1be065a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 22:05:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36d2d662194948cdb61a178ed7fb60b2 : SUCCESS in 18m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddb2eb34314b4aebb66b955a1c3aaf3c : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673ee500c5524c91a1e65878c4b42c46 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7c57f13a74045438557e804f801f90a : SUCCESS in 16m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8fc2e2ecef24b96ac9dacbe793baccd : SUCCESS in 14m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/709d982928b8427c8d5cedc9d8365f8e : SUCCESS in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a134052dda974bb5a20fa735125e4988 : SUCCESS in 13m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/05a75ab718b54940ac75a8a95e66703e : SUCCESS in 1h 39m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/13b33ac0b479456dba9de9d05a07b1ad : SUCCESS in 58m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f0ee0780d684a1bb836e1e7a859fa4c : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb403f4d833d451883e5cd060ac20255 : SUCCESS in 1h 27m 41s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d890f52b928945d78b02ef2705a0c64a : SUCCESS in 33m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/15810904ae7646a6994492ab810ebe3c : SUCCESS in 36m 59s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/51b19f0d7a2e482db15b287cef2c8a55 : SUCCESS in 43m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/704d4843e50641f5836e696a450f6420 : SUCCESS in 44m 13s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/df521ea972054945b2a346e67e7bcc4f : SUCCESS in 16m 17s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c2ce9bcd5cfb46a88e07cbb93d0087f3 : SUCCESS in 1h 33m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee62d82acfb64873aa5a89501859c6bf : SUCCESS in 1h 07m 25s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d4374bc91d1c46e59a12cd6a03e1960d : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7536445c628e20e7c403c035dd03c939b734fdbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 22:06:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4314d727af34ca5f271f89ebcaafe577439d2bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 00:44:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5cc8672595e4b219b4183e355676d8c : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7b73c15e464a5bacca6fdbce870f5a : SUCCESS in 13m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17e09949c023473bb01d769246bce7af : SUCCESS in 17m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f3477371a384cd58f25cafaeff49508 : SUCCESS in 13m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/715182d9320e4551b3e8fa7145761524 : SUCCESS in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc61f502b8c1497a87c6632d353a213e : SUCCESS in 13m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3555446aea1d493da279ab7bc1aadef6 : SUCCESS in 1h 41m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4744ad9143144b7db097e8533300d6d0 : SUCCESS in 1h 16m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f1c7dc3effd1499d97829c33fc82c504 : SUCCESS in 1h 03m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e4e8ba18db14fe0b2b8a9b4d1977bb5 : SUCCESS in 1h 35m 20s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/24cadc6eabe34dea93a5a390bde3b21d : SUCCESS in 32m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d034dc862e6244cc885ad2fd81e9ced7 : SUCCESS in 36m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/247d53fef3bd40e18d2c06bded58d809 : SUCCESS in 1h 15m 58s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/26b7d7b4066f43e19dce618fe3196812 : TIMED_OUT in 41m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a67d870efa312b7b01e2965351858a681e22f5b2","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-20 01:10:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fc5a14e6a18951a4714417c407e0b07613c8c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 11:35:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5864eefc00854c728e965ac4f14ff3d6 : SUCCESS in 18m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6258dbf8afb4c78a20c219c079c3a6b : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b775bb623ad4c95a141f30d35f7211e : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dcb3a77600f40ec816cbd1d0cfd1af9 : SUCCESS in 14m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb115ad229244083a0e2a2da52b07c96 : SUCCESS in 16m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea38ec0f90574339b40a9ea8750e1c21 : SUCCESS in 17m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82356eaa50f9488d9b6962f09cede2fc : SUCCESS in 13m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/648f4f16fad34e9aa6a9ece02b3bf954 : SUCCESS in 1h 33m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c86e1fcf7ed042eb89dd061d887099cc : SUCCESS in 1h 00m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f5e1678c5dfd4d248c3ba7434e49398a : SUCCESS in 1h 04m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cba1a71fbf9f45ed8eae42461731db22 : SUCCESS in 1h 24m 09s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4cb883820bc43fa815a0746fefbcf32 : SUCCESS in 36m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ca312a58f79e4dec8ea43e48080f8a55 : SUCCESS in 36m 12s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/be7d9bffc53142fca133a5c5a66447ac : SUCCESS in 41m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0e986386cb0746f596d951d50d9e25b1 : SUCCESS in 40m 27s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fe9346c070fa4fff817074bebb8b14e6 : SUCCESS in 17m 07s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fb7cb82f1d3a4a2fb7a089231af99872 : SUCCESS in 1h 37m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54707362bcbd496884315ac844c45341 : SUCCESS in 1h 04m 18s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/be0954876e604b2580b44e9a246688eb : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cccdd0fd879f9ea179d751b66a7b679e7c111f71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 11:36:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60d52a33e73a623de67ca5bf3efb59f795b4e196","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:31:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff5edfc94ac898a123345d8f9b9d7f0cc3f8fd69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:31:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3208e9787b6466eb56c5885ef792a29 : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e569ca54d541b28a2999d2349d437f : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9fe591a10c34a05a064464a73747d35 : SUCCESS in 16m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2e5646b34c844c3a34c4dd30da2efe9 : SUCCESS in 16m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f9443f30b7741309eb709249d1b71d6 : SUCCESS in 14m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5cea56600c48a8b2a24536ebea55c0 : SUCCESS in 11m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/01dab230fc8a4d7991b4e9873d3dd04a : SUCCESS in 1h 47m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/68ae4de87886404485ab6b303159d8ee : SUCCESS in 1h 08m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5552625e80a3485e835f556368ee245e : SUCCESS in 1h 17m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fb64694e7b5477fb58de923d61fce9c : SUCCESS in 1h 30m 13s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d74c2315bff740e8891eacc54d444448 : SUCCESS in 36m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/906e2e67bb234bffb155f390bc431e85 : SUCCESS in 36m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7c3269d0af04c059c6c56b9be4fc6f6 : SUCCESS in 1h 08m 37s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/2508c455010548b2af926c5c07502c5e : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2b6deabea9c404ed7938259760bf3aca4b03091c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:33:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5307efe6fb0844368700f84594a452ed : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"063a8ac2716a3fc5bc10274153751b4b9d416fad","revisions":{"063a8ac2716a3fc5bc10274153751b4b9d416fad":{"kind":"REWORK","_number":1,"created":"2019-09-18 15:42:54.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/31/682931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/31/682931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/682931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/682931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/682931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/31/682931/1"}}},"commit":{"parents":[{"commit":"d8b49d802fa2dcfda97067b54b94a57ab2a35ee6","subject":"Merge \"Use correct repo for initial version check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d8b49d802fa2dcfda97067b54b94a57ab2a35ee6"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-18 15:40:39.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-18 15:40:39.000000000","tz":-420},"subject":"Move remaining protection tests","message":"Move remaining protection tests\n\nSince the protection tests were moved in 5e35efd5 some were apparently\nleft behind, perhaps during feature freeze havoc, so herd the\nstragglers.\n\nChange-Id: I3542c655522cc7e48fd53ea804bfd7198ab44e48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/063a8ac2716a3fc5bc10274153751b4b9d416fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/063a8ac2716a3fc5bc10274153751b4b9d416fad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
