)]}'
{"id":"openstack%2Fkeystone~683251","triplet_id":"openstack%2Fkeystone~master~I247478f998d7fb84c486cc04a3040517721e0dd3","project":"openstack/keystone","branch":"master","topic":"protection-zuul-timeout","hashtags":[],"change_id":"I247478f998d7fb84c486cc04a3040517721e0dd3","subject":"Readjust job timeouts","status":"MERGED","created":"2019-09-19 22:42:47.000000000","updated":"2019-09-20 18:24:26.000000000","submitted":"2019-09-20 18:19:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683251-1569003547945-fe8b4c94","meta_rev_id":"ed0b0c5335cd6166addecf6cfd615a42c9309c53","_number":683251,"virtual_id_number":683251,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-09-20 18:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2019-09-20 11:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-09-20 04:59:34.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-09-20 16:24:50.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-09-20 16:24:50.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 04:59:34.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-20 11:17:00.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-20 16:24:50.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-09-20 18:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"841dcb6b7436f654fb1963649dfa48f1744aef61","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-19 22:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"415614479c1f98037a24e04bf530b36622148c85","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-20 04:59:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e43609c6c5af37fe793e6b534bcddefd1fb66fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 09:23:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f44b76ce8cc469cb9c2ece138fe8c63 : SUCCESS in 22m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f02b5663f1c4fc48c212a5dcf7c6539 : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b2d3ccd532403597467e5904b1beb1 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30196bc3c2b04531871c9f09dda0ee89 : SUCCESS in 14m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/753d04b0e41c41ad9a3a3ac1fa24d087 : SUCCESS in 13m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f857aac847634f818374eee004e91cad : SUCCESS in 14m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9586466aa14473968da6100ead2184 : SUCCESS in 14m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ead43c3c1e3748ccb4614b962f126db8 : SUCCESS in 1h 40m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1685a71b09b148ed872eab77c60c3876 : SUCCESS in 1h 04m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ad717028d3542739e1b5c3af38d3f7f : SUCCESS in 1h 09m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1008affe81d24323ace0fbdd943d717c : SUCCESS in 1h 22m 23s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b1939f0855c48eebb5c6da6fa0675b8 : SUCCESS in 37m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/900f29a9cfa8416393919a276e848f97 : SUCCESS in 37m 14s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0c44ca449bba4f7eb4b7503559ea082d : SUCCESS in 47m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/de33b2591f3c46f7b03a6b7d3afe624b : SUCCESS in 41m 08s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c48c75257dd54125ac47a522b4a94389 : SUCCESS in 16m 34s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9c9901c61ad0419e92101fde8143a78d : SUCCESS in 1h 50m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e86e886b471d4f7a95273ef8cc097679 : SUCCESS in 1h 22m 14s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/dabda19c049d401397380f7362a2360d : SUCCESS in 29m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e91e00aadebd3126ab6a8ae872ed1ed8cbf22183","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-20 11:17:00.000000000","message":"Patch Set 1: Code-Review+1\n\nAgreed totally separating the protection test cases from normal unit tests and due to the high density of protection tests it requires more time.","accounts_in_message":[],"_revision_number":1},{"id":"54247908231da71874dff46d545a226c213a4378","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-09-20 16:24:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98e29f78d61dd5d49a0808fec5fa8fc43074ff30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:29:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d365b610627b945a89073096ec8b621735c2a2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 18:19:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63adeab4b58c4ee3815bc9e15908e1cd : SUCCESS in 23m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d73340753ae440e684111fb756403896 : SUCCESS in 7m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e891b3277ad843bbb487b7420b3004da : SUCCESS in 19m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51b666b80f6f4ecdbc61481bf488d005 : SUCCESS in 18m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36761bec4ed8446bbad78158fc49e86a : SUCCESS in 17m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b4b218465447c582ec4521b1007382 : SUCCESS in 14m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d531fa62096f4cc4b4ebbaaa48d80b7e : SUCCESS in 1h 42m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3edb449b17a4471aa5aa86b088729249 : SUCCESS in 58m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f203699e83e3401d877e26f8a316bf45 : SUCCESS in 1h 09m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be7405e5cf86426b84efe854e4e4b656 : SUCCESS in 1h 27m 36s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9ba7cb5211b46a897555bd5d51f77bc : SUCCESS in 36m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/508649a1463e41f6a55b7cde502d480b : SUCCESS in 32m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c1a0f3039ca4fe7bcd99ee98746292b : SUCCESS in 1h 11m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/693e5b6497a948a1af3f8f2fa0fc747a : SUCCESS in 42m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8ff03416c571a9dc1b4dd1f3301f3ee96c64bd14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 18:19:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"12a0873435b4515231439927ad42e55122b30273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 18:24:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a1785229f54b7093390031dc4f0e71 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"013c18d329a8e39b6caa0d3a26404f2320bedfae","revisions":{"013c18d329a8e39b6caa0d3a26404f2320bedfae":{"kind":"REWORK","_number":1,"created":"2019-09-19 22:42:47.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/51/683251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/683251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/683251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/683251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/683251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/683251/1"}}},"commit":{"parents":[{"commit":"d8b49d802fa2dcfda97067b54b94a57ab2a35ee6","subject":"Merge \"Use correct repo for initial version check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d8b49d802fa2dcfda97067b54b94a57ab2a35ee6"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-19 22:41:25.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-19 22:41:25.000000000","tz":-420},"subject":"Readjust job timeouts","message":"Readjust job timeouts\n\nWith the protection tests split out, the regular unit tests mostly\nshouldn\u0027t reach the 40 minute timeout limit unless they are extremely\nunlucky, so remove those overrides.\n\nThe protection tests themselves are still extremely expensive even when\nsplit out from the rest of the tests, so bump their timeout to an hour.\n\nChange-Id: I247478f998d7fb84c486cc04a3040517721e0dd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/013c18d329a8e39b6caa0d3a26404f2320bedfae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/013c18d329a8e39b6caa0d3a26404f2320bedfae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
