)]}'
{"id":"openstack%2Fkeystone~688601","triplet_id":"openstack%2Fkeystone~master~I35e871992dd21e96e7e812e440e6cafdb61d26ac","project":"openstack/keystone","branch":"master","topic":"drop-python2-support","hashtags":[],"change_id":"I35e871992dd21e96e7e812e440e6cafdb61d26ac","subject":"Stop testing Python 2","status":"MERGED","created":"2019-10-15 02:05:57.000000000","updated":"2019-11-14 10:32:27.000000000","submitted":"2019-11-14 10:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688601-1573727332189-bdb6f180","meta_rev_id":"aa1c95ee59804c9042d7eb544e6f75e484083406","_number":688601,"virtual_id_number":688601,"owner":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":24454,"name":"suguangfeng","email":"suguangfeng@inspur.com","username":"suguangfeng"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 10:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-11-12 15:56:40.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":24454,"name":"suguangfeng","email":"suguangfeng@inspur.com","username":"suguangfeng"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-11-13 19:37:23.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-11-13 14:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2019-11-13 15:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":2,"date":"2019-11-13 16:45:02.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2019-11-13 19:56:07.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":24454,"name":"suguangfeng","email":"suguangfeng@inspur.com","username":"suguangfeng"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24454,"name":"suguangfeng","email":"suguangfeng@inspur.com","username":"suguangfeng"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 05:53:32.000000000","updated_by":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"reviewer":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"state":"REVIEWER"},{"updated":"2019-10-15 05:53:39.000000000","updated_by":{"_account_id":24454,"name":"suguangfeng","email":"suguangfeng@inspur.com","username":"suguangfeng"},"reviewer":{"_account_id":24454,"name":"suguangfeng","email":"suguangfeng@inspur.com","username":"suguangfeng"},"state":"REVIEWER"},{"updated":"2019-10-15 05:53:46.000000000","updated_by":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"reviewer":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"state":"REVIEWER"},{"updated":"2019-10-15 05:53:54.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2019-10-15 05:53:58.000000000","updated_by":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"reviewer":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"state":"REVIEWER"},{"updated":"2019-10-15 05:54:14.000000000","updated_by":{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},"reviewer":{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},"state":"REVIEWER"},{"updated":"2019-10-15 05:55:02.000000000","updated_by":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"reviewer":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"state":"REVIEWER"},{"updated":"2019-10-15 05:55:17.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-10-15 05:55:35.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-10-15 05:55:49.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2019-10-15 05:56:18.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-10-29 19:33:51.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-10-30 11:33:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-31 08:34:55.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-10-31 11:11:35.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-11-13 14:13:43.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-13 15:25:12.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-13 16:45:02.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-11-13 19:37:23.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-11-13 19:56:07.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-11-14 10:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73f9de4d78c2357b38e1e9711312ec8de56023ca","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-15 02:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39bed4fbed6211edc14b9c1b1241d341634596be","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-15 02:06:55.000000000","message":"Topic set to drop-python2-support","accounts_in_message":[],"_revision_number":1},{"id":"48ecd18e8c6fb9d4c1d4c86db0ee86c36f864be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 03:48:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ddab38235414616a864066a8b97ac16 : SUCCESS in 2m 46s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce673c7a0f5943e3bcb701849906cdb8 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b84f0104b5d44f29bfb0731c19071c51 : SUCCESS in 14m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3554030d40e4a199fa607c954442b9a : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc58cb93e214e92bc70c28d67eb4c1c : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3435abfbd0314f5189771245464299d3 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5007b2728a44abf99e485b08eb3f972 : SUCCESS in 12m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97d3502b59f14f4aa9818ad78dd22f5f : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6795eeba78eb439aa87e3089e4aa7b1a : SUCCESS in 10m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2fb40c0b759440378d8dae9ef5b3114b : SUCCESS in 1h 36m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5247c7a3e7e741e684d41d79d59cf4df : SUCCESS in 1h 02m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6729f05babb3416d9d68a3bb0f01f4c8 : SUCCESS in 58m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc304515c28947da86f389eed5d4a154 : SUCCESS in 1h 21m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c040b7456e644dfb54f490b89fefbd2 : SUCCESS in 18m 04s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/c03e28f4c8f4418babbb7aef889bc345 : SUCCESS in 30m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/69f4721edc1146b4b1d8641ddc715a01 : SUCCESS in 35m 19s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f4c4a11512144fd1aa6295260f7a5749 : SUCCESS in 36m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/eeaceb18a4ac478594437f0158221132 : SUCCESS in 41m 17s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b52dd118a77e4ece817e9b541d2110f0 : SUCCESS in 17m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1ceceb62e5844c3c9c262cf1cc2dbefb : SUCCESS in 42m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23b2ffe410ae4bd88755d9a854cfc115 : SUCCESS in 1h 02m 25s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d0e9d7a3b62f42a6985fb63af13f2005 : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c72966498962c9443a8835807246529e5f6409a0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-15 16:57:00.000000000","message":"Patch Set 1: Code-Review-2\n\n(2 comments)\n\nDropping python2 support from keystone needs to be coordinated with the rest of the OpenStack community, especially given keystone\u0027s central role in an OpenStack deployment. The TC has outlined a draft plan and will be having a meeting on this next week:\n\nhttps://etherpad.openstack.org/p/drop-python2-support\n\nI\u0027m placing an administrative -2 on this until that meeting happens and we get clarification on the coordinated community-wide plan. In the mean time, I have left -1-worthy comments inline on the commit message that could be addressed.","accounts_in_message":[],"_revision_number":1},{"id":"8e3aea07f1917f0bd7c9d0fd5ee7dc650a64a71a","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-16 02:56:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"e93594ab18bd57fe443f79469411001696149e10","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-16 05:44:31.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"90b24479401f37a8277d8b890e1e3d0bb548fd85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:29:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74a61273347f468f96023fa292d59c1c : SUCCESS in 3m 29s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49b4fa68e1d64a3bb5ef05d6d1da7f45 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ba94937ce3949089efe4a2a687b446e : SUCCESS in 22m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0272bf1643f414f9446cb18520ba24d : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b44e9e2dfb441895597dcce0620706 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64d045532c1a4a7daa56bb06a6b751ca : SUCCESS in 21m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ad7235525754fd295ea250252d33ddf : SUCCESS in 22m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d33adb2614b34176b8851bac81900bc5 : SUCCESS in 20m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe66343c9f448f5938331876923db73 : SUCCESS in 11m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3532abc9719447c9a02e2771b933b930 : SUCCESS in 1h 42m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0e4172f255cd4489905767029662b7cf : SUCCESS in 1h 04m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9e5df79ba5d24ff2976f96363aa94946 : SUCCESS in 57m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a87cd0685e4c4f1286ceedaaaf63f84b : SUCCESS in 1h 26m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d1eb3dacacf400d83fa73b685b25075 : SUCCESS in 16m 25s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/fdd660edf4a243b78e9c6ac1305a164f : SUCCESS in 33m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/05c838e045764233984da88d63c37198 : SUCCESS in 33m 47s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3256a91369454c19a951ded26a54057e : SUCCESS in 40m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/492fdb999d6841c194d01e8da7c2cabd : SUCCESS in 35m 01s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/29f5ddad2c974c40a2912659b9899a43 : SUCCESS in 12m 57s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d7678f94864b4d5286125751bee9df34 : SUCCESS in 36m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb34710bab7430296800a551acb26c5 : SUCCESS in 1h 05m 48s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f6b82eeb06cf48e29a32e9b42f09fa10 : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b39648a8128b3463d293847a0e3c76cfb70bd5bd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-24 20:52:19.000000000","message":"Patch Set 1: Code-Review-1\n\nThere\u0027s now a coordinated plan and it looks like we should go ahead and drop the python2 jobs: http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n\nCould you update this change to actually drop those jobs in .zuul.yaml, and modify the commit message to reference that email and not plagiarize Stephen\u0027s commit message?","accounts_in_message":[],"_revision_number":1},{"id":"31791dca82730547de03704e5152939ce4881a17","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-25 07:28:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d54be789cd11e06e2d4176998650a713f9c39794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 09:50:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/febb5636b99649ccb42c92e542bc8ddc : SUCCESS in 4m 02s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23c1797c866c451bbd10bd04d80cb2d3 : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b207e76958564030845359a9d65d3b9d : SUCCESS in 17m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cdc5c826353457d8f16c9e7d69ea4ca : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0f74d978c947a3b0e0059261484d9e : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d82e731a1d54f5983599440ea9170e7 : SUCCESS in 18m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8972e8b99ad04503a9687249be12f873 : SUCCESS in 20m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ab2fd53d48465ab63318e891225b24 : SUCCESS in 15m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf1b30736e7f44e8ae8fc903459ef5fa : SUCCESS in 1h 33m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8736d35265e3459b97809ea253ff7e69 : SUCCESS in 58m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8481e6be3a2b42fe8888f2cc1dd5ce71 : SUCCESS in 1h 08m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6696a796c004b119234f4e7cc25125a : SUCCESS in 1h 30m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d103c9f3522e439f8d20b2ce4822fd52 : SUCCESS in 20m 24s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebb6bdca9af44e8888e62ec3fe8adefb : SUCCESS in 32m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f3310969547f4e28a7bf2c0259637434 : SUCCESS in 48m 07s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e14100fef8ce45fdb42853fb9244dd33 : SUCCESS in 44m 40s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6f4be1088cf84c89a984dfe97029b21d : SUCCESS in 39m 25s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e646f1e272de42c1b14ebd44c6deb09f : SUCCESS in 17m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f91fb1bd30e94309a69d8fef23d1eae4 : SUCCESS in 43m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dd5d1e64c964878bc9386bd25d53c8d : SUCCESS in 1h 08m 20s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7c8d972429b44fe48d9a44dd73d11bf8 : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":2},{"id":"fa15e313f7f446edb419d137a1b6c67f96141bb9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-29 19:22:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nStill some functional tests using python2","accounts_in_message":[],"_revision_number":2},{"id":"f42d5cafbc2159a204525b964f7ec7de902b3760","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-29 19:33:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nalso can you add a releasenotes also\n\nPython 2.7 support has been dropped. Last release of Keystone to support\n   py2.7 is OpenStack Train. The minimum version of Python now\n   supported by Keystone is Python 3.6.","accounts_in_message":[],"_revision_number":2},{"id":"b5975002c178b2b864132950738cbea3503f33cd","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-30 02:59:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"afe014de7d0fae247c095cf619a59f54d97ecb80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 02:59:50.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/keystone on branch master.  The error was:\n\n  found undefined alias \u0027irrelevant-files\u0027\n  in \"openstack/keystone/.zuul.yaml@master\", line 189, column 31","accounts_in_message":[],"_revision_number":3},{"id":"256bd2a42a873a75e7cb546de804f5c6f0b0919e","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-30 03:30:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ca20d21f34aeb3b7ee59fc3803888ff41a0a6a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 05:26:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b53799f5d2341fe9cec36edc0e2f9f2 : SUCCESS in 2m 43s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/edfe8a2d47554ad593bac1b897eda453 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c8f5f45fa8644e580558722d3a1f1f6 : SUCCESS in 19m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd4668df8f3648728e1446e1445e887e : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b4e13b034ac43a4938dc74695834e86 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94fbc331d42440f28a96878fe418684b : SUCCESS in 17m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02dd80859caf493ea26456c612b5a35d : SUCCESS in 19m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9cca7853104ba7b212db0d02df4c12 : SUCCESS in 11m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4fa0806d241e499b950b092fd5e329cb : SUCCESS in 54m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5decf15e1a9b40a1aa315712bd376730 : SUCCESS in 1h 25m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c322ec66e31944b596c45593a4f831f7 : SUCCESS in 16m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ad5dc9ff055f40c5b1050eee11beb618 : SUCCESS in 31m 29s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/47eef02ab56e4b7a90f5770704410179 : SUCCESS in 37m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/77e9e7ca51224e77a471c5c15d0b9bf4 : SUCCESS in 34m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/66602df318ea4798a932326077a85e10 : SUCCESS in 12m 45s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8dae46db09bd447c8d5f54daefbfd2f5 : SUCCESS in 40m 10s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2f332372e3734e82a8d74e9167d16823 : SUCCESS in 59m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d8c6d84f8a546559203b454715ebf8f : SUCCESS in 55m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/3daa1d8c4f23497195937581a09e3b93 : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d23af4c6ece6992b11710a0e800a357ff6e09534","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-30 06:15:09.000000000","message":"Patch Set 4: Code-Review-1\n\nStill no releasenotes.","accounts_in_message":[],"_revision_number":4},{"id":"0f86e56a63de3ebe9ccefe1ccedd84ca3cf26c0a","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-30 06:38:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0630bfc02f2c498a2c00749203d4eb92b71ef0a8","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-30 07:55:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nYou need to drop this [1] since python 2.7 support is removed.\n\n[1] https://github.com/openstack/keystone/blob/master/doc/requirements.txt#L5","accounts_in_message":[],"_revision_number":5},{"id":"6f5a3602a0fa3967bc8454a7cc034148f56bf775","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-30 08:40:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fa0a248f1c2e7d3eaf3594a6b70a7a61bcd2f7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 11:27:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b228a84b2ef84c21a1644070e086d2b5 : SUCCESS in 4m 04s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b847594e1ea04f3b945da3a1f98eb6e8 : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30d5f38486b54bcdb019eb788601d503 : SUCCESS in 21m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0701414fb7914c69be97b4bf741249cc : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6627e9a2bf54d21b62c75f2185eb247 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9c6871b603e41058e396296ad05451d : SUCCESS in 19m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1278ef5b7c5f4d628b79bff1afa5b03a : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0216998c8d084f7fb0501858dbc6d846 : SUCCESS in 11m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/153bc3c979724bc1a5c9843849ef2544 : FAILURE in 2m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/57c289b748d14ace859fda636b546bc5 : SUCCESS in 52m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d02d047f0d3440f99e56f97b7e16c036 : SUCCESS in 1h 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/560875769b89419f932230e96a8cb2f1 : SUCCESS in 15m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ac81fd3bbf624cab85106d7859f356cd : SUCCESS in 30m 58s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/25b0e174b3de4ff181d09dbe883696e8 : SUCCESS in 43m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/1033fe461a05448989d86ad71785a477 : SUCCESS in 35m 23s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/14855a3d366f40b6bea92dbf1d1aeaa3 : SUCCESS in 16m 57s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f83421d997064190b93fe5070cdafa28 : SUCCESS in 38m 27s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/71bb6758a438424ca60b33f86621887c : SUCCESS in 57m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99b8d30b7a934166bbba4dc564af1d81 : SUCCESS in 57m 08s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8181e85b8839428ebc6006f2bbeb1eee : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":6},{"id":"cb21588502cae56a37e6a88963d0ed0b87f1364d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-30 11:33:00.000000000","message":"Patch Set 6: Code-Review-1\n\nI think you need to remove the `six` usages too\n\nhttps://github.com/openstack/keystone/search?q\u003dsix\u0026unscoped_q\u003dsix\n\nThoughts?","accounts_in_message":[],"_revision_number":6},{"id":"895022434a49f4fcab5186ed2ac3df10dffb6000","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-30 12:08:47.000000000","message":"Patch Set 6:\n\n(6 comments)\n\n\u003e I think you need to remove the `six` usages too\n \u003e \n \u003e https://github.com/openstack/keystone/search?q\u003dsix\u0026unscoped_q\u003dsix\n \u003e \n \u003e Thoughts?\n\nSix is a Python 2 and 3 compatibility library, there is no need to remove it in this patch.\n\nConsdering performance, we could remove it in other Code Review Request, but not this.","accounts_in_message":[],"_revision_number":6},{"id":"5f035e89a1f078e1de0c96dfc21043ab0dce0115","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-30 15:49:36.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)\n\n\u003e (6 comments)\n \u003e \n \u003e \u003e I think you need to remove the `six` usages too\n \u003e \u003e\n \u003e \u003e https://github.com/openstack/keystone/search?q\u003dsix\u0026unscoped_q\u003dsix\n \u003e \u003e\n \u003e \u003e Thoughts?\n \u003e \n \u003e Six is a Python 2 and 3 compatibility library, there is no need to\n \u003e remove it in this patch.\n \u003e \n \u003e Consdering performance, we could remove it in other Code Review\n \u003e Request, but not this.\n\nI agree with this. If anything, that should happen separately so that this patch, which is just about dropping the CI, remains easy to review. Even as a separate patch, however, I would prefer we do not go through the entire codebase and rip out compatibility code, it\u0027s a waste of contributors\u0027 and reviewers\u0027 time.\n\nrequirements check needs to be fixed, some more minor notes inline.","accounts_in_message":[],"_revision_number":6},{"id":"96192425c20d976f18ffdf754e31793b0de430ba","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-31 06:19:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a6a6542d72662eb0144553f0c06c10e6d73087f0","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-31 08:34:55.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a803c6c13711f35bdc6f2c83f35e2bc4b60a97bc","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-31 11:11:35.000000000","message":"Patch Set 7: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"f1c83baebcd3d3fe5e46b9d22c25f9b16ef43ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:17:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9bae44f7d4a14562a106d6b0aa5641e1 : SUCCESS in 3m 30s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91ef00639de24c56be7627892601061e : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b88fbeccb1f6456399dcc6803450947a : SUCCESS in 26m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/015aa322692248f68f61409bf3f162f9 : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdb268d88124b72abf52573a4fc72b4 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/383aa55dc8b64d0aaf917fd7fb2b52a0 : SUCCESS in 19m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c76ad71700ff4f6e81ad4c02de68d287 : SUCCESS in 20m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6cb235cc864fd1b1492eccebe4f210 : SUCCESS in 12m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1f9d10c2b274680bfdd7b00b8ce2a72 : FAILURE in 3m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ad9c93f697348fb85e5d53a61fec01a : FAILURE in 1h 05m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52ae53be4bed456a8027f2aabacc0a85 : SUCCESS in 1h 26m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e96e377043b4b6fa9ee80967f4f8c0d : SUCCESS in 15m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2ce4bcaaa43445a0ad29a520483eab0b : SUCCESS in 29m 11s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/b3ba3ca129324cf2ab5fc29686bd47be : SUCCESS in 46m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/5e20e350ce4c4f4983eaa62e98b84c53 : SUCCESS in 37m 09s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c76bf101aba14b23a5934c0daf82eaba : SUCCESS in 18m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bb868d32894d46a88ba776b8f6cdc6d9 : SUCCESS in 49m 37s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/375a550fecea48498352a47c76f2852f : SUCCESS in 1h 05m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d977c6bf3db41f6973213874f6d6937 : SUCCESS in 56m 54s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b2c2992fdcad439ea5be6cf72221c2e3 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":7},{"id":"8ffc1c2bde59ca00637aea7f335d46456ba6d722","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-31 16:53:03.000000000","message":"Patch Set 7: Code-Review-1\n\nThe requirements check is failing because of the change in doc/requirements.txt, please undo it.","accounts_in_message":[],"_revision_number":7},{"id":"367807667b453709dc9de3108efe36c687b5183e","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-10-31 23:52:37.000000000","message":"Patch Set 7:\n\n(3 comments)\n\n\u003e The requirements check is failing because of the change in\n \u003e doc/requirements.txt, please undo it.\n\nThanks, i got it. \nIn fact, it is automactically maintained by openstack-requirements according https://pypi.org/project/openstack-requirements/","accounts_in_message":[],"_revision_number":7},{"id":"ba8bde28b1f4192edf08cc380f51132037bf916d","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-11-01 00:46:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f098420ec2f08801eb16b234a0a124e876e82a86","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-11-01 01:00:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c1d50a4c2ad7dd2400d74911dbcf83056b6819cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:09:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b9a8f1ece70c4f2f97ccd61c60469122 : SUCCESS in 3m 04s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/66dffa04f4144a8ea121bf0f27a19e64 : SUCCESS in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10306e3a4f624112b4b35383f99c42f6 : SUCCESS in 19m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c40ce5a11f4f0a8c3e2a74f06abf9d : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2575ae1a2d534b4e87d6ba8a6e2d3a96 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2713572ef0fb4bc7b5a1030c7ae45d61 : SUCCESS in 13m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a76c7dad2e94e4ea5de729121f5dc0b : SUCCESS in 20m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf3af3b3da0403ebdc5c4075462d0b8 : SUCCESS in 11m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/19f9b00836544aa08ab39801d8dabb4a : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b181474035e448e5bdf42c9575164f4c : SUCCESS in 1h 25m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8e4b757d4be4a45b98ae64d33307942 : SUCCESS in 16m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8df0a56bedf24d20af59df9d4b07e175 : SUCCESS in 29m 24s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/592a4935ea544c799ff7eebd0d3c0d29 : SUCCESS in 35m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/8cf1a55731dc4f878c2e0c3333e3c41b : SUCCESS in 35m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5f72d0e3446747afbce2b267d5b47ad5 : SUCCESS in 16m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d7eaabf11ce64ee4a46cdcd2a9a5ee26 : SUCCESS in 37m 46s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/867fcb26bec541a483fe78094ef7b110 : SUCCESS in 1h 01m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/813d2b17518a4a0fad44989bd32f1422 : SUCCESS in 1h 10m 24s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f882d21b2bae45fd85323506d44ccace : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":9},{"id":"28936099723f59ea7d63b58a52ac4d6a8d18dac4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-12 15:56:40.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":9},{"id":"6d6b8ad383a4776f87485c8f30054049c98daf7e","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 14:13:43.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"817c3d3f031a46bc898d9f89d79b6a6bdda00fd0","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 15:25:12.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"01bd695c034bbc7f043d556120a239d227bf7170","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-11-13 16:45:02.000000000","message":"Patch Set 9: Code-Review+2\n\nlets do this","accounts_in_message":[],"_revision_number":9},{"id":"987eb6b778a448140e00f7523afc839e2190d815","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-11-13 19:37:23.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"51ca555d5f151a40d7f00c469a62f54fd5f1d2b3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-13 19:56:07.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c4771b7c2771bfd49e9401c7ef6d3c03f0193c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 23:47:44.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1ee5c86506cf14150a96bc0e5e67fe050bd3e897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 03:31:50.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1026b6f3cb714599b901d41e51e91756 : SUCCESS in 3m 25s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1789bf322cbb4322b9c9dbd5c11caa62 : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86460590411b45d989338a46530a7f0f : SUCCESS in 22m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cde957b827f4ae491e1bb44cfa25293 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7cdafc36c04302957f6dea9ef625eb : SUCCESS in 20m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/504ea6e846e54a5398913fb3788ec4bc : SUCCESS in 20m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ea8f9632b243eea76d04a7d378715b : SUCCESS in 15m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0df911c72f584f62b3139a1702c81b8e : SUCCESS in 1h 09m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a75d9ba47e8c41f78bbbb48c6fb92efe : SUCCESS in 1h 41m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/730f466b60ba4c879caaff4aa648ea86 : SUCCESS in 22m 57s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/388acaef2c6f43579458bafdf7add678 : SUCCESS in 36m 50s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/179a19a44b9a46b18cb4e4b539a1507d : SUCCESS in 40m 32s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/ff7ba2cfe3a743ea852de23148e38844 : SUCCESS in 43m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca9186d78d924938ad7b0ff17f3077a4 : SUCCESS in 1h 59m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fb8e2f0e6e8c471c8a144e3765ee0884 : FAILURE in 1h 05m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a10e2a8516143849cdc441388d7aa80 : SUCCESS in 55m 39s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/486feeb1e7cb4b8faebd74bb8b29fb7b : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":9},{"id":"fd8fd07f8ee8d6879b5fad60119c53faec823ad6","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-11-14 03:35:39.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"92263e9246b0636036ae76ea9e72129531dae75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 04:57:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83ee3b31f6f145bdb80d7915c7397054 : SUCCESS in 2m 44s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce608903582a4395b3640bab5e83903d : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba35a64c0dab4cd79ce67f6fb59b7b84 : SUCCESS in 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c769cc67996941649d29ec6be3b7d9ce : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd9a607e87240918db6185c27b83453 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf9eb601c64a4563bbf9aaaf73426534 : SUCCESS in 11m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bc2adcb93c64384899cf44153ad2985 : SUCCESS in 21m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d090f15e29a447229ee4250dbce23997 : SUCCESS in 11m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a3c82cf64cd04952a639365c482ee7d7 : SUCCESS in 54m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5a900ca65d946f49a1d4339211496f2 : SUCCESS in 1h 18m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb9c3a369a24404d9a2e4d3da7858d42 : SUCCESS in 17m 50s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/63671f84d504469cbdd426c0d59fe4ef : SUCCESS in 32m 28s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5f301fd4e22640eea3f9d4a1c4afd309 : SUCCESS in 37m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/f6e1ddbb268241fb940c3c6ef0cef188 : SUCCESS in 38m 01s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/23d5bd8199904fe08ffda6bb617d8b6e : SUCCESS in 16m 12s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d2805c0adb0b41c59c3d3e4527b2c891 : SUCCESS in 49m 03s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6502c69ecd744cf09bac7986fe1b21f4 : SUCCESS in 55m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58feafd9c41a43cc912e1525a6dbc929 : SUCCESS in 1h 19m 18s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/5a4c6a9704a047e6a6d312d9b7a756ef : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":9},{"id":"ac4505cef2b45ed3fa1f1c429c02b0b8e11e84a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 04:58:07.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"454a9475b0281ead4c57078c18d05042bbe62678","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:28:51.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cb14401836664aadb25ebfc60e65669d : SUCCESS in 3m 09s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/170bf86d69b0408d9590f27ccdb9f2f5 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be669271c65f4b498bb6b46c0a0764e7 : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97ab85cd6624e5db229d89d7e35ced0 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead47cc8d24746b3be45628326a931b0 : SUCCESS in 19m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb56387275aa4c81bcc5a049960a701c : SUCCESS in 19m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7eb1e05754a48f399db49fc1250b629 : SUCCESS in 12m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3c16a921839542f7bfa4f168f0546685 : SUCCESS in 1h 01m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d63d64c31c2d4d59a72114ed7ed06d1a : SUCCESS in 1h 19m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/371dacef815f46938a7fd6056090c7e2 : SUCCESS in 15m 50s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/0450ccb3606d4e7e87fcdac42a6b39f6 : SUCCESS in 33m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b7433dcc58f941599d4054c86c1f1632 : SUCCESS in 31m 44s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/5e82cc7cdb904f3b8c60022e6c1b18a8 : SUCCESS in 34m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/19cdf92a873b4498ad83fb65f26055dc : SUCCESS in 1h 34m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/732de68c26da4b6fa494a9d53803d709 : SUCCESS in 1h 02m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe42867caecf41acacc2543a6eb75a93 : SUCCESS in 1h 03m 47s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b9da9f7f09564028ac9270ba5c9cc773 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":9},{"id":"b491e8f8ae9d6de31de5d28208a44d267aa31038","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:28:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"aa1c95ee59804c9042d7eb544e6f75e484083406","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:32:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3e6812f98a8f44118cbf365df3e2f0d0 : SUCCESS in 59s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8993a3534140afa7c637cdd7f57da5 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aad477f3dea94517aa1848ea933bf2ed : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"a92885a98b299f0d0c01e72deacf73f6efcabede","revisions":{"b03845b4121d4d55d2fe5c9a43f728c8a9772a48":{"kind":"REWORK","_number":1,"created":"2019-10-15 02:05:57.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/1"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 02:00:45.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIt\u0027s Ussuri. We can *finally* stop testing Python 2 [1].\nWe don\u0027t attempt any cleanup but simply stop testing with Python 2,\nindicating that we only support Python 3 via \u0027setup.cfg\u0027 and remove any\nPython 2 only dependencies. The \u0027tox.ini\u0027 is modified such that\n\u0027functional\u0027 now runs with \u0027python3\u0027\n\nThis should free up a significant amount of resources from the gate and\nlet us start using Python 3 idioms in our code.\n\n[1] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html#python2-deprecation-timeline\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b03845b4121d4d55d2fe5c9a43f728c8a9772a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b03845b4121d4d55d2fe5c9a43f728c8a9772a48"}]},"branch":"refs/heads/master"},"4905c54fe933b8af9b99c52630655a56d83994e8":{"kind":"REWORK","_number":2,"created":"2019-10-25 07:28:03.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/2"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-25 07:14:28.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4905c54fe933b8af9b99c52630655a56d83994e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4905c54fe933b8af9b99c52630655a56d83994e8"}]},"branch":"refs/heads/master"},"b9f4ca9b28aafa297d0f63f277319cf7aeb1f22c":{"kind":"REWORK","_number":3,"created":"2019-10-30 02:59:33.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/3"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-30 02:53:49.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b9f4ca9b28aafa297d0f63f277319cf7aeb1f22c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b9f4ca9b28aafa297d0f63f277319cf7aeb1f22c"}]},"branch":"refs/heads/master"},"317ce5f186a8f60b04e7bb676a273ad6727cddfa":{"kind":"REWORK","_number":4,"created":"2019-10-30 03:30:11.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/4"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-30 03:28:18.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/317ce5f186a8f60b04e7bb676a273ad6727cddfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/317ce5f186a8f60b04e7bb676a273ad6727cddfa"}]},"branch":"refs/heads/master"},"a9d8f0a1555bbc674697fd41ce92772d80cdea17":{"kind":"REWORK","_number":5,"created":"2019-10-30 06:38:43.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/5"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-30 06:31:25.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a9d8f0a1555bbc674697fd41ce92772d80cdea17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a9d8f0a1555bbc674697fd41ce92772d80cdea17"}]},"branch":"refs/heads/master"},"f0dfcbafe04f33592ca71114047915bc9c1c4c59":{"kind":"REWORK","_number":6,"created":"2019-10-30 08:40:28.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/6"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-30 08:23:13.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f0dfcbafe04f33592ca71114047915bc9c1c4c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f0dfcbafe04f33592ca71114047915bc9c1c4c59"}]},"branch":"refs/heads/master"},"2b77a921ed86b14727bb60640d0e1612f7e37530":{"kind":"REWORK","_number":7,"created":"2019-10-31 06:19:11.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/7"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-31 03:34:23.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1] and [2]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n[2] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b77a921ed86b14727bb60640d0e1612f7e37530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b77a921ed86b14727bb60640d0e1612f7e37530"}]},"branch":"refs/heads/master"},"a26bb0266f424c66d3f975830335745a63d75dda":{"kind":"REWORK","_number":8,"created":"2019-11-01 00:46:11.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/8"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-11-01 00:42:31.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1] and [2]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n[2] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a26bb0266f424c66d3f975830335745a63d75dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a26bb0266f424c66d3f975830335745a63d75dda"}]},"branch":"refs/heads/master"},"a92885a98b299f0d0c01e72deacf73f6efcabede":{"kind":"REWORK","_number":9,"created":"2019-11-01 01:00:58.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/01/688601/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/688601/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/688601/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/688601/9"}}},"commit":{"parents":[{"commit":"f9a086e16599123fba8a12c1c06bec73565a0ebc","subject":"Merge \"Switch to official Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9a086e16599123fba8a12c1c06bec73565a0ebc"}]}],"author":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-10-15 01:49:06.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"palagend@aliyun.com","date":"2019-11-01 00:57:08.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIn Ussuri, Drop support for Python 2 according to [1] and [2]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html\n[2] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html\n\nChange-Id: I35e871992dd21e96e7e812e440e6cafdb61d26ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a92885a98b299f0d0c01e72deacf73f6efcabede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a92885a98b299f0d0c01e72deacf73f6efcabede"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
