)]}'
{"id":"openstack%2Fkeystone~693838","triplet_id":"openstack%2Fkeystone~master~I398b151904c9c5c9d0c3ab4358074e8adcfd1b6c","project":"openstack/keystone","branch":"master","topic":"upstream/doc/fix-oidc-configuration","hashtags":[],"change_id":"I398b151904c9c5c9d0c3ab4358074e8adcfd1b6c","subject":"Update OIDC documentation to handle bearer access token flow","status":"MERGED","created":"2019-11-12 12:44:52.000000000","updated":"2020-01-29 15:43:43.000000000","submitted":"2020-01-29 15:40:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693838-1580312437639-46a14a57","meta_rev_id":"a66d419a5e7fd94b017e94d260bfbded801087bb","_number":693838,"virtual_id_number":693838,"owner":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"date":"2019-12-26 19:47:18.000000000","_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 15:40:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2020-01-29 14:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-01-28 23:38:54.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-01-10 16:27:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":1,"date":"2020-01-10 16:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2020-01-29 14:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 16:15:54.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-01-10 16:27:38.000000000","updated_by":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2020-01-28 23:38:54.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-01-29 14:28:44.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-01-29 15:40:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72b5fb1bdbd7c0bad92bf98999c90cc3c8a8a6d0","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2019-11-12 12:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6968d8bc4c3861d84758876c9201fc879d88a9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 12:56:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0ecc24e1e44a41b6b646af42c1c769 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0529ed35068c4c6ab8893e1e72c27beb : FAILURE in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6f5eb2387b80b8109046c46894135922034d847f","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2019-11-25 18:51:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9e4f0782bde2b13ccdab01ff21d19512e0a2eed","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2019-11-25 22:22:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef687625f0be6cf99b6e47e978360e9a3ca1fb42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 22:35:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84c5c7b181a4b72bbb010dbf967cb6c : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51903c7817d4035abb65316e68012c5 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ce97aad22e525e42ba23cac1711369fb7e6f7ede","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2019-11-25 22:55:26.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e (1 comment)\n\nThanks Rafael, I change the block to bash instead of dir.","accounts_in_message":[],"_revision_number":1},{"id":"e147f9e9b96a8011baa462b87fffd43f5ce4c314","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2019-11-25 22:56:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3daf394866b5b1ab7bc8a264c2f2656b671bf0c0","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-11-26 16:41:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThis looks really good. I haven\u0027t played with having multiple identity providers but looks correct.","accounts_in_message":[],"_revision_number":2},{"id":"c96a4370fff312487491ed172e39e22544a6a708","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2019-11-26 16:46:32.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e This looks really good. I haven\u0027t played with having multiple\n \u003e identity providers but looks correct.\n\nWe also were not aware of that. It took almost a full month of debugging, testing, and inspecting code. The documentation is rather scarce. That is why we have other PR with an improvement in Keystone Documentation.\n\nBTW, to make multiple IdPs to work, we need to patch Horizon, but that will be done in a PR shortly. The discovery process does not work as expected when we have multiple IdPs configured in Horizon.","accounts_in_message":[],"_revision_number":2},{"id":"2e0f8f580ecc72471d911fc69be9c1b6a97e16ef","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-23 20:27:13.000000000","message":"Patch Set 2:\n\n(14 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c49ba085231613083583f7601bdef00b3d833e3b","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2019-12-26 19:45:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2598efc7f47b6e461a70a57a408b58cb2e815b59","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2019-12-26 19:47:18.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ec4d695c437992c1ea699afd3c7d04cd29689f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 20:28:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20085b1dbcc48c2aa3a09ef689e7476 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb640266a4fe4935909f0ee0da324120 : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":3},{"id":"009c984ca9e0781cab233552c992764b97a2cae2","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-01-10 16:15:54.000000000","message":"Patch Set 3: Code-Review+1\n\nCould have used this a month ago! Aiming to get OIDC support in openstack-ansible-os_keystone with support for this flow, so it\u0027s nice to verify that I\u0027ve not been doing anything too stupid. I don\u0027t know whether its worth mentioning the Direct Access Grants in Keycloak somewhere because that could\u0027ve caught me out - but perhaps that\u0027s a little too Keycloak specific. Brilliant work.","accounts_in_message":[],"_revision_number":3},{"id":"9c5d998fc7125d87399fb6140e8f0728cf1a5381","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-01-10 16:27:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"de60688b0d38dfa23450af37a24f62dea5c0925d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-01-28 23:38:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6a9dd04d8271c4e3850a5b342f7e2aa644427bae","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-01-29 14:28:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3c5cee74c6957510778706c902de67055b3b796e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 14:29:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"42c94056b3043e061314e38c4c90577941742772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 15:40:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eaf79068617430eb7493cfa34bd933b : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3738781d26514bfc9db7075463e7e034 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":3},{"id":"aec1517be5365a73a71035498c671298f491caac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 15:40:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"272f421c2e598d68a83036ca3a9818bdb4dfbd6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 15:43:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42710708bb8348938d3b30c1b4b921dd : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01a8c1fcabcf9d64b48fc1ca4b425aa3ca80c670","revisions":{"f016dbca8e94e066d098845c915cf31ff89ee31e":{"kind":"REWORK","_number":1,"created":"2019-11-12 12:44:52.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/693838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/693838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/693838/1"}}},"commit":{"parents":[{"commit":"36aad6ff8191c0d891f6f0b8e87b322c851e3322","subject":"Merge \"Refresh \"how can I help?\" doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/36aad6ff8191c0d891f6f0b8e87b322c851e3322"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2019-11-11 17:34:19.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2019-11-12 12:43:05.000000000","tz":-180},"subject":"Update OIDC documentation to handle bearer access token flow","message":"Update OIDC documentation to handle bearer access token flow\n\nAlso add a section of multiple Identity Providers configuration.\n\nChange-Id: I398b151904c9c5c9d0c3ab4358074e8adcfd1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f016dbca8e94e066d098845c915cf31ff89ee31e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f016dbca8e94e066d098845c915cf31ff89ee31e"}]},"branch":"refs/heads/master"},"f51b7243699c6258a9e030378cc1a4734ee64b77":{"kind":"REWORK","_number":2,"created":"2019-11-25 22:22:47.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/693838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/693838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/693838/2"}}},"commit":{"parents":[{"commit":"36aad6ff8191c0d891f6f0b8e87b322c851e3322","subject":"Merge \"Refresh \"how can I help?\" doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/36aad6ff8191c0d891f6f0b8e87b322c851e3322"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2019-11-11 17:34:19.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2019-11-25 22:22:09.000000000","tz":-180},"subject":"Update OIDC documentation to handle bearer access token flow","message":"Update OIDC documentation to handle bearer access token flow\n\nAlso add a section of multiple Identity Providers configuration.\n\nChange-Id: I398b151904c9c5c9d0c3ab4358074e8adcfd1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f51b7243699c6258a9e030378cc1a4734ee64b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f51b7243699c6258a9e030378cc1a4734ee64b77"}]},"branch":"refs/heads/master"},"01a8c1fcabcf9d64b48fc1ca4b425aa3ca80c670":{"kind":"REWORK","_number":3,"created":"2019-12-26 19:45:37.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/38/693838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/693838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/693838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/693838/3"}}},"commit":{"parents":[{"commit":"36aad6ff8191c0d891f6f0b8e87b322c851e3322","subject":"Merge \"Refresh \"how can I help?\" doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/36aad6ff8191c0d891f6f0b8e87b322c851e3322"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2019-11-11 17:34:19.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2019-12-26 19:44:57.000000000","tz":-180},"subject":"Update OIDC documentation to handle bearer access token flow","message":"Update OIDC documentation to handle bearer access token flow\n\nAlso add a section of multiple Identity Providers configuration.\n\nChange-Id: I398b151904c9c5c9d0c3ab4358074e8adcfd1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01a8c1fcabcf9d64b48fc1ca4b425aa3ca80c670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01a8c1fcabcf9d64b48fc1ca4b425aa3ca80c670"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
