)]}'
{"id":"openstack%2Fkeystone~697611","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~I5dd85a6b8368373a27aef2942a64499d020662ef","project":"openstack/keystone","branch":"stable/train","topic":"credential-list-sec-stable/train","hashtags":[],"change_id":"I5dd85a6b8368373a27aef2942a64499d020662ef","subject":"Fix credential list for project members","status":"MERGED","created":"2019-12-06 02:57:02.000000000","updated":"2019-12-06 22:08:32.000000000","submitted":"2019-12-06 22:06:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697611-1575669983545-a71cd6a1","meta_rev_id":"c7c2947719c644cc4fc630ed8377aa662fa40ddb","_number":697611,"virtual_id_number":697611,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-06 18:38:03.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 22:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-12-06 15:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2019-12-06 09:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-12-06 16:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2019-12-06 16:15:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-12-06 16:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2019-12-06 15:44:49.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 09:35:13.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-12-06 15:44:49.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-12-06 15:46:31.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2019-12-06 16:00:24.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-12-06 16:15:51.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-12-06 22:06:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7425e15e1ce1b453abf2cada8a70fa4797cbc03b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-06 02:57:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9284037fa8d9cebb06caacd940488566f002e3d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:44:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6580812cee254c41a94c11edf6e0af00 : SUCCESS in 15m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77ed6b6c1fc242559d038b4cb5585a17 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fed785d9d44c65928e1f409a4f0710 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d06799b5efeb44b79d4affe9972ff9cf : SUCCESS in 16m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49f492289ea34ad7b49b36fd4954f76f : SUCCESS in 12m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d844d162394e4e3c9b7d739089c5e66c : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a93b0afa65cd462188711e6d5c5c54c5 : SUCCESS in 11m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/20bc9106e7294170a50818f66cfe4fe1 : SUCCESS in 1h 42m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a2e4d11c56db4aba9938f2050a056a97 : SUCCESS in 1h 00m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c849e1e356d45e6ac882a8c7942c7ac : SUCCESS in 1h 03m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e058879fa8e34a72803d8f3415c5b91c : SUCCESS in 1h 42m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c887bb1f26a42108b4f1efa90990868 : SUCCESS in 17m 02s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c35f8d120a6450fba4c135e7159e9d5 : SUCCESS in 45m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8145ccdcf6d14838a5a3e5986783ef5b : SUCCESS in 35m 11s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/98af8987f0ed43c19adb0109f5bad590 : FAILURE in 40m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/aad194b7d5ae43d28d4b95ac21764dd3 : FAILURE in 48m 40s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d555ff74ff5f4344b546b8bc4cc8c6d5 : FAILURE in 18m 11s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/abc138d24a1141618a715702797f1ddd : SUCCESS in 58m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/547e263dfdfd4efba847b4d30ab5d040 : SUCCESS in 1h 05m 49s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/79016179e67a459c81bf42da20c641ec : SUCCESS in 47m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9bf6829f7da91dd96ff9e13da654c939408b4379","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-06 09:35:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d3e0fe2a421215c43570e3ced171ba2d0c812b41","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2019-12-06 15:46:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb33c956500d6357a9f6562e046b31c1aebcc3bf","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-12-06 16:00:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecd1f5ee731dfd0072a7cee442b9e851ba378ea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 16:00:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02473a9b18c5e9425393305b96ee5c9ad510bc70","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-12-06 16:15:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6b624990745fdd6ef29a8efce1cf9831ef68e2fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:34:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27d1942d7dd54ca69eb863c08e31cb74 : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc73ea2e37514337a2539e0c8cd20879 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ad4d84f22124d8fb82c4afc280d660d : SUCCESS in 14m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38db1c89f2be4f7eb57cf6fc12ddefda : SUCCESS in 23m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19b7b4abfbbe4b51b6565ea0845c4fe1 : SUCCESS in 24m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c4595162c94946b63b59af44d2e4c6 : SUCCESS in 14m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/06e8f913ca254356a251c36b139747e4 : FAILURE in 1h 14m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/719ec814534c4a35891d471614b4d775 : SUCCESS in 54m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/101bf155adc4423dac02ae8d0115dbbb : SUCCESS in 52m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b52eb99386a4eec9304c7b1b2a83a3c : SUCCESS in 1h 35m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83ee3c4f58d14b63961b45deb01266e2 : SUCCESS in 19m 20s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/789cb9b3f36e49f8a4b8eec2600a24c0 : SUCCESS in 34m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d6674db85aca4868bc9ec0a13c7a1362 : SUCCESS in 36m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc4c857510564f52a859b2cad0c0b705 : SUCCESS in 1h 09m 15s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/93a0d9f27b0d4e31ab9f41b1c2af1951 : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"706fd59e49d4add98389c920424007e595aee2ac","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-06 18:38:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed04875a530dd6b12333c4fa189e61788a8ff3d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 20:26:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c47a1656f7454299b523b072565599e8 : SUCCESS in 15m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c570638ae3d40409904dea57319f327 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dde127a7878417cab91bc1aa06533a5 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc2ad80eea0e435ca6dc64c2e6e48de2 : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/851e18bfd57a410fb24e63ecbc969233 : SUCCESS in 13m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24e3fe72d7874a118b1e4275c856ed7d : SUCCESS in 13m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0381e90eee47d0a789790be515cab3 : SUCCESS in 13m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/470c47cd36e344438e6cbc71b31c1aa2 : SUCCESS in 1h 41m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b5dcf917b29c4cc39f7d158307d8c166 : SUCCESS in 1h 01m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/966d55d70e0b4327bd0b132d690281b3 : SUCCESS in 58m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da0c14577f82448a9a52450efb13352b : SUCCESS in 1h 21m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd2812cfc97e45a7b988f3f7d45933fd : SUCCESS in 18m 44s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e4fb66ab6284cd28027c5c02d26b68d : SUCCESS in 35m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3238d48596204e61b803b825734dbabf : SUCCESS in 34m 35s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cd6a16ce5c0844e9895b3e3827761924 : FAILURE in 39m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d57d6ba63f0d4867bb0f4f09a3e1ef8d : FAILURE in 41m 04s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d277319e71794b5eaea20d8aa9cd54ac : FAILURE in 19m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3df8545541b94a83a3309e7cd57c6910 : SUCCESS in 40m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a1c4932c29146a7984ff1672ab44eeb : SUCCESS in 1h 05m 37s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b5041c5835c842889ac7b23370f3ca3d : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9f6fc076eac871af689ee3da97f2de81f72236d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 20:27:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54bc30bcb6beab8214b322a3227d00439c02a60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 22:06:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c9e746a38994b30b370e6d60e6fb375 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4bdf0131494928a31901dce8b9e8bc : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f93d02ef53e948129d9d42ff6a640a8d : SUCCESS in 20m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0356184eddf04000be1159ee44aac753 : SUCCESS in 17m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73d6d87b699544a4bae70fd85b79a41a : SUCCESS in 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b57b091e4384a4bb443d42fa4621ab4 : SUCCESS in 13m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c1f3ada829ac4b36bd55a1b804ea051f : SUCCESS in 1h 30m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b356228926f04a4dbbf2e15dc811c57f : SUCCESS in 53m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b00324c121a64063ab04bf8d2c1f6e1f : SUCCESS in 52m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/440222e3c7884a44a235dc39c0f92d2c : SUCCESS in 1h 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/505b6cc3b3e54128be46bb513f0c9142 : SUCCESS in 17m 03s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/53f8a8d4343b4308881a9bf653378c6d : SUCCESS in 30m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/07ce9b9975c74df8ba9407e39ecb9947 : SUCCESS in 32m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/649cb425f16b45cf83ebf2111062e5d0 : SUCCESS in 58m 33s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/9f05fe7c594b4b2483a1c502fc6e3ce2 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fa3965f4e1fabca47c571ab8fe8bfd17f0dcab4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 22:06:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"47839b8c7b2ba891f0fe9c2dc6fc2b2cff83ae74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 22:08:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7291a2b35d4a4fac0c48cfa5c61e9e : SUCCESS in 1m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c3267a90e646eabcd0c9a2d211322c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd3f63787151183f4daa43578aa491856fefae5b","revisions":{"bd3f63787151183f4daa43578aa491856fefae5b":{"kind":"REWORK","_number":1,"created":"2019-12-06 02:57:02.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/11/697611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/11/697611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/11/697611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/11/697611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/11/697611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/11/697611/1"}}},"commit":{"parents":[{"commit":"d5f9c681fc04e10af29005ca2c0e9b498024c042","subject":"Switch to opensuse-15 nodeset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d5f9c681fc04e10af29005ca2c0e9b498024c042"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-12-04 18:51:05.000000000","tz":-480},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-12-06 02:57:02.000000000","tz":0},"subject":"Fix credential list for project members","message":"Fix credential list for project members\n\nWithout this patch, project members and readers can list any credentials\nwith the /v3/credentials API when enforce_scope is false. enforce_scope\nis only applicable to project admins due to the admin-ness problem[1],\nand this policy is not meant to allow project admins any access to users\u0027\ncredentials (only system admins should be able to access them). However,\nwhen enforce_scope is false, we need to preserve the old behavior of\nproject admins being able to list all credentials. This change mitigates\nthe problem by running the identity:get_credential policy check to\nfilter out credentials the user does not have access to. This will\nimpact performance.\n\nCloses-bug: #1855080\n\n[1] https://bugs.launchpad.net/keystone/+bug/968696\n\nChange-Id: I5dd85a6b8368373a27aef2942a64499d020662ef\n(cherry picked from commit 17c337dbdbfb9d548ad531c2ad0483c9bce5b98f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd3f63787151183f4daa43578aa491856fefae5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd3f63787151183f4daa43578aa491856fefae5b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
