)]}'
{"id":"openstack%2Fkeystone~699834","triplet_id":"openstack%2Fkeystone~stable%2Fstein~Iddc364d8c934b6e54d1e8c75b8b159faadbf865d","project":"openstack/keystone","branch":"stable/stein","hashtags":[],"change_id":"Iddc364d8c934b6e54d1e8c75b8b159faadbf865d","subject":"Ensure bootstrap handles multiple roles with the same name","status":"MERGED","created":"2019-12-18 18:46:54.000000000","updated":"2020-01-30 18:15:04.000000000","submitted":"2020-01-30 18:11:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699834-1580407899359-9ea6cd4b","meta_rev_id":"0d96f9c15bcbfcfeb9f53a37b3fbedbed86bb5c3","_number":699834,"virtual_id_number":699834,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-01-30 13:44:02.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 18:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2020-01-29 20:22:43.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-01-29 17:49:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":1,"date":"2020-01-29 20:22:43.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 18:46:54.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-01-29 17:49:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-01-29 20:22:43.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-01-30 18:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e179b24d2508e8fe3ed38edb1c4b2f93497af064","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-12-18 18:46:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc547ad0c7a1e7d5b503b5f8fddd9bb7d4af1043","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-12-18 18:52:46.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting until https://review.opendev.org/#/c/699743/ is merged.","accounts_in_message":[],"_revision_number":1},{"id":"29962ebd529c3c04361e993fce5ce2ae1fd7ee3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 20:08:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a012af5008784f21baf32e77e82140b4 : SUCCESS in 29m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6e7f424ad8e4dc4b04714989aae38b5 : SUCCESS in 39m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a04349b135743ad943f43aad0a5e99e : FAILURE in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3043c57625f54157b0d7de1098499253 : SUCCESS in 20m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4feb2451aaa747b8958e0c9cf72f73c3 : SUCCESS in 21m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2303c163f2a74e79817c9977e2cfde15 : SUCCESS in 25m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebeb9bbd4764eff9d602549b6e50d57 : SUCCESS in 9m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e77a0388333d403bbc0f81c67d9202a8 : SUCCESS in 1h 19m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97eabb3c19484685afd75395e45aadf8 : SUCCESS in 52m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8fee5c8f98a741acb16bc72e038b9272 : SUCCESS in 52m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/331473cae3a14e39bf052844e8dbc003 : SUCCESS in 1h 15m 42s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/084f432fdc6543458ad85954a9a022eb : SUCCESS in 31m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5c575d1996254b3589401c810edf4fca : SUCCESS in 31m 04s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/55bc052c8a60422f93e5d61373fbe758 : FAILURE in 33m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/069357bdcd584551a6b6b545aaf652b6 : FAILURE in 36m 40s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c529664ba7f344aa850bda8eceb810d9 : SUCCESS in 15m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cf05c21f450d4b6c8445fa3711a0b055 : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"860aa2540d6808ea0395d8a6874dc2b98c374828","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-09 13:54:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"055c156db76d77b2d111debe39c633a2c935fbae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 15:36:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37eb9bc8347f4e67983335ba2abb159e : SUCCESS in 29m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aae18d79be9146639958a8d0b867a967 : SUCCESS in 28m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7294a6061341cfaa2a0dc33ac63154 : FAILURE in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1878da3eba84b9d967688534db7c4b0 : SUCCESS in 24m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3acc27e9fae44578ed7ab730b3a01d9 : SUCCESS in 27m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b54aca99cfc34f84ae06e7f0113b8fab : SUCCESS in 25m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4bcbb62fda4444ad61845be403b74c : SUCCESS in 10m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/968d56342fc047a99a2d9a14f15d7bd4 : SUCCESS in 1h 25m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c7cdd43c97044e44947257ab68d39c6a : SUCCESS in 56m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/45f7721b24874d1ea8203c069b139ff1 : SUCCESS in 1h 02m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93d2943039db4df58eb9743b107ed31f : SUCCESS in 1h 34m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad01d2dfa91948b2ad6b80c663198e0d : SUCCESS in 21m 38s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/39a4d28cff274fd096d2772d0a0c7cc4 : SUCCESS in 38m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b6cfbb3646134831a998421891ec1b12 : SUCCESS in 37m 39s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9be6c3b0605d4f348ec79a428ad8cb64 : FAILURE in 37m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/4a67c621de0240a1a2e89cc2a32ccf37 : FAILURE in 40m 42s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/23e38ba433794d49b0345ce9396447d8 : SUCCESS in 21m 19s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/33fe34d8bffc4e17b4ea73ed503a4bed : SUCCESS in 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36f9870d959536b9de7808f12d917df022441b5a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-09 16:52:59.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"736593121ce2ce6fe38fe47b413ec197ebae0e57","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-28 20:08:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nIf the missing new line and the cherry-picked-from-commit line will be added then the backport would be OK.","accounts_in_message":[],"_revision_number":2},{"id":"2af6494068888ca607e02e64df1fead0eccfd269","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-29 17:38:43.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ae28b812f168a7a5881fdeeb29e6cf7551a1c9ff","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-29 17:39:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8135bea325946bb8cb5dcfd6eceb5ec89b6fb1e6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-29 17:49:50.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks for the update. The backport looks good, conflict is noted and patch merged in Train.","accounts_in_message":[],"_revision_number":3},{"id":"3269925f66f33436fc0d699cac7fc37bcd75f13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 20:15:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27b10b2c222d4a34b66479420b96c09d : SUCCESS in 36m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6896c362dec4d62be3e025243c5d30a : SUCCESS in 38m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f9bab67f26425a829156d62fa511df : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28266af3988448cfa920d6a3988cd0b6 : SUCCESS in 22m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7b84938f3404265833c16d7c228cfb1 : SUCCESS in 33m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68748bcdf5f44e2fb7abc72f98dc3577 : SUCCESS in 29m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14690e89a8d646c5b353226554d857c3 : SUCCESS in 11m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a7f5de58daa40638c6539e8f1fc9eb9 : SUCCESS in 1h 31m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3374aafaac2142059020edad1fe60708 : SUCCESS in 1h 01m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/88d6c839512346ee8b9725eab66f9f94 : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22fc060d3efd480f9f55db44d11aa9ab : SUCCESS in 1h 26m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/975280f33d9a46c7a6e41c7000d07b1f : SUCCESS in 21m 28s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd129b8859ab42bfa71bb7d53ec0711d : SUCCESS in 30m 50s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/738bae54092046b0a04c6ce276bf6856 : SUCCESS in 31m 36s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/eb637960649547a4907105c301ce4e04 : FAILURE in 34m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9e2cee0213cb4c299539313b98e89f74 : FAILURE in 36m 35s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d575700df8c04018a1764bd30a45096f : SUCCESS in 16m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d47ddb3fd6b94200bf2bf5724bb173ef : SUCCESS in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b90b293f80d95b1fa54d304ddb18522eedb0b9c8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-01-29 20:22:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nBackport already merged to train, patch proposed by keystone-stable-core so approving.","accounts_in_message":[],"_revision_number":3},{"id":"44b3772b3c02958d6acd1720f6d706422642acec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 20:23:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b3a88c035ce5422884bee60059ac27ce82bf8aae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 00:38:44.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b26728de54be4a3484cf02eb8e10e2d6 : TIMED_OUT in 41m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f62b51673140ed95925534f3485e47 : SUCCESS in 7m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6b2103fc1324e758c1cd7ccc7f21230 : SUCCESS in 35m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e23417c00e794ec598fd9fa8cf0fcefa : SUCCESS in 21m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d989095faa174e2d8d8166b723306eb3 : SUCCESS in 40m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc673c0d638422cba8637ebb7b2f0c7 : SUCCESS in 11m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5afadc2ae2e84b00a3a80784922b9a8d : SUCCESS in 1h 22m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/63d045264dc348b9833bb7f88fb7f29c : SUCCESS in 59m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04a85802088c47e0a00d3fcda8c06e16 : SUCCESS in 57m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92a5b3de89154054a9ba3bc1da1c8777 : SUCCESS in 1h 18m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2d3d6f87147466a824d599efe6a4070 : SUCCESS in 17m 27s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/19f1407174d6486c9e00dd7333994692 : SUCCESS in 37m 21s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a346385f5e9d4e0a94b934319142e3cd : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a0bb2bae81096c1ac0f9d7e3f4eb6cee7b9ee7b8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-01-30 02:36:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5806063dd258263ec9dd3876cf0593ce7829d467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 04:07:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6466a8c746764053b9fff53d4b919c3f : SUCCESS in 25m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/748aaf64f7f548539d1f6a20c3cacd0e : SUCCESS in 32m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62edb8ff93e246c58e349e3a7a570ff0 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6441570ccefd4c9788701874f63236ca : SUCCESS in 26m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28afaa3820d74b2988ae14a4250a607c : SUCCESS in 20m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd41d3b582164b16890f82a67d240692 : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0cf16398d08413c8b39df1a7861d6c2 : SUCCESS in 10m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c8cf42eaad414a5d9f376b2e6bc14924 : SUCCESS in 1h 26m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8ccc2af2a0484577836c207031ba9a96 : SUCCESS in 55m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ae7785ec784c4d29be8ef0188d92a28e : SUCCESS in 57m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee3a6f6e1e6744a69065cc275b901206 : SUCCESS in 1h 16m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbc19f9e53634b169c975f7b850474ce : SUCCESS in 17m 40s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/a05a29024ff74dbcbfac24a1d6d7519d : SUCCESS in 29m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/19bce6f201c143d38a91640f53120fee : SUCCESS in 31m 27s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/359fe2961eac45e684c3d5119fbe1855 : FAILURE in 34m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/df21f329163e4d54a8221f2713a4088f : FAILURE in 36m 11s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6475d83c94574b2ab1c6e028d5ffa6d1 : FAILURE in 16m 56s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4c1b9445a6aa4ad98c70a14d166fa0e6 : SUCCESS in 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"718980ee689a82ec3bb5c3dc2258363c785fcd3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 04:07:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"079d0780fd57758451f88da5f2b5dae15ef4351e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 06:13:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ab2b724c5304827ac763bdbf1e00396 : TIMED_OUT in 41m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be86c5180cdf495e9daeab899c1584f5 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb9c0052e16345d68a064c2a5b9bcd19 : SUCCESS in 35m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/275b676903be410d896baaba29fcb09c : SUCCESS in 39m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4d98d3c2dd144ea8e96a57375e19b1b : SUCCESS in 38m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7c242eb2334acc97660219e10d7082 : SUCCESS in 15m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/039c54f1a94d4002b33d3743d0790adb : SUCCESS in 1h 51m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bdb211e933f945099952671faa472aab : SUCCESS in 1h 13m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9330184bbbbb4c8992a2f4e07aed7aee : SUCCESS in 1h 13m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db470961c267498aa687c3adf1565a30 : SUCCESS in 1h 45m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8787db48efb43d7aef28e5f032adbbd : SUCCESS in 23m 58s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/90a09fd773814c0789834980a591b880 : SUCCESS in 41m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6dfa0132b6614183bef85083b916685b : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":3},{"id":"8febdff88e3b7bb04cd314bfbac9cb8975a13f3a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-01-30 06:34:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51eb05ec162c15113d41846bfc0080e02b2b2688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 08:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/defbee5eeeec44c08207d76a1db66668 : SUCCESS in 26m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43c6a2eacff145bfb9b617015422d7f1 : SUCCESS in 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8a7e0d5a4749bc81d1daba373e25ac : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dc6247016ee4ee0826f61bbab3f65e3 : SUCCESS in 19m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b83cfbef68045a49d1238b391d8b2a7 : SUCCESS in 22m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4612168b48ca4d06969efdd6bef058cd : SUCCESS in 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c49ec9a38f4438b8bf4c1ab902127e : SUCCESS in 12m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cf748257235f44439ad112a8134cf2a2 : SUCCESS in 1h 51m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ad34aa53391b4bacaf2f0c0c8d4f099d : SUCCESS in 1h 08m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b18ff6c25064f0f8c33b29f4137af59 : SUCCESS in 1h 12m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aac307d7e79e45fc819a160be8075b79 : SUCCESS in 1h 21m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b267c87060044b084ea8a2e436a7797 : SUCCESS in 18m 08s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/60dd235bf9324b878dc131e6f9b5b8e9 : SUCCESS in 37m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8afebae2ddf44296bcdc31492a588950 : SUCCESS in 32m 34s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/66c4baa34b0b41f4a359cf7e35a96528 : FAILURE in 35m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/654edb4ef75e4ae791af9904639685d1 : FAILURE in 37m 58s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bf3f22cf545a42e4a3d41743e1f0e004 : SUCCESS in 15m 45s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/55609d73a6734dc08d0b0b3295e838dd : SUCCESS in 39m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"64996c535361af4ac893d04f1f45b502c87caa73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 08:39:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8f946f1c461b8d2cc9de9753bfba58246b0cf32d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:27:21.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cf65c4534a24d93b5c6137bbf1f901f : TIMED_OUT in 42m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0a7be5720a4c98a7c87115d1887a53 : SUCCESS in 7m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b21f3f263d9048f9bf87139dc4e16906 : SUCCESS in 21m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80ccd5634ca94008bacf3b3b90a87a05 : SUCCESS in 22m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bed4e3c06a984439b24788669fd78999 : SUCCESS in 28m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d131587992424da28d8018df36090c : SUCCESS in 11m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cec2e7ad173e4099a953fbacb8bb3d97 : SUCCESS in 1h 36m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/899e3b3c77594b99803bef16c5ecc4b0 : SUCCESS in 57m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/139a56d9d41d4ed1af36a1a75f6d20a6 : SUCCESS in 59m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0589b4e93f5459086a54844ab173191 : SUCCESS in 1h 25m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17c33c0a6eab49979399aec9c884792d : SUCCESS in 23m 07s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/48ea86deb3f742b28721ff8fc5ba3559 : SUCCESS in 46m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d39657db55184407b26b0e6bd3a2e6c4 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":3},{"id":"116ed7d43b051494b1b845394ce73b6918d2f27f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-30 13:44:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"373a1e4e4e95238d17825feb6efa8a0451102d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:31:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e669ad7c785a42bd93dbe667c7e68c9c : SUCCESS in 34m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f521fac33f5475bb62741dcb9b75fde : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e63ea5330caa47ab844659fa07d1a299 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51ddd9af54cf4d86b2ad59b762d0138d : SUCCESS in 29m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aed4181ae2144c4b91c6cddfbdbd1f54 : SUCCESS in 30m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d248a5ced974ba6abc3418ba5df26a0 : SUCCESS in 29m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dba7563a86146998c68cc314309eb3d : SUCCESS in 9m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f3a6fdec45924c78b597f1212b136989 : SUCCESS in 1h 25m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b81aaace2484a20a4eaba9c4ee9fac8 : SUCCESS in 59m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8509bbe17864e93a2f6ec437b130a10 : SUCCESS in 1h 08m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ac5bf3c05d04f089f5e0f7cf3979f9a : SUCCESS in 1h 42m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23a68ed229b745e08aaecf827b4f52e9 : SUCCESS in 21m 28s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/579ff4e86db04422833d6fd3b8f6edb4 : SUCCESS in 36m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8f7ab2f216a1403ab0f30203c8023b2d : SUCCESS in 36m 22s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/11514eba12174bd58ba7ca1b6b1dd802 : FAILURE in 40m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5a794444c63144d7a69e2847bfc8fa31 : FAILURE in 39m 21s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7ab32da089d44402aa434da6303a6929 : SUCCESS in 23m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/320bbefca1fd4efab6ec23ebd1068a78 : SUCCESS in 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0ad33916a2f4064602d51e93cc02d8b797a1584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:31:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"635081aceb1785d6ad5c606853efd37f258c0775","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:11:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c20c46c0bd50addd60ae12548036e72ecf985ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:11:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd259f6e3ea7406dadf49ab54a28bc56 : SUCCESS in 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919f5d14fb0d435b9770b68651e2e639 : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3166372f01a4ab1ab2826d24fbc4f3a : SUCCESS in 25m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fedb84de8d34f109075bf98b830e4d6 : SUCCESS in 28m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/801e8319955741b286cfb8615162c41e : SUCCESS in 29m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecd7c83fa194581ac509e06d1d44710 : SUCCESS in 11m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a092295ff48d478684a2713be8dd1ad2 : SUCCESS in 1h 40m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5b4c61e4f4cf470484ce355233fb3ea6 : SUCCESS in 1h 09m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7d9f2e04435450fa232c6f38298b675 : SUCCESS in 1h 10m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/810a644fa05748bb99d07d4b0eb731d0 : SUCCESS in 1h 30m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d75b2496c224a378e147eeabf0f1356 : SUCCESS in 17m 55s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/1debf375cf5746c8b99e64630086309c : SUCCESS in 35m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/164f9c5506234fe2af162e686d38748b : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":3},{"id":"0d96f9c15bcbfcfeb9f53a37b3fbedbed86bb5c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:15:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8602d52ac8a34d539f17249e2efd7bd2 : SUCCESS in 2m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a174aa095ecc4b62b78ca37f0bc43f57 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1ba238e49195890c0232554005d4efa670467694","revisions":{"2c047f1a580c5dd05ac23902ccde2ae28dbd14e9":{"kind":"REWORK","_number":1,"created":"2019-12-18 18:46:54.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/34/699834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/699834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/699834/1"}}},"commit":{"parents":[{"commit":"5ad8a33d5a43c3d3e63f434c10223d9f42070e0e","subject":"Merge \"Docs: Make robust with using real links\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5ad8a33d5a43c3d3e63f434c10223d9f42070e0e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-12-18 17:59:53.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-12-18 18:45:55.000000000","tz":-360},"subject":"Ensure bootstrap handles multiple roles with the same name","message":"Ensure bootstrap handles multiple roles with the same name\n\nThe bootstrap logic doesn\u0027t take into consideration multiple roles\nwith the same name. If bootstrap is unable to determine which role to\nuse and accidentally uses a domain-specific role with the same name\nas a default role, bootstrap will fail in unexpected ways.\n\nConflicts:\n      keystone/tests/unit/test_cli.py\n\n      This change conflicted with another test that was added in later\n      releases. The resolution was to not include the additional test,\n      which isn\u0027t related to this change anyway.\n\nCloses-Bug: 1856881\nChange-Id: Iddc364d8c934b6e54d1e8c75b8b159faadbf865d\n(cherry picked from commit 0552bf9804adea7ab9b9b8b9485fd6cbff598260)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c047f1a580c5dd05ac23902ccde2ae28dbd14e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c047f1a580c5dd05ac23902ccde2ae28dbd14e9"}]},"branch":"refs/heads/stable/stein"},"ba014fd189025df9c7632cb31c58f5027a8336ed":{"kind":"REWORK","_number":2,"created":"2020-01-09 13:54:19.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/34/699834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/699834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/699834/2"}}},"commit":{"parents":[{"commit":"5ad8a33d5a43c3d3e63f434c10223d9f42070e0e","subject":"Merge \"Docs: Make robust with using real links\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5ad8a33d5a43c3d3e63f434c10223d9f42070e0e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-12-18 17:59:53.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-01-09 13:53:28.000000000","tz":-360},"subject":"Ensure bootstrap handles multiple roles with the same name","message":"Ensure bootstrap handles multiple roles with the same name\n\nThe bootstrap logic doesn\u0027t take into consideration multiple roles\nwith the same name. If bootstrap is unable to determine which role to\nuse and accidentally uses a domain-specific role with the same name\nas a default role, bootstrap will fail in unexpected ways.\n\nConflicts:\n      keystone/tests/unit/test_cli.py\n\n      This change conflicted with another test that was added in later\n      releases. The resolution was to not include the additional test,\n      which isn\u0027t related to this change anyway.\n\nCloses-Bug: 1856881\nChange-Id: Iddc364d8c934b6e54d1e8c75b8b159faadbf865d\n(cherry picked from commit 25cf359e5fb914b855922121f20e23bd14626b8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba014fd189025df9c7632cb31c58f5027a8336ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba014fd189025df9c7632cb31c58f5027a8336ed"}]},"branch":"refs/heads/stable/stein"},"1ba238e49195890c0232554005d4efa670467694":{"kind":"REWORK","_number":3,"created":"2020-01-29 17:39:28.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/34/699834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/699834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/699834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/699834/3"}}},"commit":{"parents":[{"commit":"5ad8a33d5a43c3d3e63f434c10223d9f42070e0e","subject":"Merge \"Docs: Make robust with using real links\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5ad8a33d5a43c3d3e63f434c10223d9f42070e0e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-12-18 17:59:53.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-01-29 17:39:09.000000000","tz":-360},"subject":"Ensure bootstrap handles multiple roles with the same name","message":"Ensure bootstrap handles multiple roles with the same name\n\nThe bootstrap logic doesn\u0027t take into consideration multiple roles\nwith the same name. If bootstrap is unable to determine which role to\nuse and accidentally uses a domain-specific role with the same name\nas a default role, bootstrap will fail in unexpected ways.\n\nConflicts:\n      keystone/tests/unit/test_cli.py\n      Conflict exists because stable/stein doesn\u0027t have\n      https://review.opendev.org/#/c/675228/ but it\u0027s unrelated to this\n      specific bug fix.\n\nCloses-Bug: 1856881\nChange-Id: Iddc364d8c934b6e54d1e8c75b8b159faadbf865d\n(cherry picked from commit 25cf359e5fb914b855922121f20e23bd14626b8e)\n(cherry picked from commit 51ff7be731450c183b3e3eb6d34493e986cc2635)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ba238e49195890c0232554005d4efa670467694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ba238e49195890c0232554005d4efa670467694"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
