)]}'
{"id":"openstack%2Fkeystone~704453","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~Icfce3b14abb55c6fef3de1b314cee22fc8b1d08c","project":"openstack/keystone","branch":"stable/train","topic":"assignment-role-filtering-stable/train","hashtags":[],"change_id":"Icfce3b14abb55c6fef3de1b314cee22fc8b1d08c","subject":"Fix role_assignments role.id filter","status":"MERGED","created":"2020-01-27 22:44:48.000000000","updated":"2020-01-29 03:56:33.000000000","submitted":"2020-01-29 03:52:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704453-1580269979632-ff1353d7","meta_rev_id":"75b1e1208e810cbeee19c40ee6f5be7dbe3044b5","_number":704453,"virtual_id_number":704453,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 03:52:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":2,"date":"2020-01-29 00:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2020-01-28 20:21:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-01-28 19:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":1,"date":"2020-01-29 00:14:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 19:02:10.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-01-28 20:21:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-01-29 00:14:11.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2020-01-29 03:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99df0dafd592b342406281d94f426febe10e886b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-01-27 22:44:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"664643ba36b5423bdfc34988145244f8b1af6c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 00:20:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feb1fdb769da44058a3a152442409fc8 : SUCCESS in 20m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2490b70bf2444ba9333afdd12b78b22 : SUCCESS in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbe9fe9506b4e5fbde0aa5d61f6d1d6 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1de68a0f5ef45038eadaf5ff3059938 : SUCCESS in 17m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ea05bab441c41ca8782be3678141bc1 : SUCCESS in 14m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92abc766146a428cbc61da77c2ed0a4c : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0d2be7e296442d8f0eb8eb12844d20 : SUCCESS in 12m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b86461c44b1648dba5c92c8753ba251e : SUCCESS in 1h 33m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d130fbda74cb4d96848561eaac98a7f3 : SUCCESS in 59m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7798e1bae71a436db8944c418b30110a : SUCCESS in 53m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51747498a73c42b594c3eea8eec73790 : SUCCESS in 1h 26m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03fefa0db6994dd3bf10d084681f13f5 : SUCCESS in 18m 35s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed01df5ec68c46c8aaa5cb1126403958 : SUCCESS in 35m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5498405f7b604fdb9f3ea8486375d8ff : SUCCESS in 33m 31s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/65e2e5603f5946939e250d5f2795ebc7 : FAILURE in 36m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fcc70678159541e79358063facc9d02f : FAILURE in 38m 57s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e1b34a596ab441fb9fbd20589946e4a7 : SUCCESS in 19m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a011a2aa210f491f90b91624d6e6a4be : SUCCESS in 43m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f72c83b88b94f44a7fddefdbd07abd7 : SUCCESS in 1h 05m 38s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/2da86c02a13148d792cc290077692f94 : SUCCESS in 39m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6a170da925d117d69604f4deec0daf4a383ac5e3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-28 19:02:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e0776c2b911d9b89f50f47cf19acd708e007f3c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-28 20:21:55.000000000","message":"Patch Set 1: Code-Review+1\n\nClean cherry-pick and merged in master. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"438fab1ea2f2d722e417d948265826804e949565","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2020-01-29 00:14:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nkeystone-stable-maint members\u0027 backports count as +2","accounts_in_message":[],"_revision_number":1},{"id":"944fae57d7dfff1c39269c39cb895a9a2ccdc4f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 00:15:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3f8af0f0cfb10e8595b5d9b2ac8f08ceb92050d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:52:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e99d7061eb7c437b97da6d4c28df9b3a : SUCCESS in 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e1a1e677bb4f16a5b3a478e13e6d05 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/768ab62412d142c7aadd03075eca65f5 : SUCCESS in 24m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e98da8e8b44cd0a1233c8bbd11fe3f : SUCCESS in 23m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df651190f9324144a49eb4f6e845329f : SUCCESS in 20m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c160ed7a36c54b6185f2854300729377 : SUCCESS in 13m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f3ec79231f84046aca717787692aace : SUCCESS in 1h 23m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/975c091513034c5ca13eb312840caabe : SUCCESS in 52m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/de31db6edc184b15b7a7edf59b6c79f6 : SUCCESS in 55m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c7555f0f9044b409c840c609f05972e : SUCCESS in 1h 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0834e14898a04aeea13dfdac124929dd : SUCCESS in 16m 55s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/dffe9e890286488496d198c2c5d58b47 : SUCCESS in 32m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/48413d9747fc46e5ba120c97347fff47 : SUCCESS in 31m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b061bc269efc4596882388cafee128ae : SUCCESS in 1h 12m 46s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/da9d8e48c19d4da1ac3c3311994b4502 : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":1},{"id":"974c273a42e0ee51ba045adc7d932fd3726ac88e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:52:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f55fd7dd8230b580c473dc472fc39149dd7a2aff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:56:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1b8a33ecde4e008218e76599728d4a : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c10de3e99e644e9eba1f57b69a19c714 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d413f1eba2d1e6b16ecd57fa27de528dd0f67cb","revisions":{"4d413f1eba2d1e6b16ecd57fa27de528dd0f67cb":{"kind":"REWORK","_number":1,"created":"2020-01-27 22:44:48.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/53/704453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/53/704453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/704453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/704453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/704453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/53/704453/1"}}},"commit":{"parents":[{"commit":"1b967e0aad133e8260acc2ebfd11699bc6a13b02","subject":"Merge \"Fix line-length PEP8 errors for c7fae97\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1b967e0aad133e8260acc2ebfd11699bc6a13b02"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-01-01 00:22:34.000000000","tz":-480},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-01-27 22:44:48.000000000","tz":0},"subject":"Fix role_assignments role.id filter","message":"Fix role_assignments role.id filter\n\nWithout this patch, if there are multiple role assignments on the system\nand they are not all the same role, querying for role assignments with\n/v3/role_assignments?role.id\u003d{role_id} may leak some role assignments\nthat don\u0027t match the role_id, making the returned results incorrect.\nThis patch fixes the issue by using a list comprehension instead of a\nfor loop over a list that was being modified within the loop.\n\nChange-Id: Icfce3b14abb55c6fef3de1b314cee22fc8b1d08c\nCloses-bug: #1858012\n(cherry picked from commit c2d88306621f890a857acd6831ea8bf073f55537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d413f1eba2d1e6b16ecd57fa27de528dd0f67cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d413f1eba2d1e6b16ecd57fa27de528dd0f67cb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
