)]}'
{"id":"openstack%2Fkeystone~706088","triplet_id":"openstack%2Fkeystone~stable%2Fstein~I9dc634e073657ff337751ec67363a57bd10e20d4","project":"openstack/keystone","branch":"stable/stein","topic":"k2k-tests-stable/stein","hashtags":[],"change_id":"I9dc634e073657ff337751ec67363a57bd10e20d4","subject":"Add voting k2k tests","status":"MERGED","created":"2020-02-05 19:05:00.000000000","updated":"2020-02-06 06:55:15.000000000","submitted":"2020-02-06 06:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706088-1580971869321-bbef5f2b","meta_rev_id":"d7e17c3745315796ca8968b31deea51654c73897","_number":706088,"virtual_id_number":706088,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 06:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2020-02-05 22:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2020-02-05 22:35:58.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 22:25:14.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-02-06 06:51:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07041d5d26d15116756e1ec7838a9232b6c100b6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-02-05 19:05:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eceea0c91a8cfd72a72df034a8b5c506987fbba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6295f32ecb849cd86f5d9f792dfd049 : SUCCESS in 34m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77384d937e484f7797f46b2c57dac32b : SUCCESS in 35m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fcfc5b8cad74a2cb6d3ce47d183be56 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ebf146880364fbb94bdaecb291f3a77 : SUCCESS in 24m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7486980965444560898b0cc87a21048f : SUCCESS in 26m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3de0a13f7a34a2ba55a51cf67aed238 : SUCCESS in 26m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a21117cc40b41fb83d7eec3b0abfac1 : SUCCESS in 14m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/af149648067d4a9c9b7d07a5c7ed60f2 : SUCCESS in 1h 28m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d45d77db123c4db18720bdd9935c400b : SUCCESS in 59m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8afef0185cfc461c9126041e3f5cab02 : SUCCESS in 1h 06m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ccdcd521fe843f58383329782199f34 : SUCCESS in 1h 25m 19s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/f398c831416c4d7fb38f20ed4dce78fc : SUCCESS in 34m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e01d6921079140cea2e7ba3151d2bc83 : SUCCESS in 36m 34s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6d7690b735aa443896e990a1bcae1cd8 : SUCCESS in 35m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3f93da85e97443ca8085c34308c3b68f : SUCCESS in 40m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/bcc71eabdf0c4a23924933cdb02b8208 : SUCCESS in 35m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/61f6cdd1aa2e4131829eb7f0ef432979 : SUCCESS in 15m 44s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/465891570c2846e48fcdb4ef6fbee624 : SUCCESS in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"780081e7d1b98861c33fb1c546fb715dc3ca4f84","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-02-05 22:25:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f35e0ce7e2544c0393c78ab22df38b347cb5f9d5","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-02-05 22:35:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2152a1258663346c9ece5a0cc755fa4e3fd6ae39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 22:36:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fdf7512f47d8dd81469a305b57c65fbc560cad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:12:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29507424946446ee851fffd485e03197 : SUCCESS in 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65641bd19b09493981560d3c3db8a934 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d005d62d3d74af1a50cfce9b580c865 : SUCCESS in 22m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/939b312c941744b3bea439902536eefb : SUCCESS in 28m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb517fc560a643fc813f2475b987426d : SUCCESS in 29m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3187472d249046e2863c875698102a03 : SUCCESS in 11m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bdb4585a18fa4b36b5ac3a09f74dea8b : SUCCESS in 1h 30m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab64ba8402be490fba60b320157cc903 : FAILURE in 9m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bc9cf092c811403eaca67c6a99a40066 : SUCCESS in 1h 09m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93a4bfe08024461d8ea7256ca5333711 : SUCCESS in 1h 19m 12s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/0aa17629b3b54a319e112e995e3e8eb6 : SUCCESS in 35m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5c87699156504a3bad860fab6cbf588c : SUCCESS in 35m 58s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d4f92118633c45f68d2a89fec44d1bcf : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4857dd4fd5baeba0a1a11ac76fcee4d2d810a2cd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-02-06 01:23:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c09e2d9c3285a957414c87d1f7119ead8175741c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/effada3301d642b5bce5cb61a20ab3c8 : SUCCESS in 29m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d24ff1133014a8f89ccd66c65c84809 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc03b3e0e3634d5ca663ec071cd60a52 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ce50cddf0c24b4c9e8c83592d8f74c5 : SUCCESS in 27m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4adffc4a0fe14e6895302c2990ef84e0 : SUCCESS in 23m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50f5d74bc0ed42139127a2ae5c883285 : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82cf1825a6f541e288a19736f210dbe3 : SUCCESS in 10m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0fc4b16c9ebf49988877f3e77d15be16 : SUCCESS in 1h 36m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a655354406f64112938bc23a02a8a651 : SUCCESS in 1h 01m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7f299885e814046bf5e7f15c2c86782 : SUCCESS in 50m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b75b15e214bb4724966d9d5f627b3c8e : SUCCESS in 1h 17m 16s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd9f46f302ad4062a013c29ba2c7f991 : SUCCESS in 34m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9253e26a73084874916ea357e602f808 : SUCCESS in 34m 20s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1881a3e5b9b549ff9820f03bd12a1d03 : SUCCESS in 34m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c210d42c28e646f0ab0b3aa7935a7e21 : SUCCESS in 38m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/8f4e6a03d88c471a952605c5c8ace198 : SUCCESS in 37m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1a882b7aefd640a19730d3f7b85bf771 : FAILURE in 18m 15s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/144f0c4613a246f0a086f0e89b0c53be : SUCCESS in 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08df1df7cdb82e4caf427a6a4267f42091156c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:06:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc6c1bb2086f21cedae0771209a6b65c8c7ef943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:51:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71800cd2a6704a5a96f01b8ab3307378 : SUCCESS in 29m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec4c18f4ffa14505bf6460f2cd561c9d : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0530b53497064afc96a8e8a00c748140 : SUCCESS in 23m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08c745cd211a416899229df77687094a : SUCCESS in 25m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/257e94f712a34c928cfb1837af34e6a7 : SUCCESS in 25m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38c3102cd7c415ca0a8674960e07081 : SUCCESS in 9m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2d152bda8c544fffbc92a6a99c1428f3 : SUCCESS in 1h 42m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb7daa1ff43142b989d6c444eadac424 : SUCCESS in 55m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e61e731f12df46a8b5da007d90f3b1c4 : SUCCESS in 55m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa88b689d91a4b3592e4df490f7b299b : SUCCESS in 1h 34m 20s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c5965b788e04f42aee1794058077341 : SUCCESS in 31m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d5b1a6a5da6a4ed6a7f452c78b020a54 : SUCCESS in 36m 16s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e047e2883e59445a8e491a7cd9978bb6 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"499be228dc03d68c20f520c501f3e3b8cec8f44d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:51:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"046aa61bad01a00426c07f1d19e5c0f5102e4b62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:55:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4709f90202c746318a96e555a260662c : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f537ed542698ebeb20d110d766b26320733effe","revisions":{"8f537ed542698ebeb20d110d766b26320733effe":{"kind":"REWORK","_number":1,"created":"2020-02-05 19:05:00.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/88/706088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/706088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/706088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/706088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/706088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/706088/1"}}},"commit":{"parents":[{"commit":"0e6c07e46fdfdd7662dd3eee7a110f72dda128e6","subject":"Added keystone identity provider installation to Devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0e6c07e46fdfdd7662dd3eee7a110f72dda128e6"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-17 17:12:00.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-02-05 19:04:26.000000000","tz":-480},"subject":"Add voting k2k tests","message":"Add voting k2k tests\n\nWith the addition of K2K-specific tests in the tempest plugin and a\nconfig toggle in the plugin to disable use of the external IdP, we can\nsafely add a voting federation job. This also fixes the devstack plugin\nto install the xmlsec1 tool which is needed for K2K.\n\nChange-Id: I9dc634e073657ff337751ec67363a57bd10e20d4\nDepends-on: https://review.opendev.org/689222\n(cherry picked from commit fb0be8e59917b5b637926d9526a64676af07ebf0)\n(cherry picked from commit ac74320876ac6c771d69b85ebb30058fe4eb3692)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8f537ed542698ebeb20d110d766b26320733effe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8f537ed542698ebeb20d110d766b26320733effe"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
