)]}'
{"id":"openstack%2Fkeystone~710734","triplet_id":"openstack%2Fkeystone~master~I848f168215800b42480d14f1f1905d8843ed3faa","project":"openstack/keystone","branch":"master","topic":"710734","hashtags":[],"change_id":"I848f168215800b42480d14f1f1905d8843ed3faa","subject":"Correcting api-ref for users","status":"MERGED","created":"2020-03-02 10:17:11.000000000","updated":"2020-03-17 20:36:04.000000000","submitted":"2020-03-17 20:30:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710734-1584477024646-ee8e90f4","meta_rev_id":"7e2da0eaf814a3fdc564654b4eb524318554831f","_number":710734,"virtual_id_number":710734,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 20:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2020-03-10 20:38:48.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-17 14:26:26.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-17 14:26:26.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 20:38:48.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-03-17 14:26:26.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-03-17 20:30:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3108dfd92a6c4337631b94c053efbea5ca85f49","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-02 10:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02ecb938e986f89c604dc4e9fa6f2fe2bee776e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:00:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4f5e02d9a7804c71aeb9224479a716db : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2c4ee4ba13443cb3e1c4e44fa91328 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a1cefa80af4364bb4102d874ce6207 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9ada492ce7f336ce797968a239372817817a134c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-03 23:30:16.000000000","message":"Patch Set 1: Code-Review-1\n\nThese options are actually optional though, if you create a user without a default project ID or without options then those attributes won\u0027t show up in the response. Was there some discussion that I missed/forgot about \"optional\" appearing in the api-ref response?","accounts_in_message":[],"_revision_number":1},{"id":"c5cf2ae6f1f4502ae080dcc6e8828071d6ddbe9a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-04 05:38:54.000000000","message":"Patch Set 1:\n\n\u003e These options are actually optional though, if you create a user\n \u003e without a default project ID or without options then those\n \u003e attributes won\u0027t show up in the response. Was there some discussion\n \u003e that I missed/forgot about \"optional\" appearing in the api-ref\n \u003e response?\n\nNot exactly but there are many other options that are \"optional\" too like domain_id and enabled in user api-ref. They don\u0027t show \"optional\" in the response.","accounts_in_message":[],"_revision_number":1},{"id":"ab4ab34cfe90c881840b70baaf8fc0b5231d8ede","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-04 17:54:34.000000000","message":"Patch Set 1:\n\n\u003e \u003e These options are actually optional though, if you create a user\n \u003e \u003e without a default project ID or without options then those\n \u003e \u003e attributes won\u0027t show up in the response. Was there some\n \u003e discussion\n \u003e \u003e that I missed/forgot about \"optional\" appearing in the api-ref\n \u003e \u003e response?\n \u003e \n \u003e Not exactly but there are many other options that are \"optional\"\n \u003e too like domain_id and enabled in user api-ref. They don\u0027t show\n \u003e \"optional\" in the response.\n\nThose are both always present in the response:\n\n$ curl -s -H \"x-auth-token: $token\" -H \"content-type: application/json\" http://localhost/identity/v3/users -d \u0027{\"user\": {\"name\": \"foo\", \"password\": \"foobar\"}}\u0027 | jq .\n{\n  \"user\": {\n    \"id\": \"efbebf7ced7e4fea9f460283d9b7cede\",\n    \"name\": \"foo\",\n    \"domain_id\": \"default\",\n    \"enabled\": true,\n    \"password_expires_at\": null,\n    \"options\": {},\n    \"links\": {\n      \"self\": \"http://192.168.122.96/identity/v3/users/efbebf7ced7e4fea9f460283d9b7cede\"\n    }\n  }\n}","accounts_in_message":[],"_revision_number":1},{"id":"42b2d1f3edac1e6640c510ad16bd7ba84bb4303e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-05 05:19:58.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e These options are actually optional though, if you create a\n \u003e user\n \u003e \u003e \u003e without a default project ID or without options then those\n \u003e \u003e \u003e attributes won\u0027t show up in the response. Was there some\n \u003e \u003e discussion\n \u003e \u003e \u003e that I missed/forgot about \"optional\" appearing in the api-ref\n \u003e \u003e \u003e response?\n \u003e \u003e\n \u003e \u003e Not exactly but there are many other options that are \"optional\"\n \u003e \u003e too like domain_id and enabled in user api-ref. They don\u0027t show\n \u003e \u003e \"optional\" in the response.\n \u003e \n \u003e Those are both always present in the response:\n \u003e \n \u003e $ curl -s -H \"x-auth-token: $token\" -H \"content-type:\n \u003e application/json\" http://localhost/identity/v3/users -d \u0027{\"user\":\n \u003e {\"name\": \"foo\", \"password\": \"foobar\"}}\u0027 | jq .\n \u003e {\n \u003e \"user\": {\n \u003e \"id\": \"efbebf7ced7e4fea9f460283d9b7cede\",\n \u003e \"name\": \"foo\",\n \u003e \"domain_id\": \"default\",\n \u003e \"enabled\": true,\n \u003e \"password_expires_at\": null,\n \u003e \"options\": {},\n \u003e \"links\": {\n \u003e \"self\": \"http://192.168.122.96/identity/v3/users/efbebf7ced7e4fea9f460283d9b7cede\"\n \u003e }\n \u003e }\n \u003e }\n\nAs per the response the \"options\" field is also present. I think only default_project_id is optional.","accounts_in_message":[],"_revision_number":1},{"id":"d5c9ab9deb6513581d5eb4bd0c939fb142cc09cc","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-05 17:36:12.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e \u003e These options are actually optional though, if you create a\n \u003e \u003e user\n \u003e \u003e \u003e \u003e without a default project ID or without options then those\n \u003e \u003e \u003e \u003e attributes won\u0027t show up in the response. Was there some\n \u003e \u003e \u003e discussion\n \u003e \u003e \u003e \u003e that I missed/forgot about \"optional\" appearing in the\n \u003e api-ref\n \u003e \u003e \u003e \u003e response?\n \u003e \u003e \u003e\n \u003e \u003e \u003e Not exactly but there are many other options that are\n \u003e \"optional\"\n \u003e \u003e \u003e too like domain_id and enabled in user api-ref. They don\u0027t show\n \u003e \u003e \u003e \"optional\" in the response.\n \u003e \u003e\n \u003e \u003e Those are both always present in the response:\n \u003e \u003e\n \u003e \u003e $ curl -s -H \"x-auth-token: $token\" -H \"content-type:\n \u003e \u003e application/json\" http://localhost/identity/v3/users -d \u0027{\"user\":\n \u003e \u003e {\"name\": \"foo\", \"password\": \"foobar\"}}\u0027 | jq .\n \u003e \u003e {\n \u003e \u003e \"user\": {\n \u003e \u003e \"id\": \"efbebf7ced7e4fea9f460283d9b7cede\",\n \u003e \u003e \"name\": \"foo\",\n \u003e \u003e \"domain_id\": \"default\",\n \u003e \u003e \"enabled\": true,\n \u003e \u003e \"password_expires_at\": null,\n \u003e \u003e \"options\": {},\n \u003e \u003e \"links\": {\n \u003e \u003e \"self\": \"http://192.168.122.96/identity/v3/users/efbebf7ced7e4fea9f460283d9b7cede\"\n \u003e \u003e }\n \u003e \u003e }\n \u003e \u003e }\n \u003e \n \u003e As per the response the \"options\" field is also present. I think\n \u003e only default_project_id is optional.\n\nOh, you\u0027re right. So we should also remove the \"Only present in the response if set on the user entity\" part of the description.","accounts_in_message":[],"_revision_number":1},{"id":"06abe6179fee9a451cee9b8e9a94a5cddcb24b87","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-06 06:54:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d67f27ece92fd611ad8afba6842eb521f1a0bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 07:18:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/52c5d977ee3a4b879f63a582cae95e1a : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ee2143ef624562bfd13274408e0aca : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3399224e44c4693b7dfd2a1cc2be9dc : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":2},{"id":"870a61cb3a603ea03b58f207a9e6b0dc7a40ba0b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-09 13:13:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5cb4a713ca58853009313b5eb213596035a361a6","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-10 17:13:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7a9e879549c8e22a82341ebe81124468ba65c198","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-10 17:29:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54a17b50a5925ec1c3ec823437962a6a44b027d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:04:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf01e70bfad94b958b9721aca0c35f61 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f739bd0ec8534927a2dac64254cb0cb4 : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7651c379c744d239088df01ebbfb01f : SUCCESS in 17m 38s","accounts_in_message":[],"_revision_number":3},{"id":"78b49a44dc82fbec25e237023efa64ae4c3a8d5a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-10 20:38:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9034d405aff43f7605aecc0249ba67e8ee9949fb","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-03-17 14:26:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a204a56aeb6cb807a8021137934725edf3bb055f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 14:27:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"767d787f00ec6bc0122144345acf72e2db305393","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:30:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/062677c26eed411ca50e51c14c60041c : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceff9e46e85349ecb9e5d0da3e038e1d : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5c0d4e144e48259dc915a240411fe0 : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":3},{"id":"0ff2399b95e731978dd2f78a82e8bf1aa27fee06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:30:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"50b26a00753549889538cbd56021173b7c4dd396","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:36:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8610f96caedf4d9a968825a9f627cb6d : SUCCESS in 1m 06s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c022f3e384d24809a47281d6d7be108b : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"271c09bb538b838e05865df034ba0f98f4a26259","revisions":{"c444e539fa726fae4ddd40dd9ef1fa1d7afe85d5":{"kind":"REWORK","_number":1,"created":"2020-03-02 10:17:11.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/34/710734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/710734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/710734/1"}}},"commit":{"parents":[{"commit":"97d83016afb93dd6981ce8bb07d8019384d8df22","subject":"Merge \"Tell reno to ignore the kilo branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97d83016afb93dd6981ce8bb07d8019384d8df22"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-03-02 10:13:27.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-03-02 10:13:27.000000000","tz":330},"subject":"Correcting api-ref for users","message":"Correcting api-ref for users\n\nThe response should not show any \u0027optional\u0027 with the parameters.\nThis patch removes the \u0027optional\u0027 from the same and also updates\nstring with the format to be used while creating api-ref in\nkeystone.\n\nChange-Id: I848f168215800b42480d14f1f1905d8843ed3faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c444e539fa726fae4ddd40dd9ef1fa1d7afe85d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c444e539fa726fae4ddd40dd9ef1fa1d7afe85d5"}]},"branch":"refs/heads/master"},"6841c6416ecf212f2c15cd4b8283467aa6db8b85":{"kind":"REWORK","_number":2,"created":"2020-03-06 06:54:00.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/34/710734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/710734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/710734/2"}}},"commit":{"parents":[{"commit":"97d83016afb93dd6981ce8bb07d8019384d8df22","subject":"Merge \"Tell reno to ignore the kilo branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97d83016afb93dd6981ce8bb07d8019384d8df22"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-03-02 10:13:27.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-03-06 05:25:32.000000000","tz":330},"subject":"Correcting api-ref for users","message":"Correcting api-ref for users\n\nThe \"options\" is not \"optional\" in user api responses.\nThis patch removes the \u0027optional\u0027 from the same and also updates\nstring with the format to be used while creating api-ref in\nkeystone.\n\nChange-Id: I848f168215800b42480d14f1f1905d8843ed3faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6841c6416ecf212f2c15cd4b8283467aa6db8b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6841c6416ecf212f2c15cd4b8283467aa6db8b85"}]},"branch":"refs/heads/master"},"271c09bb538b838e05865df034ba0f98f4a26259":{"kind":"REWORK","_number":3,"created":"2020-03-10 17:13:29.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/34/710734/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/710734/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/710734/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/710734/3"}}},"commit":{"parents":[{"commit":"97d83016afb93dd6981ce8bb07d8019384d8df22","subject":"Merge \"Tell reno to ignore the kilo branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97d83016afb93dd6981ce8bb07d8019384d8df22"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-03-02 10:13:27.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-03-10 17:13:19.000000000","tz":0},"subject":"Correcting api-ref for users","message":"Correcting api-ref for users\n\nThe \"options\" is not \"optional\" in user api responses.\nThis patch removes the \u0027optional\u0027 from the same and also updates\nstring with the format to be used while creating api-ref in\nkeystone.\n\nChange-Id: I848f168215800b42480d14f1f1905d8843ed3faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/271c09bb538b838e05865df034ba0f98f4a26259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/271c09bb538b838e05865df034ba0f98f4a26259"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
