)]}'
{"id":"openstack%2Fkeystone~712724","triplet_id":"openstack%2Fkeystone~master~I7dd0db1135898c4098093f54d759ff2384bf7714","project":"openstack/keystone","branch":"master","topic":"last_active_at-for-federated-users--712724","hashtags":[],"change_id":"I7dd0db1135898c4098093f54d759ff2384bf7714","subject":"Record last_active_at for local users using keystone as IdP","status":"ABANDONED","created":"2020-03-12 15:59:37.000000000","updated":"2022-01-21 21:00:08.000000000","total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56a777f67c3f731a8cb696706b52cffd5e05128f","_number":712724,"virtual_id_number":712724,"owner":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"date":"2020-06-11 11:01:57.000000000","_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"date":"2020-06-22 08:19:35.000000000","_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-21 13:41:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":-1,"date":"2020-05-27 21:35:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 13:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 21:35:30.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-06-11 11:01:57.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"}],"messages":[{"id":"e0a0fa766d4a261b4282b3ac01f0eee8c8e2ec57","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-03-12 15:59:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d412653726bee44a508ef06184dc8951b8a6ea5f","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-03-12 16:00:46.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a7b4fff8481cef0d86d6c28615f1032fdc4abcd8","accounts_in_message":[],"_revision_number":1},{"id":"317e5f22f4f8834a356191993c4a4c85f7ac3113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:54:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/912f094d3bf54f5dbe5febf799e84373 : SUCCESS in 26m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7a5aa073838441ca7ff786a4ff63e59 : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8bfe77d6134fb5a5d70348e479b5d1 : FAILURE in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7abbf24d406d4f1d8cd0e1f2d91ee6ed : SUCCESS in 18m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0fb8af999bf41ee9886ca8c1a20f93b : SUCCESS in 16m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50f2c80b0ed54a178d664750168c5ee4 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8dc55a2a063473c89d69bcbaa130cc1 : SUCCESS in 16m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/36bc372b0a274f6da7fba1d220b4e3fd : SUCCESS in 58m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c3a55cac4284d47b106182d5bf566dc : SUCCESS in 1h 20m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cefdf279cc0b4b66a6ee4e45d02af59f : SUCCESS in 37m 54s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/04c6055b648840488abc2ccd53915730 : FAILURE in 35m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/29e021582b554a4ba54a767b0ec17aa2 : SUCCESS in 49m 40s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/46d806e3b70243e59466e8dd2e682991 : SUCCESS in 19m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b03cb42d08a14cada82162dd37242f7b : SUCCESS in 51m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86341235a66f4860b526c952f409fbc3 : SUCCESS in 1h 11m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a4a141781abb40eb9f8ba954ea7ec655 : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dcfe2ad786df542e2a1f3c8fee4dde4bcdca84e2","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-03-13 08:44:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50e9a65b8a411089a7b0d107517b6d2d7ac613ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 10:18:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8df3405738014d298afb570a3ac34bbe : SUCCESS in 16m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4b9627ae14242a196a64e055e4504cc : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5eb9aef06a44bc0a8f927f6628e4a2e : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d37da011adf4e65b7422d604c78061d : SUCCESS in 13m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00750614eda8490ab9bc3e568493e476 : SUCCESS in 13m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1df37aecfb9742729e7bee329f5b0bb6 : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd2f266a8e34eb7892d0f2b15611c3d : SUCCESS in 13m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1aeb2d9cef1e4566b792abc04a1133a7 : SUCCESS in 1h 00m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76590ed69452400d9926c42e1fe349c6 : SUCCESS in 1h 32m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1689a7d475724dd0a0af2c71fa57bf19 : SUCCESS in 31m 43s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d254726200b448349a4dedcd538522ba : FAILURE in 36m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d492c1b9194a4a91bc9c7a8b383cd13c : SUCCESS in 43m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/44e20ee8a40c4b0a8d591c1f2bfd72be : SUCCESS in 18m 12s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/02e7a092583b4ee6842a44423cf3d20c : SUCCESS in 39m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83f6cd04747f496c9be36a25ef6281b5 : SUCCESS in 1h 01m 43s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/1e41df35e2a241ecb6269f3823029619 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"245e84526c054ef3564069757dcb046cf3d6215c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-03-16 09:31:35.000000000","message":"Patch Set 2: Code-Review-1\n\nHi Walter. Thanks for the change. Could you add a unit test supporting this change.","accounts_in_message":[],"_revision_number":2},{"id":"67859e25272c28c002e27a06b1465b800315cade","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-03-16 09:36:41.000000000","message":"Patch Set 2:\n\nHi Vishakha!\nThanks for the review.\n\n\u003e Could you add a unit test supporting this change.\n\nNot sure. It\u0027ll take more time than finding and creating the fix, and is lower down on the priority ladder.\n\nUnit tests are great. Especially when someone else who is familiar with the code creates them :)\n\nI\u0027ll see if I can find the time.","accounts_in_message":[],"_revision_number":2},{"id":"72c11e95861d1bcac531555fb93b0a01e606486d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-30 13:45:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8975f4624a056454cf50e60e5e85b655527f78ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:04:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b457f8186544bffb8e0da9c79f8d687 : FAILURE in 16m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f3a559b7e184071b576e9f222cd14ad : FAILURE in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e521f145d8be4321ae9eecf9f5c9bd16 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79b4281186224130864d7d6d8cfdc9c1 : FAILURE in 14m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76c590a272e345bd801914890a5e8be1 : FAILURE in 14m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/964894b479234d51a71962460556558d : FAILURE in 13m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7801693f03449aaa898d4e2319b26c6 : SUCCESS in 13m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f9cb817ecb04c068d85fd6872215557 : SUCCESS in 1h 02m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8869205ed1264cfca26bec6563d9be74 : SUCCESS in 1h 18m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9c28be46ae3b42f59440024e819d111f : SUCCESS in 36m 28s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e5bca91fffdd49d1a0b99425dd9b6b19 : SUCCESS in 37m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/b88d6462ae754a7ab8a0d5171550c5b3 : SUCCESS in 37m 12s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/18c7a9ff4dc54df79897ba7f92216302 : SUCCESS in 17m 06s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/49627b3d290d426d8abb25db56c011c6 : SUCCESS in 39m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4f7ee0addbb4b6d9f5c2140038b14c5 : SUCCESS in 59m 53s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/60e1d58a432f4a8ebbd9d86bde72a5d1 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":3},{"id":"41dca11c46626f5a34db643c251cd7b8ce94b379","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-30 17:28:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"26775b44957d08508c81950f20d6a7749d53788c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 18:56:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af47d3e8db8a4ba5bcbde8813ade9dd9 : TIMED_OUT in 52m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0240e5d4bf804dcf82f6f733bfcd3639 : FAILURE in 26m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae292c8df774c799a98ca7dbfd72634 : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e28231f4f8f84f9b867c3e71857f0795 : TIMED_OUT in 42m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9061a8ca818d4adfab9e8f1a8cf1f79c : TIMED_OUT in 42m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fbcf74ec6ea4b10a07ed41cf4f89832 : FAILURE in 15m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f237ae6d23349e4be15c42efa5c21d9 : SUCCESS in 22m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8621c441b8748dbbf3da2f129dbe683 : SUCCESS in 1h 19m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18add2ac086f49d79b3b6760fcc01c02 : SUCCESS in 1h 26m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ca3faca9915b4a7ab2f3352a7d76771c : SUCCESS in 32m 04s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ef4e8171e0c64e85bbf9e824bf11bb82 : SUCCESS in 51m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/f72047766d2d413bb454e5cdcf52f6e3 : FAILURE in 49m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/04263d7f542a4aacbcddfb1250907331 : SUCCESS in 19m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b1d1b2540f8f4badb00f6477cc64babc : SUCCESS in 50m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e7c68235a344839afee0de70ffade7a : SUCCESS in 1h 09m 22s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/22f54b00534b40c6ac0a6020b61d7b58 : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":3},{"id":"bea8f86f58c8e588fac4143b69828ebf7bd14478","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-04-30 20:07:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dcf080092a37564a80f404fb36827998ac1ea7e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 21:20:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef819b6e1aec498390a8cefa3406b739 : FAILURE in 39m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3565e5d503c542a8bc4729b9efa4cce7 : FAILURE in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1d496f6eee4699bd58c54556d45c79 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a255b6836eaf4b1a92dffa39de7b4253 : FAILURE in 21m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcb36e3dd40a4b6f865ffeac2d394e9b : FAILURE in 20m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cd35b58d6da4bcb911c06f2ca6dea30 : FAILURE in 18m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d02d0ae5344325a3f82dc70376860b : SUCCESS in 13m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/34d03e09e7ac4ac5af71b89652ce2238 : SUCCESS in 54m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1db3afb6ef14566879d1c6f5d332eae : SUCCESS in 1h 12m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/85a4ef0426ff48dfaf7b63829f524e7a : SUCCESS in 29m 48s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/a4118d9afaf0431697d5567190fa7431 : SUCCESS in 32m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/87a9f38348ca4bc3958557f64ee04551 : SUCCESS in 33m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0c613d2477b94a968dfc60260f8be927 : SUCCESS in 13m 44s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e730535ad47f44db8c3b605b77113f11 : SUCCESS in 34m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0404aa833f4b466f928849712a74dffc : SUCCESS in 55m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/70e08ff2c41f47ae80454bbc3a8d67fa : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":3},{"id":"5c2bdcebdb9d8966db90d44eca1e16d6930c288a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-01 05:00:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ecc009aa06c0a1e27cc9b39caf8891db094dbbb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 06:18:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c50e8c2fe31045a58c573fb9ea389ebf : SUCCESS in 24m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7d650ec13844e039261016b82393de2 : SUCCESS in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498b6e097af043138c78a1751be2f0fb : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ff6fb8b3d04a7f8329aff3d9f46640 : SUCCESS in 19m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a82536ccce934b76b5089d4279b9f104 : SUCCESS in 19m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ba63d3aded4f5eabeb6812338a50c2 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9ca9842fea4e0fbf2bfcb290268bc6 : SUCCESS in 13m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/94e9f20a0b424220b0de2b567a72cc1f : SUCCESS in 1h 02m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69e6bb84ae254ab5805c769b5789d97a : SUCCESS in 1h 15m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/46de3f347d6040eca4ddffe227195c2a : SUCCESS in 29m 26s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5a23734c47324a4e808c1cd7fea03998 : SUCCESS in 35m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/644f7fe6f38e48f2a2f8a35853c1cdd3 : SUCCESS in 31m 17s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/17770581c20f47bfa22df9dc807c0c0b : SUCCESS in 17m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8a19cd2185fb4fdf8164851a3f0a38e6 : SUCCESS in 36m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d88feb06e503421bb019d75be19b2468 : SUCCESS in 57m 46s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b6ead5e51bc742dcbfd67ef45972e690 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":4},{"id":"2f0b5b606740892183fb5e2c13562550a150b273","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-05 16:56:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nPlease include a release note.","accounts_in_message":[],"_revision_number":4},{"id":"203067d96116db4aceb1c24b0d4caf54412b102f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-05 17:37:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aef02d78b3889a0607643e8836cc91f939c828de","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-05 18:10:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d45daba6c3e6fa2ffa3a63d3d632c40f3ab6311","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-06 10:33:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf681a92a6e809fa38820dd609ca403a0b838432","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 17:56:06.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nWalter, could you perhaps clarify the problem you were having and how the patch fixes it so that Vishakha can be sure to capture it in the unit test? Feel free to go ahead and update the commit message, or leave a comment. I\u0027m not sure if you were already working on your own unit test, but if you were feel free to post that and it might also add some clarity.","accounts_in_message":[],"_revision_number":4},{"id":"293f19e9d798c3213b2c2ad6b4986549e97b7be5","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-05-06 18:37:41.000000000","message":"Patch Set 4:\n\nHi!\nThank you both for taking the time to get this merged.\nAs far as I read it, Vishakha understood things right, but the terminology of SPs/IdPs/Federation is not something I use on a daily basis. So, here goes:\n\nI\u0027m using the Keystone as a Service Provider (SP) and our own OpenID server (OIDC) as Identity Provider (IdP): federated authentication.\n\nThe mapping docs say:\n\n\u003e If the type attribute is not set to local in the local rule and no domain is specified, the user is deemed ephemeral and becomes a member of the identity provider’s domain.\n\nWe\u0027re using \u0027local\u0027 (or real) users, not \u0027ephemeral\u0027 ones (because I needed the users to be in the right domain).\n\nEvery time the users log in (federated auth), the last_active_at is NOT updated. (That probably makes sense for ephemeral (throwaway) users, but not for real ones). So, after 90 days, the users started losing access because of the disable_user_account_days_inactive feature.\n\nFor my use-case the postgres DB showed that this simple changeset did its work.\n\nI hope this clarifies things :)","accounts_in_message":[],"_revision_number":4},{"id":"f3cf2ec41451493c3f006fcd05a5f5d594d0468e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 18:51:01.000000000","message":"Patch Set 4:\n\n\u003e Hi!\n \u003e Thank you both for taking the time to get this merged.\n \u003e As far as I read it, Vishakha understood things right, but the\n \u003e terminology of SPs/IdPs/Federation is not something I use on a\n \u003e daily basis. So, here goes:\n \u003e \n \u003e I\u0027m using the Keystone as a Service Provider (SP) and our own\n \u003e OpenID server (OIDC) as Identity Provider (IdP): federated\n \u003e authentication.\n \u003e \n \u003e The mapping docs say:\n \u003e \n \u003e \u003e If the type attribute is not set to local in the local rule and\n \u003e no domain is specified, the user is deemed ephemeral and becomes a\n \u003e member of the identity provider’s domain.\n \u003e \n \u003e We\u0027re using \u0027local\u0027 (or real) users, not \u0027ephemeral\u0027 ones (because\n \u003e I needed the users to be in the right domain).\n \u003e \n \u003e Every time the users log in (federated auth), the last_active_at is\n \u003e NOT updated. (That probably makes sense for ephemeral (throwaway)\n \u003e users, but not for real ones). So, after 90 days, the users started\n \u003e losing access because of the disable_user_account_days_inactive\n \u003e feature.\n \u003e \n \u003e For my use-case the postgres DB showed that this simple changeset\n \u003e did its work.\n \u003e \n \u003e I hope this clarifies things :)\n\n+1 thank you! Makes sense to me :)","accounts_in_message":[],"_revision_number":4},{"id":"32f09ea95478ad88db08f98c437380db281b8356","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-19 11:27:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ac6985df032cfa4520c15352bfb64cc61a3ad6c8","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-19 11:29:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dc72846c00372d3b5388f9712a6ca76f769abd0d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-19 11:31:49.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"299a47d1ed83361057328edb9536ff018be1647a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-19 11:44:57.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cfe000a5e7b0fee76db40c10cd0515cbebeca4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 13:07:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed5203a520f4e0ea248e346a2e0fbeb : SUCCESS in 39m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d78c42e192d4277be8b635c3024aa75 : SUCCESS in 32m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eeb13ae97bb487da75ddd79b034d72e : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5488b354ecb4ba5aee9b6afc90262c5 : TIMED_OUT in 41m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d36defc3a5294f2493314cd43e8568d6 : SUCCESS in 18m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d2a17efab7483091a8087cce72d331 : SUCCESS in 19m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/40b4e6f6b50d45388277fa17e8ad4e40 : SUCCESS in 1h 29m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73f4de6ef4944bf1aa7e57e736a09863 : SUCCESS in 1h 33m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f007ad1bc6d04629998570b8be19d230 : SUCCESS in 9m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/eca80f33bc134da5bd57dfd152891742 : SUCCESS in 44m 15s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1ef4de1c37204f239ae4adb1f01e8f2e : SUCCESS in 36m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/848e36e0cb2040ae9ed8496ef416b2c5 : FAILURE in 43m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/653b972355c04057bd29a9ff532779f1 : SUCCESS in 12m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/13ce920595f54555971c7cd54d4a2d8c : SUCCESS in 35m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be430ca310b840e9a41e5894bebddb1e : SUCCESS in 1h 05m 55s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/32ff10ff9c3d48458709a633df6e806b : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":7},{"id":"82aac3f4a5aca6ace67ead95c5432c8b71741405","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-19 16:58:17.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"75e5be0f68e1df96bb815dc859d7780557bc010a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 18:28:22.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e594bc3245864ba0a074aa6788ec0e37 : SUCCESS in 24m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa3af1799f9743d3b31542b99d1ea211 : TIMED_OUT in 41m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28e7bbe9dec4a0e8029a88aa756d450 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99d94ba4148b4ce68a2e927398af5bb0 : SUCCESS in 35m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5a0e25ba0b740f9aa0ac285cf335bd2 : SUCCESS in 35m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18bc2fda44d4318bdd825da110e5bc4 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/5160e0f0c12f476cbc9767b4458a9ffb : SUCCESS in 1h 09m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b113c94e177d4db9b14f478c89e14cdd : SUCCESS in 1h 26m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43b4c390ee504b3ab8f0d29467d88eee : SUCCESS in 9m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ffe3f7e1a1954f1fa7e809703550171e : SUCCESS in 41m 28s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3c8f4e7fa6404597bcf7f77b6b9a295b : SUCCESS in 32m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/6deda22027404e1fba19bcac7b63a11e : SUCCESS in 37m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c7423073d5b04b0ab99d1fd0c37df6af : SUCCESS in 17m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fa9c558b86ad45ea815f159a3628b459 : SUCCESS in 42m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38aa27ccaeb740839ca5ec6c2f57e67a : SUCCESS in 1h 06m 41s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/846f8146c3de406dae564f38e656e742 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":7},{"id":"f7decba5909812a2f2c9a2bccc33ec71cc9c61cd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-19 19:47:02.000000000","message":"Patch Set 7: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"382888a7923a9efbf76b1d5a1710e7006e477f0f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-21 11:44:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6033dd177eedc11d12fb38b44a7044cf5c51d989","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-21 11:49:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ad599de6b3fea226d7734df15b0877639828522f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-21 12:10:01.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bda54599c1b3adcd689821cad983a2a0e5b0092b","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-21 12:13:30.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8028844f2f753ebd6f7217f38550fc85602c7b12","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-21 12:15:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d926fa81a4080e612162c4e2ea7a084cad82e80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:41:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57262a2aa2bb4cf49b5c8cda6e97d9ab : SUCCESS in 19m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f768f469bbf74a25b045f1d769f77fc9 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e337a8641bc481b8d305c4856569758 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db69d8f937404fa29ca198a5dcd49ef1 : SUCCESS in 22m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71832da6af5748959b215ce5502eea42 : SUCCESS in 21m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15653252177d4b19b1c9fc270b9aa40d : SUCCESS in 16m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/68719827b9214badbf97367959ae9406 : SUCCESS in 1h 03m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4da5698e4a1341018056bc0f6d007112 : SUCCESS in 1h 25m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/965aec1e95004bc5bd64910034eb6b49 : SUCCESS in 7m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/77e55979b5ee49ac8ce2b913cf60c040 : SUCCESS in 38m 46s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5370fb6b3ded42c19f885ceaeaacda91 : SUCCESS in 42m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/c19abf3317b24ccead14497e3e2ce48f : SUCCESS in 41m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1e37313b901041ff9805e93e6d310d6f : SUCCESS in 21m 40s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5ff433446ce4488da08dded77dd738e7 : SUCCESS in 43m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95be790b3777467f9f2c43552bd20fdc : SUCCESS in 1h 06m 09s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/fc510516729842659212bce0d5841039 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":10},{"id":"a74d4ea406f267b65b8745dc96d657fba70331ad","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-27 21:35:30.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4a8be8bd25689ab31689bb8920ce80fc140ed4ba","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-06-11 11:01:57.000000000","message":"Patch Set 10:\n\n\u003e Hi!\n \u003e Thank you both for taking the time to get this merged.\n \u003e As far as I read it, Vishakha understood things right, but the\n \u003e terminology of SPs/IdPs/Federation is not something I use on a\n \u003e daily basis. So, here goes:\n \u003e \n \u003e I\u0027m using the Keystone as a Service Provider (SP) and our own\n \u003e OpenID server (OIDC) as Identity Provider (IdP): federated\n \u003e authentication.\n \u003e \n \u003e The mapping docs say:\n \u003e \n \u003e \u003e If the type attribute is not set to local in the local rule and\n \u003e no domain is specified, the user is deemed ephemeral and becomes a\n \u003e member of the identity provider’s domain.\n \u003e \n \u003e We\u0027re using \u0027local\u0027 (or real) users, not \u0027ephemeral\u0027 ones (because\n \u003e I needed the users to be in the right domain).\n \u003e \n \u003e Every time the users log in (federated auth), the last_active_at is\n \u003e NOT updated. (That probably makes sense for ephemeral (throwaway)\n \u003e users, but not for real ones). So, after 90 days, the users started\n \u003e losing access because of the disable_user_account_days_inactive\n \u003e feature.\n \u003e \n \u003e For my use-case the postgres DB showed that this simple changeset\n \u003e did its work.\n \u003e \n \u003e I hope this clarifies things :)\n\nHi Walter. Thanks for the reply. I just wanted to confirm how do user login after 90 days in  OpenID server (OIDC) as Identity Provider (IdP? Do you use any dashboard and login with the help of id and password and get the error that the user is disabled?","accounts_in_message":[],"_revision_number":10},{"id":"0534c634ebf84061370ac5bbb41d5a49d77703be","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-06-22 08:19:35.000000000","message":"Patch Set 10:\n\nWalter wrote:\n\n \u003e \u003e Every time the users log in (federated auth), the\n \u003e \u003e last_active_at is NOT updated.\n \u003e \u003e (That probably makes sense for ephemeral (throwaway)\n \u003e \u003e users, but not for real ones). So, after 90 days, the\n \u003e \u003e users started losing access because of the \n \u003e \u003e disable_user_account_days_inactive feature.\n\nVishakha wrote:\n\n \u003e Hi Walter. Thanks for the reply. I just wanted to\n \u003e confirm how do user login after 90 days in  OpenID\n \u003e server (OIDC) as Identity Provider (IdP)?\n \u003e Do you use any dashboard and login with the help of\n \u003e id and password and get the error that the user is\n \u003e disabled?\n\nI\u0027m not sure I grok your question entirely.\n\nI\u0027ll rephrase your question and answer:\n\n \u003e How do users log in when they have been disabled after\n \u003e 90 days of inactivity?\n\n\nBefore logging in the \u0027wdoekes\u0027 user, who was _not_ inactive:\n\n    keystone_db\u003d#\n      select substring(u.id, 1, 3) as id_, l.id as l_id,\n        l.name, (current_date - u.last_active_at) as inactive_days,\n        u.enabled from \"user\" as u\n      inner join local_user l on l.user_id \u003d u.id\n      where l.name like \u0027wdoekes%\u0027 order by inactive_days;\n\n     id_ | l_id |       name       | inactive_days | enabled \n    -----+------+------------------+---------------+---------\n     3ca |   99 | wdoekes          |             7 | t\n     0ba |   96 | wdoekes-test     |           231 | t\n\n\nAfter logging in the \u0027wdoekes\u0027 user:\n\n     id_ | l_id |       name       | inactive_days | enabled \n    -----+------+------------------+---------------+---------\n     3ca |   99 | wdoekes          |             0 | t\n     0ba |   96 | wdoekes-test     |           231 | t\n\n\nUsing the wdoekes-test user (whch has enabled\u003dtrue,\nbut lots of inactive days), I get this:\n\n- horizon says the user has Enabled \u003d No\n- the DB has the user enabled \u003d true\n- logging in does not succeed, with \"(keystone.auth.plugins.core):\n  2020-06-22 09:50:36,499 WARNING User is disabled: 0ba...\"\n\n\nBy logging into Horizon with a different user, I can update\nthe field, using the \"Enable User\" button. After that, I _can_\nlog in.\n\n\nThe \"Enable User\" changes the last_active_at to today:\n\n     id_ | l_id |       name       | inactive_days | enabled \n    -----+------+------------------+---------------+---------\n     3ca |   99 | wdoekes          |             0 | t\n     0ba |   96 | wdoekes-test     |             0 | t\n\nWhen updating this value in the DB directly (without going\nthrough Horizon and Keystone), flushing the keystone\ncache (redis/memcached) is mandatory:\n\n- update \"user\" set last_active_at \u003d \u00272020-06-01\u0027 where id like \u00270ba%\u0027;\n- redis-cli\u003e flushdb\n\n\nDoes that answer your question?\n\nWalter Doekes\nOSSO B.V.","accounts_in_message":[],"_revision_number":10},{"id":"56a777f67c3f731a8cb696706b52cffd5e05128f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:00:08.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"066606e50d5de1ef6245e222cbc4b580068c0982","revisions":{"76a8322a90122bafde7800cfc6807f1be1dc16ce":{"kind":"REWORK","_number":1,"created":"2020-03-12 15:59:37.000000000","uploader":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"ref":"refs/changes/24/712724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/1"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:55:23.000000000","tz":60},"subject":"Also record last_active_at for local users using IdP to log in","message":"Also record last_active_at for local users using IdP to log in\n\nIf a user tries to log in after disable_user_account_days_inactive days,\nthe login is denied with 401, the log showing a \"User is disabled\".\n\nBefore this changeset, local users using an federated identity for login\ndid not get an updated last_active_at. Hence the login was denied 90\nafter account setup.\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76a8322a90122bafde7800cfc6807f1be1dc16ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76a8322a90122bafde7800cfc6807f1be1dc16ce"}]},"branch":"refs/heads/master"},"c0852384d65030ac62c44e3a70a4db2b8184585f":{"kind":"REWORK","_number":2,"created":"2020-03-13 08:44:45.000000000","uploader":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"ref":"refs/changes/24/712724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/2"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-13 08:44:26.000000000","tz":60},"subject":"Also record last_active_at for local users using IdP to log in","message":"Also record last_active_at for local users using IdP to log in\n\nIf a user tries to log in after disable_user_account_days_inactive days,\nthe login is denied with 401, the log showing a \"User is disabled\".\n\nBefore this changeset, local users using an federated identity for login\ndid not get an updated last_active_at. Hence the login was denied 90\nafter account setup.\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0852384d65030ac62c44e3a70a4db2b8184585f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0852384d65030ac62c44e3a70a4db2b8184585f"}]},"branch":"refs/heads/master"},"2ccaab87f797e714a4589f75fdf051366bb0df58":{"kind":"REWORK","_number":3,"created":"2020-04-30 13:45:14.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/3"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-04-30 13:05:07.000000000","tz":330},"subject":"Also record last_active_at for local users using IdP to log in","message":"Also record last_active_at for local users using IdP to log in\n\nIf a user tries to log in after disable_user_account_days_inactive days,\nthe login is denied with 401, the log showing a \"User is disabled\".\n\nBefore this changeset, local users using an federated identity for login\ndid not get an updated last_active_at. Hence the login was denied 90\nafter account setup.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2ccaab87f797e714a4589f75fdf051366bb0df58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2ccaab87f797e714a4589f75fdf051366bb0df58"}]},"branch":"refs/heads/master"},"48af2e417228033a37250bd9c4ccc1a7d9f7b790":{"kind":"REWORK","_number":4,"created":"2020-05-01 05:00:15.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/4"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-04-30 13:57:46.000000000","tz":330},"subject":"Also record last_active_at for local users using IdP to log in","message":"Also record last_active_at for local users using IdP to log in\n\nIf a user tries to log in after disable_user_account_days_inactive days,\nthe login is denied with 401, the log showing a \"User is disabled\".\n\nBefore this changeset, local users using an federated identity for login\ndid not get an updated last_active_at. Hence the login was denied 90\nafter account setup.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/48af2e417228033a37250bd9c4ccc1a7d9f7b790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/48af2e417228033a37250bd9c4ccc1a7d9f7b790"}]},"branch":"refs/heads/master"},"d85cda323a91c98b35cf9c4ef0069f0e86ce4f53":{"kind":"REWORK","_number":5,"created":"2020-05-19 11:27:21.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/5"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-19 11:02:46.000000000","tz":330},"subject":"Record last_active_at for local users using keystone as IdP","message":"Record last_active_at for local users using keystone as IdP\n\nIf a normal local user in keystone tries to log in after\ndisable_user_account_days_inactive days, the login is denied\nwith 401, the log showing a \"User is disabled\".\n\nBefore this \"last_active_at\" for local users using keystone\nas IDP was not updating. If the option disable_user_account\n_days_inactive is set in keystone.conf, and if the user logged\nin for federated_auth before the days set in disable_user_\naccount_days_inactive, the last active at was not updating and\nthe user used to get disabled after disable_user_account_days_\ninactive.\n\nAfter this patch user can view last_active_at set in the database\nfor local user using keystone as idp.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d85cda323a91c98b35cf9c4ef0069f0e86ce4f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d85cda323a91c98b35cf9c4ef0069f0e86ce4f53"}]},"branch":"refs/heads/master"},"cc4c12202cbc9d2249bc78c141712e8929f4560a":{"kind":"REWORK","_number":6,"created":"2020-05-19 11:29:35.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/6"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-19 11:29:03.000000000","tz":330},"subject":"Record last_active_at for local users using keystone as IdP","message":"Record last_active_at for local users using keystone as IdP\n\nIf a normal local user in keystone tries to log in after\ndisable_user_account_days_inactive days, the login is denied\nwith 401, the log showing a \"User is disabled\".\n\nBefore this \"last_active_at\" for local users using keystone\nas IDP was not updating. If the option disable_user_account\n_days_inactive is set in keystone.conf, and if the user logged\nin for federated_auth before the days set in disable_user_\naccount_days_inactive, the last active at was not updating and\nthe user used to get disabled after disable_user_account_days_\ninactive.\n\nAfter this patch user can view last_active_at set in the database\nfor local user using keystone as idp.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cc4c12202cbc9d2249bc78c141712e8929f4560a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cc4c12202cbc9d2249bc78c141712e8929f4560a"}]},"branch":"refs/heads/master"},"822c5443479fac82f1b2403a4a29aa2760518d93":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-05-19 11:31:49.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/7"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-19 11:30:03.000000000","tz":330},"subject":"Record last_active_at for local users using keystone as IdP","message":"Record last_active_at for local users using keystone as IdP\n\nIf a normal local user in keystone tries to log in after\ndisable_user_account_days_inactive days, the login is denied\nwith 401, the log showing a \"User is disabled\".\n\nBefore this \"last_active_at\" for local users using keystone\nas IDP was not updating. If the option disable_user_account\n_days_inactive is set in keystone.conf, and if the user logged\nin for federated auth before the no. of days set in disable_user_\naccount_days_inactive, the \"last_active_at\" was not updating and\nthe user used to get disabled after disable_user_account_days_\ninactive.\n\nAfter this patch user can view last_active_at set in the database\nfor local user using keystone as idp.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/822c5443479fac82f1b2403a4a29aa2760518d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/822c5443479fac82f1b2403a4a29aa2760518d93"}]},"branch":"refs/heads/master"},"34195166ec203f218f255d8e66e80ad40bf8af35":{"kind":"REWORK","_number":8,"created":"2020-05-21 11:44:39.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/8"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-21 11:36:20.000000000","tz":330},"subject":"Record last_active_at for local users using keystone as IdP","message":"Record last_active_at for local users using keystone as IdP\n\nIf a normal local user in keystone tries to log in after\ndisable_user_account_days_inactive days, the login is denied\nwith 401, the log showing a \"User is disabled\".\n\nBefore this \"last_active_at\" for local users using federated\nidentity was not updating. If the option disable_user_account\n_days_inactive is set in keystone.conf, and if the federated\nuser logged in for federated auth before the no. of days set\nin disable_user_account_days_inactive, the \"last_active_at\"\nwas not updating and the user used to get disabled after\ndisable_user_account_days_inactive.\n\nAfter this patch the federated user can continue to log in as\nlong as they are refreshing their active status by logging in\nregularly.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/34195166ec203f218f255d8e66e80ad40bf8af35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/34195166ec203f218f255d8e66e80ad40bf8af35"}]},"branch":"refs/heads/master"},"46b7822b66a53c07b93e2edd234c69ccc50677f1":{"kind":"REWORK","_number":9,"created":"2020-05-21 11:49:54.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/9"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-21 11:48:53.000000000","tz":330},"subject":"Record last_active_at for local users using keystone as IdP","message":"Record last_active_at for local users using keystone as IdP\n\nIf a normal local user in keystone tries to log in after\ndisable_user_account_days_inactive days, the login is denied\nwith 401, the log showing a \"User is disabled\".\n\nBefore this \"last_active_at\" for local users using federated\nidentity was not updating. If the option disable_user_account\n_days_inactive is set in keystone.conf, and if the federated\nuser logged in for federated auth before the no. of days set\nin disable_user_account_days_inactive, the \"last_active_at\"\nwas not updating and the user used to get disabled after\ndisable_user_account_days_inactive.\n\nAfter this patch the federated user can continue to log in as\nlong as they are refreshing their active status by logging in\nregularly.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46b7822b66a53c07b93e2edd234c69ccc50677f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46b7822b66a53c07b93e2edd234c69ccc50677f1"}]},"branch":"refs/heads/master"},"066606e50d5de1ef6245e222cbc4b580068c0982":{"kind":"REWORK","_number":10,"created":"2020-05-21 12:10:01.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/24/712724/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/712724/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/712724/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/712724/10"}}},"commit":{"parents":[{"commit":"ba2e4b83e88d5c1576e9d2cd6a760367deef3a62","subject":"NIT: Fix spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba2e4b83e88d5c1576e9d2cd6a760367deef3a62"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2020-03-12 15:34:43.000000000","tz":60},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-21 12:09:29.000000000","tz":330},"subject":"Record last_active_at for local users using keystone as IdP","message":"Record last_active_at for local users using keystone as IdP\n\nIf a normal local user in keystone tries to log in after\ndisable_user_account_days_inactive days, the login is denied\nwith 401, the log showing a \"User is disabled\".\n\nBefore this \"last_active_at\" for local users using federated\nidentity was not updating. If the option disable_user_account\n_days_inactive is set in keystone.conf, and if the federated\nuser logged in for federated auth before the no. of days set\nin disable_user_account_days_inactive, the \"last_active_at\"\nwas not updating and the user used to get disabled after\ndisable_user_account_days_inactive.\n\nAfter this patch the federated user can continue to log in as\nlong as they are refreshing their active status by logging in\nregularly.\n\nCo-Authored-By: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\n\nChange-Id: I7dd0db1135898c4098093f54d759ff2384bf7714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/066606e50d5de1ef6245e222cbc4b580068c0982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/066606e50d5de1ef6245e222cbc4b580068c0982"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
