)]}'
{"id":"openstack%2Fkeystone~713566","triplet_id":"openstack%2Fkeystone~master~I15ef258381e522f3986ad43eda7e60da2289af8e","project":"openstack/keystone","branch":"master","topic":"debug-samltest","hashtags":[],"change_id":"I15ef258381e522f3986ad43eda7e60da2289af8e","subject":"[DNM] debug external idp tests","status":"ABANDONED","created":"2020-03-18 03:51:00.000000000","updated":"2020-04-21 17:20:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"526b1179aebb3452076f7b1ae08effe29e4a6c1f","_number":713566,"virtual_id_number":713566,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-03 20:50:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 20:50:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b688db20cd2776aca2f32ce1207d4f347329becf","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-18 03:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fcc5a7cd888d1e130a682c82dd49e2a2a7b3109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 05:19:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/775421b9249b487f9835077c49819629 : SUCCESS in 36m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d49a942c73a4b03bec8dbbf6d1e9128 : SUCCESS in 24m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2686d81dc83c4265b749780c347a9e0a : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4db8acf86c674af5b33075c6c5f24ed0 : SUCCESS in 17m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd4204329bf74a4d94f4eb5e5ef1c49a : SUCCESS in 19m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab3750ce07de4630a3d8fadcb3731ec5 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b1eacd54f146d4b9878b2ab0373ab5 : SUCCESS in 12m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/13a9a3850fdb4b83b2f238e3c4c2acbf : SUCCESS in 1h 00m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81f9ef3a3a294734806e0a95b96aff9d : SUCCESS in 1h 22m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/924b08d3e1934be7bd8dd599758e198a : SUCCESS in 33m 16s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cca7cd975c03407ea376d74a0c5d9fe0 : FAILURE in 32m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/1ea0dca0ff6848d3876fffde25c6995e : SUCCESS in 39m 41s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2ed21c9273c6421d985841b419b08675 : SUCCESS in 17m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/de1cb3486f794c54a90252abcf5d0d92 : SUCCESS in 40m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92ccd8e943c24b5292acf8b8ebbb2440 : SUCCESS in 1h 00m 06s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/20d0690ffb6d453a87f01ef64c4ff0dd : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ab608d138df0132bb2ff08836e035def84dfa041","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-19 03:07:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d26dbafb223ba4b487043a7e16e3b2b317fea9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 04:28:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a12d32a3d4c848f9b08aa09dd96dbd6c : SUCCESS in 25m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aae871c4e8e48ff80554f3c96c7b053 : SUCCESS in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f7aefd984a143788af915ff341212b1 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3164261126c4b52b272ffecdcb2b3bb : SUCCESS in 33m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e11163e9115b427d84d1380b0ccb0ff2 : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90e12fd315cd43cdb5d242c42644be75 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b86a6b784a5b44e8911ae2b5a3a45b94 : SUCCESS in 13m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c5333b2156f2433687d50ba353985d1d : SUCCESS in 57m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/461fe237c9f34154a7d0ba72dbb86323 : SUCCESS in 1h 19m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d3a23502127c479d8536396bec3510e6 : SUCCESS in 30m 14s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6187efc80abe4f779b3c79fa2f35bc12 : FAILURE in 28m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/03656ed665964407a337f3422cb04e94 : SUCCESS in 33m 08s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ccf5867606c94bf482e30d318e5c5fc2 : SUCCESS in 14m 04s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/618199d8a0794e7b903e34221414befb : SUCCESS in 36m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af5254bca2634c429d3c2738d6c4b0ae : SUCCESS in 1h 02m 04s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8cd260fe70ed4b639cefbbe4394955ab : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2c4727e6c0c85eb0235ae4ae650fb552d2cb187c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-19 04:41:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4800b11a3e58a9cdb6e2541bed805f625cb6fcb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 06:04:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c433ac340e4c4630bf29651d55d50145 : SUCCESS in 29m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdd41a62c6824b68a42679f6e1e34100 : SUCCESS in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffac8e93fedd4eb68b2f74b565353863 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87e50f6673714011b69692e567567ce2 : SUCCESS in 12m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df0c19390423437d9f926073cd6c6fda : SUCCESS in 12m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaa98d3d0b894b35a545270cdb55d0e2 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763c0fb2b9e54658ab6d3dc3a4134d13 : SUCCESS in 15m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6288c98e27124dc8bb5e2786af53aaf1 : SUCCESS in 1h 01m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ade14941f3e41bbb871297665a88e29 : SUCCESS in 1h 18m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/38cb06062afd4510bb893a4738dce476 : SUCCESS in 31m 28s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fc9caa731d2143cfaa35437d53909804 : FAILURE in 28m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d3e0a131c8db413db81191539dceac9f : POST_FAILURE in 33m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/38cc33e843d940299f2dafb0b88595da : SUCCESS in 15m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8ea9d05d52ed49e1acfdabe845650407 : SUCCESS in 39m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c377c88a76348b39c051294e7e98225 : SUCCESS in 59m 49s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/06fb8d8f578948938c1fff466b3710a3 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":3},{"id":"369e51e7de7066950503b308b17ee4d149016c6d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-19 18:49:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c3a1b6e46718b1ee7ea5207f1db4dd7250b1a971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:44:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b12ebaa6e4f24837b52a2fb390c694ed : SUCCESS in 27m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/800c95823fb5461fb155bb945931f335 : TIMED_OUT in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63085686e4914cb59fe47998a2e963ec : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/473fffc54f6e4e4eae7071015fde7df7 : TIMED_OUT in 41m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16f60a9806144854806d9942c4635b96 : SUCCESS in 40m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2a027827c1c4f8283f9ff80bbcd283f : TIMED_OUT in 41m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f83e2bea292c49268283f9fc976ab12f : SUCCESS in 17m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4901bba9666d4b5881e7f9c0a512f125 : FAILURE in 1h 08m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bf6d7fc81b84b21bf1524b9c499d9dd : SUCCESS in 1h 50m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9a4c6524fb6f4f33aebe6f4835afa3e5 : SUCCESS in 34m 45s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2655773e5b894ad9b56c89953fbbc10f : FAILURE in 25m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/74a42f025a8943fda9d951654e274df4 : SUCCESS in 37m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/152588c9bc294ba685615dadc509c219 : SUCCESS in 16m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1ada0720786643d8b00054c33d2c14ef : SUCCESS in 39m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4837955734f4436a9ed62132851894e : SUCCESS in 1h 04m 08s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ff993b8fdf9f4fc7ab1757d95839d977 : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":4},{"id":"311a861fd157c84d2e272acc1eeeb12b74560b30","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-19 20:52:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae4cba5fcb400a9041a666e08b48cc0fe39c791d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 22:36:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e592116054c49fb9a6ce1432435c018 : SUCCESS in 29m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3908f97af49f4d90976dda54aef314c6 : SUCCESS in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc0620e0659488dbf7ff35f95aad452 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80de7f4eb3624e9783928707a262c72b : SUCCESS in 21m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5cff06158e448b7b17e9d5fb8ebf7bd : SUCCESS in 24m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f049ac6735942f88a8b8e17e4fb8e08 : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ebf0ce7e194118b8a5929290e5ddcc : SUCCESS in 14m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/375eb7d39afb45ceb26676b6266f82fd : SUCCESS in 1h 04m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/347b631ffded4daba09e02a9e5e9ccc5 : SUCCESS in 1h 41m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4a54e34a43f944b2bbbaecf500586363 : SUCCESS in 39m 54s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d17a2b30667f431f9472499b4e212e3b : FAILURE in 29m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/3757909e9f7d42869bf2010b70cb9a10 : SUCCESS in 41m 17s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/198e26b092f44580b7415517373c184a : SUCCESS in 18m 52s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3cd1bf6dbbda4c4dac816e4bc0f684f1 : SUCCESS in 47m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33dc41a966084f53b4208f32a215d5e3 : SUCCESS in 1h 04m 38s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/59a63fe22d464ebda6500161cdc24b18 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":5},{"id":"166dc27c43413786a91811e28cc3f7e3ddcf5a38","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-04-02 16:15:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6281a0351a111f5b18e9f0016d3fb25214b75212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:09:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abefb911fa674e79a7b2ea07a66638c7 : TIMED_OUT in 53m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21018f5cf7f9436f91faa6b610b64a92 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698ebe15be464893a1181fc88c5535cd : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9527a287745f4dc9908b0b42ccc5c53d : SUCCESS in 17m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/205b37381c024134ab4bdd75ac5aa03c : TIMED_OUT in 41m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab1fff2c920740158d5a002789b12d96 : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb24731d1b04bfc972388b32595fb5a : SUCCESS in 18m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fb2d90a40bac44c18d19ba4f42926a88 : SUCCESS in 1h 04m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95ae6458dfe94775977324065cc26a82 : SUCCESS in 1h 35m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a5855c71ffc14ea48d3d3680ecd65d64 : SUCCESS in 42m 25s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2639dd21d03e445490e3db8032afe815 : FAILURE in 34m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/4298db9ce39d478b8c9eb722446dc2ae : SUCCESS in 38m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/736b8bf1c8e14d3d9d1470f1e39e8ded : SUCCESS in 20m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9853b9934421429dba105c28cb96a989 : SUCCESS in 39m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fa5b28bbd774915802973e09d8b5f62 : SUCCESS in 1h 07m 11s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/1f38b710ad5f4ecbab52c421cbad6b14 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":6},{"id":"eb0e205d148f0202f3b221d195d4ca521f4890ed","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-04-02 22:04:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"459e8e76cd289c32f3ee97c46f4f42188adcc1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:10:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af4323467a724a0fad6e312b44cf92a8 : SUCCESS in 24m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c41fb3992974a52bc5b6c8076ebd436 : SUCCESS in 41m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b3e5553e6c42f4b79a1ee6d7100203 : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d8ab89bbe5e48b6a08c30968d18fb91 : SUCCESS in 22m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7e45471a4b84c80bf9b580e84976bdb : SUCCESS in 28m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ad8ba2c228d43e0bb61b728a49c3854 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa9017e88444b76a2ccabb41cabc7cf : SUCCESS in 15m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aada3f1cf33a4525bad6f29e3699a2cb : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5ab09c7abfb4c0ea3dcfcb36723525f : SUCCESS in 1h 32m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b6c2473d7a804b64aa5af910532c5c46 : SUCCESS in 33m 22s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6a1576b358894a4a82a1584bde0570c9 : SUCCESS in 42m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/f5b7862a05fa411a96da3fd6797cbb19 : SUCCESS in 39m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2ee9b3511db44edd988d100cd1911c38 : SUCCESS in 17m 12s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1174b2b3c8e84165bb30197ab9e814bb : SUCCESS in 55m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21fa44fe1ad5467396fa52ebc94cfb05 : FAILURE in 2h 03m 15s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f1162dde71f14eb5b027bff5fdceb13e : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":7},{"id":"03cf2b747333f7524018087b510d488fd44a624e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-04-03 17:22:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3f70c8401a6791484e110b20eda84abb031de527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:50:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9209774aa7934e88a095d45355179233 : SUCCESS in 15m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55fc9e670c6c43cfb47637454d11121a : SUCCESS in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e57fd4610db4f0f8b491108c17de01d : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ad4021a6c146d0b0ba1e1d65ebe45d : SUCCESS in 19m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b97191258a94fc5badd74661792b208 : SUCCESS in 14m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01bed51dd09a4034925d5deb0755b003 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5f8390eea2490faf6e714fbddb9849 : SUCCESS in 17m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1516a3ae70c34db699e403014ac2ceed : SUCCESS in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3575caaebaa64f0eb3c041fe9a5a316b : SUCCESS in 1h 20m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b76783b53a3f4c07b115ad00e198980b : SUCCESS in 34m 32s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f7cf2b833b4d4ef08db7d71a4a6507d7 : SUCCESS in 32m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/b9f832cc855f4bfe99671ff15c62e6ca : SUCCESS in 43m 14s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c322ad8f7a3a4cd282a86eb4cd6bf00a : SUCCESS in 15m 19s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/88fcaee7e3414a28b2ca680f89f318a8 : SUCCESS in 41m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3898cbed0f6545c99521632724027797 : SUCCESS in 1h 15m 48s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d055f34cd72846a2833fb6cbf6d9b4e8 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":8},{"id":"3fc25b1357ba2b39751b71d7ff76f43b48857c41","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-04-21 17:20:27.000000000","message":"Abandoned\n\nsuccessfully debugged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e614f86258a31c0014c9bb96075519d752a8f10e","revisions":{"cc4282d1a67d561fede00d47e89564efa8d5145f":{"kind":"REWORK","_number":1,"created":"2020-03-18 03:51:00.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/1"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cc4282d1a67d561fede00d47e89564efa8d5145f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cc4282d1a67d561fede00d47e89564efa8d5145f"}]},"branch":"refs/heads/master"},"b596e9b47e1b5c3676b6a91c8250042aaf379da7":{"kind":"REWORK","_number":2,"created":"2020-03-19 03:07:50.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/2"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-19 03:07:42.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b596e9b47e1b5c3676b6a91c8250042aaf379da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b596e9b47e1b5c3676b6a91c8250042aaf379da7"}]},"branch":"refs/heads/master"},"95e7d08621ee4b03fbdef7989215b673b940a91b":{"kind":"REWORK","_number":3,"created":"2020-03-19 04:41:11.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/3"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-19 04:41:05.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/95e7d08621ee4b03fbdef7989215b673b940a91b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/95e7d08621ee4b03fbdef7989215b673b940a91b"}]},"branch":"refs/heads/master"},"ed2b316a2911c9a89234d8e2c0634f6cc3e3765f":{"kind":"REWORK","_number":4,"created":"2020-03-19 18:49:30.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/4"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-03-19 18:49:25.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ed2b316a2911c9a89234d8e2c0634f6cc3e3765f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ed2b316a2911c9a89234d8e2c0634f6cc3e3765f"}]},"branch":"refs/heads/master"},"4aa4772119c1a5ed00fa03707ff8d679435313d5":{"kind":"REWORK","_number":5,"created":"2020-03-19 20:52:37.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/5"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-03-19 20:52:32.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4aa4772119c1a5ed00fa03707ff8d679435313d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4aa4772119c1a5ed00fa03707ff8d679435313d5"}]},"branch":"refs/heads/master"},"9282ffb2518e886d717e896bb08320b1aef4f82c":{"kind":"REWORK","_number":6,"created":"2020-04-02 16:15:37.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/6"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-04-02 16:15:25.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9282ffb2518e886d717e896bb08320b1aef4f82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9282ffb2518e886d717e896bb08320b1aef4f82c"}]},"branch":"refs/heads/master"},"0ec8f3be4959175b2d78aa69c6459e5dfcc546f7":{"kind":"REWORK","_number":7,"created":"2020-04-02 22:04:01.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/7"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-04-02 22:03:55.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ec8f3be4959175b2d78aa69c6459e5dfcc546f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ec8f3be4959175b2d78aa69c6459e5dfcc546f7"}]},"branch":"refs/heads/master"},"e614f86258a31c0014c9bb96075519d752a8f10e":{"kind":"REWORK","_number":8,"created":"2020-04-03 17:22:11.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/66/713566/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/713566/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/713566/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/713566/8"}}},"commit":{"parents":[{"commit":"62cd34d40d85af85b34146f9827da3414c09418f","subject":"Merge \"Copy shibboleth logs in federation jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/62cd34d40d85af85b34146f9827da3414c09418f"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-03-18 03:50:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-04-03 17:21:54.000000000","tz":-420},"subject":"[DNM] debug external idp tests","message":"[DNM] debug external idp tests\n\nDepends-on: https://review.opendev.org/717352\n\nChange-Id: I15ef258381e522f3986ad43eda7e60da2289af8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e614f86258a31c0014c9bb96075519d752a8f10e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e614f86258a31c0014c9bb96075519d752a8f10e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
