)]}'
{"id":"openstack%2Fkeystone~717631","triplet_id":"openstack%2Fkeystone~master~Ibd229b66e23d5da2ef9fc3457c2546d98827409b","project":"openstack/keystone","branch":"master","topic":"717631","hashtags":[],"change_id":"Ibd229b66e23d5da2ef9fc3457c2546d98827409b","subject":"[DNM] Openstack_groups Logs","status":"ABANDONED","created":"2020-04-05 20:22:24.000000000","updated":"2020-04-21 16:48:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6b256e40628e02dd0feb7a244ea5379e133bb81","_number":717631,"virtual_id_number":717631,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-07 19:25:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 19:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1339cd925413443217a21a8f068d4795ed567892","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-05 20:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2490177a33b64ea8101c4f3b55d29b274f879e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 21:34:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc77c27b17054cb19328a73b02b0378f : SUCCESS in 20m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b733f4dc263f4b02a6f4283e373f24d2 : TIMED_OUT in 41m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61333da3a24455ca51f212331524160 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b874f04206f4bce88392b5cc52e686f : TIMED_OUT in 41m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3347575d3020400cbd7008a895b5a9b9 : SUCCESS in 14m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6e4c676f76c412d91b6f250ce1bbde7 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac479632f4604b1e8da619bd2ccfb9fa : SUCCESS in 13m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/05e3a128c3fe4a55bd7f79feafc0cfcb : SUCCESS in 1h 05m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a5414b9437f49dc9e7a38f26c230984 : SUCCESS in 1h 07m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/335e42f37ada405fba7468f533695348 : SUCCESS in 40m 08s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f0510e1b6a9445d79d74ed60cecb01b1 : FAILURE in 23m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/aea0afd33d304dd8975ed9174a06ec74 : SUCCESS in 31m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3c49caf0b4da4cdf96787634aaf16ada : SUCCESS in 13m 37s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8afbfa7ce99c43789e01ad64dc1e95bb : SUCCESS in 38m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/432bb0b272b847e1a8f986702f3133aa : SUCCESS in 1h 10m 19s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/06dec067a4ac483383d4dbb1c746e7fe : SUCCESS in 37m 11s\n\nWarning:\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py","accounts_in_message":[],"_revision_number":1},{"id":"708ac4b1674aa306a4323bed5c6bbcadd5a48e9d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-06 06:40:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cab8e89791a6f379c8ad5331eeb62dfa3cbb260a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 08:51:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34fbefe2ab294253b2843e259605d9ff : SUCCESS in 17m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84b7b62128f144eba2f46e9118d3c727 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4eaa9b709b140dd8574370f5855714f : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b924861896564de59416f76f32e9e5ef : TIMED_OUT in 41m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/864e391685ec4cf09ec8942673a97ea4 : SUCCESS in 14m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0df549206049412b91562bf8c0775411 : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0d4fefa5774159ae25f352e50e3a4d : SUCCESS in 14m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2df6f224338640aba6da5d09c174c8a0 : SUCCESS in 1h 03m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94d918b55f124b0eb4a3b52c4accc83a : SUCCESS in 1h 29m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/458c8f5b515b45fbb836b13a242e749f : SUCCESS in 34m 23s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/086623606f2b48fb85b18fd43ef7a209 : FAILURE in 35m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/4b2915b667c6441b8a5698b17404ce91 : SUCCESS in 31m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9f50ca31153c4431b646c77f31231970 : SUCCESS in 17m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6712d24150d04502af5df5370808254e : SUCCESS in 40m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5058051e76d04eca96afa54b9d261c89 : SUCCESS in 55m 22s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/01ffb9c34f194403b7aa3a2bc30fca13 : SUCCESS in 38m 09s\n\nWarning:\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py","accounts_in_message":[],"_revision_number":1},{"id":"6312551a85b1b5645d18738dd573ea431b41e468","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-06 09:45:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f62adb0330b7a0b847175dfde1c88e3e291cf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:57:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/667afa5c484044faac39328b246f5676 : SUCCESS in 32m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed2edb55f3524bd9b3d40ee6e1c098ab : SUCCESS in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e9f044a91f4935811b103c422b2cef : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/013656452bb44b639594f6d428114d66 : SUCCESS in 15m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b7acdec7413420f8323a35e82b47738 : SUCCESS in 28m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d6186ad97b40a6b9e12d447a482a27 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10f6746cdca4c17b3ca9e893913b3c0 : SUCCESS in 15m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e4448c3323e47dda30ceaafeea5f929 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5d0563fc6df4746ad001174dee315a0 : SUCCESS in 1h 29m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1122bf96c0cc40feaa0f4c0f1e333499 : SUCCESS in 37m 25s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f5ac9cbb63de4191a20d1f7701c10df4 : SUCCESS in 1h 11m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/69299782565549d9a4b2951f7dcc0357 : SUCCESS in 39m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/96a82601d74c4659814b07d80fc6628f : SUCCESS in 20m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bc499e682cc64023a3a47fb3e678b52a : SUCCESS in 42m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8df4fa808e1402aaf2934952e95d7f4 : SUCCESS in 1h 03m 41s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/3a8340b9cd6d4d9ba552723b665b58f6 : SUCCESS in 41m 16s\n\nWarning:\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py","accounts_in_message":[],"_revision_number":1},{"id":"d6935314277d11dd5d455f37af582240b58f9ef9","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-07 11:10:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"028a63c0672b0e8b6ea705f5b9cb6fc34b012a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 19:25:42.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a137131964544d9a6c18552a0b07026 : SUCCESS in 48m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1dd7f5391424a0ab6b24652803e4902 : SUCCESS in 30m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd326abbd694f0791efe02d168c48fd : FAILURE in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6d1a5f3fb840b68c96d4422032f10f : SUCCESS in 17m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0b74408b03a4461b5a152a550e22622 : SUCCESS in 16m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be6f26fb478c4357a99cd505780eb5bc : SUCCESS in 20m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e58333b97841abb8e7cd8de2925313 : SUCCESS in 16m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e869763d2844ba69eea95bd9334c99c : SUCCESS in 1h 13m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/243bdd65babc43b698ca609c54b0c4c9 : SUCCESS in 1h 50m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1b209cefc37a472fbef20ad673402518 : SUCCESS in 32m 34s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/81f62da20f944574bba34526e7293af4 : SUCCESS in 34m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/fe1da91fd7ba4942a9086dfefa3faf27 : SUCCESS in 48m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/75d0dd17385948fd88f77a86016e4c18 : SUCCESS in 21m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/070f15f4716341609b91216da5c66ef8 : SUCCESS in 47m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dded4f8755e84381b9dcc9e6a85911e3 : SUCCESS in 1h 03m 38s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ce1a1b6a4a6c4fe2b2a6d0a2502967d5 : SUCCESS in 40m 53s\n\nWarning:\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py","accounts_in_message":[],"_revision_number":2},{"id":"1c86747622736296bf381fa8bf62f9b389ccf079","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-21 16:48:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8669fef3e8c7fbcec57129c453ca0330ee6827a","revisions":{"f89564c13abb99f7c2f2b22e3eaefbecb47c3b8b":{"kind":"REWORK","_number":1,"created":"2020-04-05 20:22:24.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/31/717631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/31/717631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/717631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/717631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/717631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/31/717631/1"}}},"commit":{"parents":[{"commit":"389d8f5a4edbbe05e69740c00c50ff3fbaf36d51","subject":"Merge \"drop mock from test-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/389d8f5a4edbbe05e69740c00c50ff3fbaf36d51"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-04-05 20:21:03.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-04-05 20:22:16.000000000","tz":330},"subject":"[DNM] Openstack_groups Logs","message":"[DNM] Openstack_groups Logs\n\nChange-Id: Ibd229b66e23d5da2ef9fc3457c2546d98827409b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f89564c13abb99f7c2f2b22e3eaefbecb47c3b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f89564c13abb99f7c2f2b22e3eaefbecb47c3b8b"}]},"branch":"refs/heads/master"},"a8669fef3e8c7fbcec57129c453ca0330ee6827a":{"kind":"REWORK","_number":2,"created":"2020-04-07 11:10:40.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/31/717631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/31/717631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/717631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/717631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/717631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/31/717631/2"}}},"commit":{"parents":[{"commit":"389d8f5a4edbbe05e69740c00c50ff3fbaf36d51","subject":"Merge \"drop mock from test-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/389d8f5a4edbbe05e69740c00c50ff3fbaf36d51"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-04-05 20:21:03.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-04-07 11:09:58.000000000","tz":330},"subject":"[DNM] Openstack_groups Logs","message":"[DNM] Openstack_groups Logs\n\nChange-Id: Ibd229b66e23d5da2ef9fc3457c2546d98827409b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a8669fef3e8c7fbcec57129c453ca0330ee6827a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a8669fef3e8c7fbcec57129c453ca0330ee6827a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
