)]}'
{"id":"openstack%2Fkeystone~722633","triplet_id":"openstack%2Fkeystone~master~Ib79a6454dd7070832f639469767a807eebe5d12d","project":"openstack/keystone","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"Ib79a6454dd7070832f639469767a807eebe5d12d","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-24 08:43:01.000000000","updated":"2020-05-06 05:54:42.000000000","submitted":"2020-05-06 05:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722633-1588744339854-62275d76","meta_rev_id":"e09ba4d0a19bd18bb098b1cbdb740d7db3fb1a08","_number":722633,"virtual_id_number":722633,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"date":"2020-05-06 02:54:05.000000000","_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 05:52:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2020-04-29 13:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2020-05-03 03:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2020-05-02 18:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-28 16:54:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":2,"date":"2020-04-28 16:57:06.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2020-04-29 13:07:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:54:37.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-04-28 16:57:06.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-04-29 13:07:08.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-05-02 18:59:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-03 03:38:45.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2020-05-06 02:54:05.000000000","updated_by":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"reviewer":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"state":"REVIEWER"},{"updated":"2020-05-06 05:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6db1879afa72e0a6116b8c5af0c721c5e7573880","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-24 08:43:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"220eca69f8cb520a6b1d0dbf034cf4a7870dd47a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:59:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b876029dbe44729855975a808d903b9 : SUCCESS in 44m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8111274da4114b64b14a8387930dbe88 : TIMED_OUT in 41m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523f793ab0604197a5592865ecf09146 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bc8379964c04af6bda79e3045a2e7f5 : TIMED_OUT in 41m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7924bccaeb80406aa9a92983ae5aa038 : TIMED_OUT in 41m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d09b0246f2403183b63e709c625c6d : SUCCESS in 18m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c9c71494435a4d4d8014ab01054d6406 : SUCCESS in 1h 24m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21dee161d5f24764b7be8011429f34c3 : TIMED_OUT in 2h 08m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6914cc16771045b59747dd750e110a56 : SUCCESS in 43m 58s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cfeb40f67eb949f99cd5a7186d6ac499 : SUCCESS in 36m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/11f435fa55bc419c87e1f27fd39c2964 : SUCCESS in 44m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/14440334835a409a8eb09f3c9931c62c : SUCCESS in 18m 16s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/95a09f9487cd4a1f9cb4fc25c738d4a2 : SUCCESS in 53m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b0df2e81b849528626c34369fa419c : SUCCESS in 1h 47m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/efbc1815dae44846b06cb5101667581f : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":1},{"id":"42065901993be491b562aaef17622bcb99c8f1c5","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-04-24 15:16:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83b0ef0c855c2bc00e86b1b780f1c5156e1e8f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:22:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c23b8f267fe4c9cb912aac0c640e0fc : SUCCESS in 27m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1c4d335e11e4f4c8b737b48641fc8f9 : SUCCESS in 25m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a1e77f8640434dae4ac0782b31b882 : SUCCESS in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6637db275084c95a5f4f57cd40bd474 : SUCCESS in 24m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7bb4cdb257e42a08b563022b88cbceb : SUCCESS in 17m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bca9d49f8ed4d8db9c486b9e9451275 : SUCCESS in 15m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b83887837ef2480bbb303b52426c93c9 : SUCCESS in 1h 13m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d8e2e96a64c467f8e5cc4bef63c8668 : SUCCESS in 1h 24m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0a8c235cc9d64877be18ea590fbb1c32 : SUCCESS in 36m 13s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/07481ae0c4b84053b92e19b4d24a0f7c : SUCCESS in 39m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/c4c7a0706307440ba79302c9729a7db5 : SUCCESS in 37m 24s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c443c7d308ae4434a2f4fe00e9ea953e : SUCCESS in 18m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bd41fb5d2fe345fcaebeb1a7234e8f08 : SUCCESS in 45m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ea2c1aedd4143b684f085e876128892 : SUCCESS in 1h 02m 08s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/5b9909dc644d4566a17e51b7c7e11b88 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"44c38415ac8c0364a495cac07edc4d8619241300","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-04-28 16:54:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"841cd3ccddcf0ae5609ffb5e021b402b927d2c42","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-04-28 16:57:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"399946dba4c8dabc7722720c19837459be92bf8c","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-04-29 13:07:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39ebdad3735d271c8416b15803ec57262b83d777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:08:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3880f22404cb1e9226f73aa4869ffafcefe37196","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-02 18:59:30.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck - event lost","accounts_in_message":[],"_revision_number":1},{"id":"930274b17600593299a5c259d501027a0f537ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 20:23:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d4ad70d565e4b7dbd19cdb454730c26 : SUCCESS in 15m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d9e6db0333e4e5999cd6d6999952594 : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0e1a686afd47d0b7b7f35603beca5f : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5389dbd6b60b4f6c817227a546ceae5d : TIMED_OUT in 41m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58a627491adf4f1a9f8dc464011952fc : SUCCESS in 12m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b99fe70a46de49e0b418b2f3e2df4e88 : SUCCESS in 17m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/542e1a4d1de04d22851264bf7af2680d : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e8cb94472b045de80b99021d3e2e1b6 : SUCCESS in 1h 19m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b913e8453062478ea96e9b497a718ab3 : SUCCESS in 30m 39s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/abbc8affdb1040199f13957c6d7d271d : SUCCESS in 33m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e63031b47d0347fcb8accde593302daa : SUCCESS in 30m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2e43473ad0544441bf58225c908d75e8 : SUCCESS in 22m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/43bd3532132749779ad59ada6db5cba0 : SUCCESS in 44m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04c302b2e62d4131b03e04e90ae9d3aa : SUCCESS in 1h 15m 14s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b2d635bf905341838421cd77b83f7b6a : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4d2c70eaf80f19a15f241b0d4f81a53da36925b7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-02 20:26:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"346319b26c04947b462ed61a12a261be877e89c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 22:01:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efb4596e8a964150be216e2e97ef93ab : SUCCESS in 25m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdff485c8b7c41afbea136ac18e89aa5 : SUCCESS in 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00fc695ab2ed4112957e6414eed53e1b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/096de4886e934368b168aa91aec93ca8 : SUCCESS in 24m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8037b6290f3148f5a5d5631d3f3c015e : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5204eccf82bb4196918e5ee83de820eb : SUCCESS in 15m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3d2a23611b2e4d4ea18db2a07b2ccf0f : SUCCESS in 1h 10m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aecf17e6624640c2af78ec561d7a974e : SUCCESS in 1h 32m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ea83fee6e5544ceeab3c52addae3d245 : SUCCESS in 29m 46s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0535371325c64fdf8c492d1cb90283e8 : FAILURE in 40m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/27c62c196db1489d80752be65a7cdae8 : SUCCESS in 40m 50s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/cad0a9991e1a497b93c8054d62dd4867 : SUCCESS in 14m 18s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4a1646c2420740318e2b38426130f237 : SUCCESS in 39m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4902bf5bfe844129fa4df00d296b8dd : SUCCESS in 51m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/47d1df9591424d1fbb3fadc5e436701f : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2fad4e9a25c2fc5cf55d66d3ff7845e2942acf5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 22:02:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccef5a965b2d9668c50c7d7e56d4e48d9fd639f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 23:26:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67318850606b415fa90ed4a4bf978353 : SUCCESS in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec754e5f7cba42e3995c178748c4a9e7 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98df59090e74d2994c354f672f10b1a : SUCCESS in 33m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cfb52b564b548bcad8c0b3aa8539e33 : TIMED_OUT in 41m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03801f2ad11c4b7b83da525f43888cba : SUCCESS in 11m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/437015503bd54c1093fba99fd043d6da : SUCCESS in 1h 02m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d6735377e8a4598b070bc2bacf79ec4 : SUCCESS in 1h 23m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c9b7b4354d5b41c9bc7bd94db6b5ac88 : SUCCESS in 30m 40s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e75b16c96d78491a839a6ad4061a4359 : SUCCESS in 38m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac1c6e2c537a45fda34608e02fd9ee0b : SUCCESS in 1h 00m 31s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/0119c1c2173c4ab5bc6b6b4747447a67 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bcb95e87099514b67d6255ebac9423610951cd1a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-05-03 03:38:45.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck timeout","accounts_in_message":[],"_revision_number":1},{"id":"12181f319c8b69323b75d05b8e37e0681fb40020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 05:17:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34bc423be09d414680cb519167d8f9f5 : SUCCESS in 16m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/772e8314c6a8444484005750ea3d3eb8 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5adbd1f8a6544e01ab4b15827b39a74b : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23fc774f19054c4f9063865609af0152 : SUCCESS in 13m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fb1ba3ec0d74e71b1e71da29669fcc1 : SUCCESS in 17m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fc602b552142b5be7e1ac0b128a502 : SUCCESS in 12m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/75c8a29586f1415288e675cc3f6e0b90 : SUCCESS in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/945b1f8056c7425bb07f84a1073685d9 : SUCCESS in 1h 12m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0060be0d81bd45038264747abcbb2e11 : SUCCESS in 28m 02s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e346bb253efd434e970a5d4d09e999a3 : SUCCESS in 33m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/84dafa9426a0450784bfa63f8d057b17 : SUCCESS in 41m 24s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fd993c4d8e864b1887adfead1a53b9b2 : FAILURE in 13m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0f282176cbc9428a8d81b26881cca5a4 : SUCCESS in 36m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c20bb72509cb4d3882d4b3e631799143 : SUCCESS in 1h 38m 10s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b967be879ae546b8af6183db06fcdf68 : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5c8eb22a0c201326f219b50913a5b819efe9d517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 05:18:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a599b9c1829580f9b590d478442490c2147ce97a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:37:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a1d29554723492c895691acc195c683 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7baa524c854edab69736b6c61ccc7d : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a4f407038fd4e01acebff2b560b6bb0 : TIMED_OUT in 40m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b88001df0e442d28d4caf41d8f0f65a : SUCCESS in 13m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981d890279c1483eaf189c4d94adc0a4 : SUCCESS in 10m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/96513bbc8f924c248710db871b118f5f : SUCCESS in 1h 06m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38307aa4b3834a269f7b6ec0648176f4 : SUCCESS in 1h 16m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2109370a46a042ccb4d3578be42678dc : SUCCESS in 30m 15s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/3135d483f49a432d92d2d1694d161f35 : POST_FAILURE in 27m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22cf1944a1a7413686416b15142ca4f3 : SUCCESS in 55m 59s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/5196b9a7aff441038438cab078367864 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f7823ae0e4e51d7244704e142bc63597c60805a4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 08:15:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c0e8b00ca5e8daf861ae475a01e96f47e4917320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 09:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d259e31db514c8a835e75399dcd313e : SUCCESS in 15m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7083dc419b64a52967ca65850960b4f : TIMED_OUT in 41m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f3a581c53549e598bc10b42163c705 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810c2d8bde6f42e59ca9bbc9a30509b1 : TIMED_OUT in 40m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9583083dbe7f48d59dfbd7b4c5988314 : TIMED_OUT in 40m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34469a76e802444faa36467c6ef17cbc : SUCCESS in 12m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9cfeba4e54fd493190a480ed07f1010f : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e70eb34aff9415081eba8cc6624d32a : SUCCESS in 1h 12m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b3f3c961ae274f22922c9c53db132d8a : SUCCESS in 32m 21s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/10f4b2a376a24177bd4bfaeba9951101 : POST_FAILURE in 32m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/c1925fb50c58440bbcaa7accbfc4c720 : POST_FAILURE in 35m 39s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/24b753ac7eb94c46bc1dec04c7f5c01f : SUCCESS in 14m 04s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d0e770cd21f643909527dbd734875012 : SUCCESS in 38m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78b19f4e61424a2f9837ea35c6aebe4d : SUCCESS in 1h 01m 56s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/c6228118da814c968c286e42e0eafbe8 : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3fccd7bc60b209740940b106a3801c88a1c243fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 09:42:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e70dada3bee5f29daf60e01e4841e23713ceeb7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 10:58:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58f1959efaa942c8b77b6dfd114f629c : SUCCESS in 27m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6ff690fdb434161b6fa17ae2237a764 : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728e5d4438c9442da3841da743ede7d0 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b8bcbd41f46446195b73be8ccaf803f : SUCCESS in 21m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e159897e60c0457888698d220ff4dce7 : SUCCESS in 18m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/607fd006eff6436aa873c6fa36bc0ede : SUCCESS in 12m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/854611e583c44b658d7d5f28777ea48e : SUCCESS in 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8187e989ea2840bdbaf3401588d0008d : SUCCESS in 1h 14m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8e862efb442e404b82331715a35b48b5 : SUCCESS in 29m 51s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ba1ce9ef0fe74b34ab497d829de3d4e7 : POST_FAILURE in 33m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/30c8c0d981a94b77bf50635fee9a2f90 : POST_FAILURE in 27m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/93f7b03456264b26877a0c587cb46b13 : SUCCESS in 13m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fbe0d573b1224b1c824642472bb900bf : SUCCESS in 34m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f95e3f882854acd9da62b6ac82f3f4b : SUCCESS in 59m 16s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/e916ff8174e64f6aa43d423b8d9d0c71 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7170c8dfb103c3357e0f4d010e1c72c4db5b47b5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 15:16:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3503be538ba232230ad2f9c0fcfd6e0553e24ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 17:11:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bbc41b393a948c8803eda40006c1239 : SUCCESS in 48m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c594f82a5d3e4acc978ee4d5cb7dcc21 : TIMED_OUT in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad0dcda5a32445c96cf57a0df94a233 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d706f4f3d56e4482b2c3a0ef689b453a : SUCCESS in 34m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3180e6598b9241cb836c8d3fde1482bc : TIMED_OUT in 41m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f621624f3f33456f805c04dbee37f09a : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/d6d013b43e6b402784d3c7e661727656 : SUCCESS in 1h 02m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9bc380ea0104d3e82cd3366fffb58ba : SUCCESS in 1h 50m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/384937b6b12347bb8ad5d38f31c62501 : SUCCESS in 29m 06s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fbb4ed0f5a2f48ce8b52575e791576c5 : FAILURE in 35m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/6ab073701cfd4717a7550e170638aba7 : SUCCESS in 32m 14s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/080007ee89bb43ebae696482ddf38e52 : SUCCESS in 15m 45s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6e647409da574286a2987eaf8d8dcb8a : SUCCESS in 44m 29s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ca6861a6a80468c95089d4506cb9a96 : SUCCESS in 1h 07m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1018f5f485e64b198914c438180622b0 : SUCCESS in 53m 54s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/0ef9585a93304a309bd74ae18a51d174 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cbefa2d7658e7b98d52d86c9f318a32705554b93","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 17:19:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46d1e4235cb25b2b29ace4788c51f45908ac8d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:27:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0936cd0ab341d99fcd27a856a40f68 : SUCCESS in 17m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f14d407ef1c74a85aec523c95fb55a99 : SUCCESS in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda164af7beb425980e340e9465ae59e : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d85aa7ac674f2393f21aa0511cba67 : SUCCESS in 31m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14a5a19402cd4134afafc02d5e60540d : TIMED_OUT in 41m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea71c634540d4e898eacb7268a84dcf8 : SUCCESS in 18m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/8bc7d7ecffeb46d38726d5bbd67671b6 : SUCCESS in 1h 10m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b69998b85dc24f0c8bc6506e3baaf6cd : SUCCESS in 1h 27m 37s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c4f487f9ec2647379e9baa697a406f6b : SUCCESS in 39m 30s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/40699c5ecccf4e37983adbeb4ca19bc3 : SUCCESS in 41m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e22a7d8bb4454b14a60420407655c2df : FAILURE in 36m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/981868283519416f8ea213ad9e337be5 : SUCCESS in 17m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/52aa0c8a5c964b2db21f161ec8af464f : SUCCESS in 42m 30s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0406411e236740aa93da8c6435c4bdce : SUCCESS in 1h 24m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14f398b4fda3474b9278bed8e95ffc96 : SUCCESS in 1h 20m 14s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/83da6718255c4c94aa053cabdd9c81ee : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":1},{"id":"864e9ca48baa16385af04b058901a50f8d00bdb8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 19:31:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58cd5d1cf054f283e803d5a102456beadcf0af28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 20:59:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65917118f6646448f4e8fcb55f34a09 : SUCCESS in 21m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ccd660e9e2a40c3bb6d53bfd5a3779e : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f942051449b409ebd867a5952df08da : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5146a3187924f4f96089861e2bdc493 : SUCCESS in 15m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac7d3c081e6e48ca82c52a8251c7e637 : TIMED_OUT in 41m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e69cf578bbe48bc8eb0df1a2b12efcc : SUCCESS in 15m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/152b6a14dd5b46f48f5f422e9c37f2a6 : SUCCESS in 1h 00m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/068f90b9f38741f3997bab3a4d18c34b : SUCCESS in 1h 21m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/95374df9738a456a82707f7b9beac982 : SUCCESS in 32m 47s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/40211fe935a54cfcb3a96a67393fdfc5 : SUCCESS in 31m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/cd84a32fc6154efb8ed25bfd2505675f : FAILURE in 29m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f120d73412d34077b4c1892deb4836b5 : SUCCESS in 17m 52s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a1ee215076e94db8b338cb236af235eb : SUCCESS in 43m 13s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1bc586970ff741cfabfdf9175f0d5144 : SUCCESS in 1h 00m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fec6c21e954c41bd9dd809b0b7001575 : SUCCESS in 58m 59s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8e65be57178d44d1859709d848cd62dc : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1c8796a5f27546fb214a9684876e1c1c8bc9834f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-05 06:40:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe36ce8001f368d1e6fb3dbe10fa59dca1267ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 08:38:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/541f34ebfdb74b948a8f7d1f8c3db565 : SUCCESS in 26m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbc3e4e01f2047f1ad8ad657fec94de8 : SUCCESS in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd08b5f91c1b483da05b548295146711 : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfe60d0197d4737a5cb9da86cd57750 : SUCCESS in 28m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bff8e0135d14d40aabd73340b47c41c : SUCCESS in 25m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5908641a6d624eacb8f583a88129bff6 : SUCCESS in 17m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/9df6702d8fb64001a8f1ae73d44b9471 : SUCCESS in 1h 01m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/418a6ca990314e5797930eee38ecc279 : SUCCESS in 1h 24m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/39d520c975c04f3581a36a82e58911b0 : SUCCESS in 32m 11s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/eed267dc30e84b3d940c28464267f056 : POST_FAILURE in 33m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/3a52a0c534324833912cc53cef9c6a68 : POST_FAILURE in 36m 15s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6dff48adf09149c9b25b198c12e49fea : FAILURE in 15m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/610bfc8f810947eda983a6ab05d35b14 : SUCCESS in 43m 14s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/820caaf4f89c4008956d4510923e8dd9 : SUCCESS in 1h 02m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7a5b9e28eda4b85b3f59a672033413f : SUCCESS in 57m 57s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/3a374d456277445cb51ce786c4b3b3fd : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d3b1bd50e03ec613b4e75ecf6692ae8df2652240","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2020-05-06 02:54:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54e3b3568985898ea0137ad2bebe8c3b5c4adaee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 04:25:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/506434541a7947be89d80f07f023f9a4 : SUCCESS in 26m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5ef1ab0c27042e18b74d904835c25af : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6433bd1570e4260aeb8d2e6f182815b : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0bce89137ff46d2a3a149179cc70102 : SUCCESS in 30m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e3408e97e874265a9421ba44e92dc37 : SUCCESS in 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aba10de5d2243c0889dcf47ab68cc10 : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/4d5b748b010c4bae9e38adc2e743a885 : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/988c72d115364e75900b8f42d0341497 : SUCCESS in 1h 27m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d1ffad3214d1444a90f4c6ea343a1c7b : SUCCESS in 32m 16s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cb44e07cddef44a98b375247fa19ffb7 : SUCCESS in 32m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/9f83d1f6c691435995ae103828bf619f : SUCCESS in 35m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c5884300c14f424eb1a6646851fd1ece : SUCCESS in 14m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e33c139322c04a55973a27bf3f23b5af : SUCCESS in 43m 01s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d88c48a64a8d41cda0b486aabffebb82 : SUCCESS in 1h 06m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e546e5c9a64026a218d698c6f7851e : SUCCESS in 1h 04m 29s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/e1ca34b3ee5f438abe7dd6060397838e : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f4d00483a691f1bf470e3bc4722ab6418278fe14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 04:25:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bb86b86959e3fe36d9e5dead7a35b39623a655d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 05:52:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78fc8e9fbda945b897fbd53b97301a1c : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79952d88c193457a9e726f9e4a70d8db : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07b3de38e23a484ca6e9977e8fa7b57f : SUCCESS in 14m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762d72a2991d4908b3d0d2d994a84713 : SUCCESS in 13m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ca2f9da94d4aa6a9967b8b3c7c34ae : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/255e25027ac34433b85b42ebb7c1094c : SUCCESS in 1h 01m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30d1117e5f3142b88844b5558dd85014 : SUCCESS in 1h 22m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fdddfff9c7cd484ba40efdef78fe594a : SUCCESS in 30m 25s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/9bc61ab1bbeb4997a202a06ba30a1715 : SUCCESS in 40m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/007321e0e16b475eb8ed09ca743dd7bc : SUCCESS in 58m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2409d58bcae5416abc537b2fa69ad709 : SUCCESS in 1h 08m 17s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b33222f8f2e84e4caec29dd14bea8437 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"590e86723efdeb59660263c817143b4e36fc6269","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 05:52:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1186214c1e281939067daeb2906eeea93f1299cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 05:54:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238b064fb0cd47ef991e14a90c4774c2 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a20542af3ae190629aa11e4ebb49c56cb02b0630","revisions":{"a20542af3ae190629aa11e4ebb49c56cb02b0630":{"kind":"REWORK","_number":1,"created":"2020-04-24 08:43:01.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/33/722633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/33/722633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/33/722633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/33/722633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/33/722633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/33/722633/1"}}},"commit":{"parents":[{"commit":"0b26e64f45def5b2c6e8dc03258cdd0649c9b7d9","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0b26e64f45def5b2c6e8dc03258cdd0649c9b7d9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-24 08:42:59.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-24 08:42:59.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ib79a6454dd7070832f639469767a807eebe5d12d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a20542af3ae190629aa11e4ebb49c56cb02b0630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a20542af3ae190629aa11e4ebb49c56cb02b0630"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
