)]}'
{"id":"openstack%2Fkeystone~724954","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~Idb10267338b4204b435df233c636046a1ce5711f","project":"openstack/keystone","branch":"stable/train","topic":"SEC-bug-1872737-stable/train","hashtags":[],"change_id":"Idb10267338b4204b435df233c636046a1ce5711f","subject":"Check timestamp of signed EC2 token request","status":"MERGED","created":"2020-05-02 03:51:35.000000000","updated":"2020-05-03 07:56:47.000000000","submitted":"2020-05-03 07:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724954-1588492489707-6bb723e0","meta_rev_id":"17e5ca7ea11d555bfe635379ac95e59ed8d6f17e","_number":724954,"virtual_id_number":724954,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-03 07:54:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2020-05-02 16:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2020-05-02 19:39:30.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-02 16:08:06.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-05-03 07:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bdcb3465866a709673f5bc97f7b0442dc5c4faa","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-02 03:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88a17741d954950d9d8c69e65eb603b5397576b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 05:22:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e35dd15232c4fbf9dfc224929875758 : SUCCESS in 14m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08fdc3d561844f2fb7b237595f3cf689 : SUCCESS in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ecb2c60a864364a997b787f47180c1 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3ec9d9b681b452692d66c9f8a9492ef : SUCCESS in 17m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3af9ccd77153454b9373aabc4e959838 : SUCCESS in 20m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fedbf48888884963a42ebdd903f495d5 : SUCCESS in 19m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6143566384414e518cd90805a489b0a5 : SUCCESS in 12m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/983cb96127a7440ea23d1acfbe7da1ef : SUCCESS in 1h 24m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd97fe3cb46a4727af7ae7a896967823 : SUCCESS in 56m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/037ce912927041968988903597830c25 : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/050fc6089fc24f9fbb4292d5a5244bf2 : SUCCESS in 1h 26m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6686b1aaad0448c698cd354dd839f36f : SUCCESS in 17m 02s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/27445981d654402bbff0346e705a3c91 : SUCCESS in 29m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e779d262e7304712a423a9905b97c63d : SUCCESS in 33m 18s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2a06af723898424ea400dce9ecb7b44b : POST_FAILURE in 34m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/7220ba33800c42f285d0aba650f65d36 : SUCCESS in 37m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/76d7e3642dbf45cb8fac18001e77bb29 : SUCCESS in 37m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9a7b17f214704314b964e6ee4e9e7db6 : SUCCESS in 16m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8993b070e72b456d9bdbed68b7941535 : SUCCESS in 38m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/054a999114934d49853f29b6c139a34f : SUCCESS in 1h 04m 03s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/eb733f82cda8466da613ec1d0c64fe47 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0303a443588ce4a7544b5706751573fe2a5a70b2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-05-02 16:08:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8edcbab7315217bf917f065851a4f67462ab483","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-02 19:39:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be374bb0431a8f74f58c7201a0e9025d539a8097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 19:39:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"857425f6df6f5e6154af9e83ce64a0d0bcd0b7cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 21:28:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8304526fc53f4d3283962e19562e2f87 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f570c49c6c441294b915c37e4fe7cd : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99986593e89b42fc93fa6ca489022fa8 : SUCCESS in 12m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32a0019f477b4bc1b0cf6c1c84282637 : SUCCESS in 16m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85b6e1d5e1b3451192e06fa5b93cef98 : SUCCESS in 20m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad95a296b3446558e68e4ff7dafa704 : SUCCESS in 12m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0bfd2a35bdab492295c7c7ce1eb6f4aa : SUCCESS in 1h 39m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc9246ce761b4c96869073261015a1e3 : SUCCESS in 54m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9e82c128c2e243ddb224233b9311ca30 : SUCCESS in 1h 02m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/902c53103a1a4cc38289011e8763cb5d : SUCCESS in 1h 25m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c417a238c7354fd18558e7af6777c9ab : SUCCESS in 19m 33s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d55b912177e04fabb4fc23c5724b2c1a : SUCCESS in 31m 14s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9eb3f2f632f44ab4a1295cef05b852c7 : SUCCESS in 31m 35s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/ec73071f97374a7ba3863257883e6398 : FAILURE in 37m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a601ef116b5492f9abec0e4a4861340 : SUCCESS in 1h 04m 01s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/0f94085be6834cfa90335678842114cd : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ffe21035e8439a09e6c2d0b2c8ad49e433805137","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-02 22:18:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62a3ed5eb81794d63743e25c4f026296d3877e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 00:02:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce5ea222aeb248998e1c92f30f67e2ea : SUCCESS in 16m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87fdf4911277471aa7792c45060d458f : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c49d9cc73cd41d2bf039b1d2aab64d4 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/543d05102e76449299ab6be5090073bc : SUCCESS in 30m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbbd3ef58f324e64a8f5e308a780d7c6 : SUCCESS in 13m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7e5cca4636c4002ac9aedfebe8bdd39 : SUCCESS in 17m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24de41128386477e84b66940647cc534 : SUCCESS in 13m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c3ad928580b14af9a9284882642d8dc9 : SUCCESS in 1h 36m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9f28a1a93c544829ace2df772dbe7c08 : SUCCESS in 1h 03m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9f0ceec593f944f78be4946b53fb074b : SUCCESS in 1h 02m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfba463029434780a765b94a97796f53 : SUCCESS in 1h 17m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbd9a07b46514f83960a09cfbe77b111 : SUCCESS in 16m 34s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/79aa3d811226469a96d651c2541c28d4 : SUCCESS in 34m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/38beb67ffaab4149b98377dd4493f430 : SUCCESS in 42m 33s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ea63a70f8c0844828a43dc54a611d651 : POST_FAILURE in 37m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fdec5d44b221490c8b3445fb2715d54f : SUCCESS in 43m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/578b6a24030e419891566acbe5e338a8 : FAILURE in 37m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/559aacfc0f664a358d1a5618c767fbec : SUCCESS in 21m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/562f797e5bfb47c390fe6d12f12dde0d : SUCCESS in 50m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/057bfa1988f8484d8756d049b4717798 : SUCCESS in 58m 39s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7ce47deb74564196be58d058fd8ae2db : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1aa4128f4e3b30e9c6d4c2ae8e44001eb39bd002","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-03 04:33:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26436ffcad298d5d8409fa6797f492dc847bc9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:22:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e05409ee2a547539a4a6efd62749dce : SUCCESS in 28m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25a54e6524f94c1a90f5621b20bc984d : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f279e83d5bc3451e88af90bc6bf1f662 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5cba97bcae1404db534c6e444fc1584 : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db785f40b0a9467481d4593b4ff13aba : SUCCESS in 13m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/028a64be4df4451fbc5095626d359c9f : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44be2c15bb534421bb9bdcacef52bce3 : SUCCESS in 10m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/093edcd00ac04faf83ba4a05d8b23a0c : SUCCESS in 1h 34m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c5bf016293304810a59a31c9a9a3fb8d : SUCCESS in 56m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/934504ec291540ddb7de4a47f3683ddb : SUCCESS in 56m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcaca8b99b994cd08f623c6395738d76 : SUCCESS in 1h 31m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2510a897cf5489f962675f4feed4695 : SUCCESS in 19m 28s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/be71808e38a54ff0a91adb94e70c7318 : SUCCESS in 30m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a4db3e4f60854403bdc8ac04fa4da416 : SUCCESS in 34m 20s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/a58f2e7b4c02448e9b4eded5f2de0b22 : POST_FAILURE in 27m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1094515803c04f3ba58d16844e630839 : FAILURE in 28m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/a2a7965a621b4809a1696c2818e4fd6a : SUCCESS in 40m 17s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fcf4ed8f113b40e5b887d97958897892 : SUCCESS in 12m 44s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/454b5c572c1249c186f0375658d9db6a : SUCCESS in 43m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fe3c81603904261ac2eb56c34478cea : SUCCESS in 1h 03m 32s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/1fadcb4e1ffa4fad9e6c3f87512d6054 : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f853b44f22def4101bb69b4fc13460fe0f3c5d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:22:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5da9e97972a17af708e74d4c6246563b62fca015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 07:54:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e54e4d9b3eee946e169dc7f8ed5c5590a003e354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 07:54:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751fdcd8a2a6463289f4653a7947f2b3 : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59317d4641c64893b32e1da19939e433 : SUCCESS in 7m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e46ba871943a4efaa40b6d731579107b : SUCCESS in 20m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14be420e2ae44116a99ac4e44214aa41 : SUCCESS in 37m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5f63b6640cd4ee6b85cb3d5034aff37 : SUCCESS in 17m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961f3aa3236146f99b3e201916d93bf4 : SUCCESS in 13m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba81cf49438047459c7fb22ac6ee473a : SUCCESS in 1h 27m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/626c72ca0bd64c1d9fbd79bcb9d02b88 : SUCCESS in 59m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/402b80d00ce5480693ed4f6a4a648cba : SUCCESS in 1h 00m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e2304fffd224a21b5baf80981d35472 : SUCCESS in 1h 26m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25afda2ba7a9494bbc2016b53101d166 : SUCCESS in 20m 23s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/15f2661df92c409380fcd1f6c47da861 : SUCCESS in 33m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f8b2cc805546429caeef87acd8a106c0 : SUCCESS in 36m 12s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/2283629643c84591a800c7ffa1c94fb0 : SUCCESS in 41m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58b9bea060ee4b72a03246a2e177dbfa : SUCCESS in 1h 10m 27s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/fb28a95b1e384e1a8076fbac7e7dce24 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a5baa77c97b779379b265e582e570f5744995981","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 07:56:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c6cec96e0f4847bef05a039d0abf7d : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3660a8825a6b4cd8af805aa4edcc7348 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3f65d6fbcd18032a8ad3dfa3aaded264a282158","revisions":{"e3f65d6fbcd18032a8ad3dfa3aaded264a282158":{"kind":"REWORK","_number":1,"created":"2020-05-02 03:51:35.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/54/724954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/724954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/724954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/724954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/724954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/724954/1"}}},"commit":{"parents":[{"commit":"40b7de87e04fbc49ab3a96b5114a9ab09c2734b7","subject":"Tell reno to ignore the kilo branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/40b7de87e04fbc49ab3a96b5114a9ab09c2734b7"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-04-17 00:05:43.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-05-02 03:51:04.000000000","tz":-420},"subject":"Check timestamp of signed EC2 token request","message":"Check timestamp of signed EC2 token request\n\nEC2 token requests contain a signature that signs the entire request,\nincluding the access timestamp. While the signature is checked, the\ntimestamp is not, and so these signed requests remain valid\nindefinitely, leaving the token API vulnerable to replay attacks. This\nchange introduces a configurable TTL for signed token requests and\nensures that the timestamp is actually validated against it.\n\nThe check will work for either an AWS Signature v1/v2 \u0027Timestamp\u0027\nparameter[1] or the AWS Signature v4 \u0027X-Aws-Date\u0027 header or\nparameter[2].\n\nAlthough this technically adds a new feature and the default value of\nthe feature changes behavior, this change is required to protect\ncredential holders and therefore must be backported to all supported\nbranches.\n\n[1] https://docs.aws.amazon.com/general/latest/gr/signature-version-2.html\n[2] https://docs.aws.amazon.com/general/latest/gr/sigv4-date-handling.html\n\nConflicts due to six removal in e2d83ae9:\n\tkeystone/api/_shared/EC2_S3_Resource.py\n\tkeystone/tests/unit/test_contrib_ec2_core.py\n\nChange-Id: Idb10267338b4204b435df233c636046a1ce5711f\nCloses-bug: #1872737\n(cherry picked from commit ab89ea749013e7f2c46260f68504f5687763e019)\n(cherry picked from commit 8d5becbe4b463f6a5a24a1929dd0f48dab6ae027)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3f65d6fbcd18032a8ad3dfa3aaded264a282158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3f65d6fbcd18032a8ad3dfa3aaded264a282158"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
