)]}'
{"id":"openstack%2Fkeystone~725176","triplet_id":"openstack%2Fkeystone~master~I52ebbcae0d77c0420848a37babc73bb75dd58e0b","project":"openstack/keystone","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I52ebbcae0d77c0420848a37babc73bb75dd58e0b","subject":"Switch to new grenade job name","status":"MERGED","created":"2020-05-04 06:15:42.000000000","updated":"2020-05-06 09:27:13.000000000","submitted":"2020-05-06 09:01:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725176-1588755689031-e143bcec","meta_rev_id":"b2779f8bb407be012ca3b77b187325e697cd2f5b","_number":725176,"virtual_id_number":725176,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-05-06 05:58:08.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 09:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2020-05-05 16:34:03.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-04 12:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-05-04 11:11:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2020-05-04 13:15:21.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":1,"date":"2020-05-04 14:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2020-05-05 16:34:03.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 11:11:55.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-04 12:21:53.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-05-04 13:15:21.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-05-04 14:56:16.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-05 16:34:03.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-05-06 09:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e082939d4b95d02fb18cb9e9dc1e6328061d70ad","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28828d74b775a7abcf299b22c3351efcd244e26f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:20:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"75e2fce176c2bf3b3476bb814ca3beb06fdc91a6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:38:12.000000000","message":"Topic changed from grenade-py3 to native-zuulv3-migration","accounts_in_message":[],"_revision_number":2},{"id":"fc22dacf43739850269497b65416e9727fd2ced2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:51:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff96b3befe44452e88c1e6f209c36214 : SUCCESS in 15m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ef175170ea449a7baaff7b024b66bfe : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85c9c43fdc54d30b86690efe63d4f71 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59814c0f897048e6a536121b567e9d8f : SUCCESS in 12m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3933b1bf63c4cf18638c91e46c2c9cb : SUCCESS in 14m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4231691cd65e48b8929ce178b0f9f069 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8874d90b5b46420ea6225dee89bae205 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/7332db304f204a4d8231bbe2c8b0cd55 : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b2fd60e3fce40ed9cb4d86bbed90e72 : SUCCESS in 1h 42m 14s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c756ce0031cc4f52bc380f2c223356d4 : SUCCESS in 37m 31s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/58fd4cc5c4a64ad1bcc4db1117e550c7 : POST_FAILURE in 32m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/5f2798233f364a978e4b17ed38e14df2 : POST_FAILURE in 34m 23s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2361ccb94bbb4e3487c72aafba3e405f : SUCCESS in 17m 51s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/486625e1fffe41c293c0ed5d62361ca6 : SUCCESS in 41m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c8290e41ccb46f3b09eade539a9fb5d : SUCCESS in 1h 03m 40s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/75e9bbd0d59847bb83c96960410a452b : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a1d89969dc3f14dc0c74f1a10f6b1020da7722b5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 11:11:55.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6114193fb59136bf7b4082b10083820f449556e1","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-04 12:21:53.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Andreas for these updates.","accounts_in_message":[],"_revision_number":2},{"id":"a69544121c3ff37abd93d5ef875ab75f80b8b793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 12:41:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef4f402a514f4e698a2c9a157a058cb0 : SUCCESS in 22m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c4851c83939455e94940e2303c643df : SUCCESS in 23m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817a756d1bfa49de87166a60f5f7b731 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1195e6844a3b4025a385ddf24a3ffafa : SUCCESS in 20m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59c211288e954feba2bb60505fe2289e : SUCCESS in 21m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc37873d502e44baa0d1575b37569d99 : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98960711327f4779b9ce900e6397af76 : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/727356583ce8444ca531f70986143a4c : SUCCESS in 1h 02m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b79e57ec996d406482c3a6b3240fe001 : SUCCESS in 1h 27m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4889780f021d40948f330b78a2e0d2da : SUCCESS in 30m 52s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ec9ad80302be4a288bbf98dbb3547b4c : SUCCESS in 31m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/4cbcdf3343ad43bf9e3a6d76b529f057 : SUCCESS in 33m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/cd9d369dd8334cf89dbbbf750ee21e54 : SUCCESS in 13m 23s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/976758ba1dc64f4eb493619b8eacefe9 : SUCCESS in 45m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eca0e1cfd60b43789158693ff8048cae : SUCCESS in 59m 42s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/99c1e4308b604788bb4056008879b602 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0acde2f8020484e2c32ec0be648d3e19e0970286","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-05-04 13:15:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a927415bde4f68eb200bf0a8f4656a6f4473ca37","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 14:40:43.000000000","message":"Patch Set 2:\n\nI remembered I did change in keystone  - https://review.opendev.org/#/c/725149/","accounts_in_message":[],"_revision_number":2},{"id":"15b5232a9823b520980ad3c61eaede99b2496a6d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 14:56:16.000000000","message":"Patch Set 2: Code-Review+1\n\nabandon mine, lgtm.","accounts_in_message":[],"_revision_number":2},{"id":"7456392c9bc06cac5c916fe322b1e884bb125482","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-05 16:34:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d76ce3dbc2597127d0a255f5836eb675522c8cf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 16:34:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"038fb87ebe789c7511b23197da4413bbf71a4e60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 19:01:05.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3264f9b95e2c4b1d9b991c734d9f524b : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0861ed1098b346d591b45e3451c16ef6 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f81ab9b1c48a45a697f577b1af8beebd : SUCCESS in 19m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/169e6d3721ca4b60ba8463b4136e7b09 : SUCCESS in 20m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa79c6f24594d3fba81b569f032acdc : SUCCESS in 16m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/abd3ed94445f4ef6ad6e45cdc8bb8589 : SUCCESS in 1h 16m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf7ff99401384b3ebe9a1eab475338c3 : SUCCESS in 1h 27m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e770e880ee7042559fd97085d89dc13f : SUCCESS in 39m 17s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/cf878da8b0c84a4a87f62754bd021ff6 : FAILURE in 40m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1640997fcebc47389bb235e77fd58280 : SUCCESS in 59m 31s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f5fb911d3f724b9bbac74a809b01fae3 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c84a1352689a9e9f29708fd62486c4eb81218708","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-05 19:02:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bd7e8c9817a1c36e04928e5193384045fc714596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 20:30:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57fac656362945f2a3c419bad904c8a3 : SUCCESS in 17m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d32616da289c470b91064541969e6e15 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d7e009ec7c4174b3fb25d29008e76e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/458675a9c8bc41169198fe54fde35f36 : SUCCESS in 17m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b15a56f5cda492886fdc625c150334c : SUCCESS in 15m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0a85f123fe84119afd56520ec3bf21e : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0496b2821f3475bb84b756541df07f7 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/aa892b2115534a5bb42fb77b65818700 : SUCCESS in 1h 01m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73c83fe6ad07451c9fdb58b23b02faed : SUCCESS in 1h 26m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3e5343dacd2f42f188c80b45a747176c : SUCCESS in 31m 34s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d24e236058764b3fb36c0f7b19f2c3cf : SUCCESS in 33m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d4fb79d4698144b791f9038b2f2fd17a : SUCCESS in 40m 00s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e8e5cac950ca463fa4119536f9cf606e : SUCCESS in 14m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f7c7e9ef222f4a479e56b31fbdcf2f1d : SUCCESS in 39m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7241984b8e9f4ecf836fb73a856060cb : SUCCESS in 1h 01m 52s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/eb6d210d67794dcb9a6753798715acfc : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":2},{"id":"26468bbd15953ef3c6cf2957053ac32d4ae77646","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 20:31:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c531255fe98605a9ee7a62c526b038fcc65489a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 00:58:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca1acde1412844c98f5f5ced5d57c523 : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d269e32a424709b12e984dbce8146b : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5107568ae27f4fcfbce40d8e2a3f1a04 : SUCCESS in 15m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f008bdd182a945229fa3c8da3f1a33fd : SUCCESS in 20m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53812ee15f494014a79a1ffab9a0802b : SUCCESS in 16m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/6eafcc2c0f5f4aff98f4dc3a5f7ad880 : SUCCESS in 54m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7bd25fcd7464009b4d9c50c25fa3458 : SUCCESS in 1h 24m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8d4adf0b74714efd850784cdac14f2f7 : SUCCESS in 31m 35s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/fbf997eae20f4bd5b8bc90c12151b51e : POST_FAILURE in 37m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72228a7330ad4332a400076b9ae9b1a9 : SUCCESS in 1h 04m 13s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/95fb350ecb744ec19cfa2f1c949353c8 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":2},{"id":"66b0139ad4684edf70b91fac770f424e73895044","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 05:58:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cec146244b0c32097b3f81ae5da3ef996da3cfb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 07:34:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31c6b6fec04d448e8b79b561f92a9162 : SUCCESS in 18m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc884112e7a94aca94cb564ddb00158e : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0685fe5ad74aadbe1dd77ea174f263 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f30b9c18e5949ee9ba914f1a00ff5d8 : SUCCESS in 16m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231ab92fe610432eb66f6b9a5586c48d : SUCCESS in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33dd1f0c53654c68969cd78def49ad90 : SUCCESS in 16m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/75efb254a73746c2a704448eafe665c7 : SUCCESS in 1h 10m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d070f88d9dc842e2bdc60256adf643d9 : SUCCESS in 1h 30m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/451510d7ddaa45d08e78c9ead39b031a : SUCCESS in 39m 14s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/187c812bda9f4ae39fab33f5a0160a23 : FAILURE in 40m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/1a30701425044980921ecae6b7c4f93d : SUCCESS in 39m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/cc394341730f4125a97de51d0130d238 : SUCCESS in 20m 14s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/69475da57de34574859181566f0154eb : SUCCESS in 41m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25655a96b85046a49d72101bf2931b70 : SUCCESS in 1h 01m 07s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ea237dc852d14871b2ed6b168fbfc25b : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9671ed49c9b1d6c86c127da6a38ab0fd724e49d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 07:34:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2cb59dff8c99188392334ab97a4b4fdb7bfa4b5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:01:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc2ad06f499f404ab044113036fa2b39 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70aed4756054569b9278187ea2d87b7 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c57be8eb53843119b1940b33f73f663 : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2fbd6179b5c4cf68884aa4c1ceabb0c : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a43df0a38d84b248eb40f217650c887 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/7dd1d48122fb450c9bd63dccba486bec : SUCCESS in 58m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79f5d38170b84779ae145490056dc1fe : SUCCESS in 1h 26m 14s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5442672849c845909a4c18ba82221123 : SUCCESS in 32m 47s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e52bb90db98a4035b648e490ec987579 : SUCCESS in 36m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/779033bb162a43d498d40463825314be : SUCCESS in 1h 00m 33s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/db175c2c6dc1458c80cbdf52703e28ee : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b7539da04ebf71550a355202b5bdda65cf8ff0e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:01:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d6dc34e5a17a56ce38b2ca32df7fab7a0a4a744c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:06:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d33ca09006343a68dae7ff1512c7131 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b2779f8bb407be012ca3b77b187325e697cd2f5b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 09:27:13.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 1f0603598f8f648a8920c419fc56c0bbee2ce7fe","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b4e53792d75bc02a47b0caf2b53d3cb55b53737","revisions":{"1014773de5197918f364a90d3fc29adb502032d0":{"kind":"REWORK","_number":1,"created":"2020-05-04 06:15:42.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/76/725176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/725176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/725176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/725176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/725176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/725176/1"}}},"commit":{"parents":[{"commit":"963cc3a692ad545e640db2b82a7eff08cabceba1","subject":"Merge \"Update master for stable/ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/963cc3a692ad545e640db2b82a7eff08cabceba1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:14:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:14:22.000000000","tz":120},"subject":"Switch to new grenade job name","message":"Switch to new grenade job name\n\nThe integrated gate has been switched to the new grenade name,\nuse that here as well to avoid duplicate grenade runs.\n\nDetails:\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014602.html\n\nDepends-On: https://review.opendev.org/725148\nChange-Id: I52ebbcae0d77c0420848a37babc73bb75dd58e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1014773de5197918f364a90d3fc29adb502032d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1014773de5197918f364a90d3fc29adb502032d0"}]},"branch":"refs/heads/master"},"2b4e53792d75bc02a47b0caf2b53d3cb55b53737":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-04 06:20:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/76/725176/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/725176/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/725176/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/725176/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/725176/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/725176/2"}}},"commit":{"parents":[{"commit":"963cc3a692ad545e640db2b82a7eff08cabceba1","subject":"Merge \"Update master for stable/ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/963cc3a692ad545e640db2b82a7eff08cabceba1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:14:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:20:44.000000000","tz":120},"subject":"Switch to new grenade job name","message":"Switch to new grenade job name\n\nThe integrated gate template (integrated-gate-py3) has been switched\nto the new grenade name (grenade-py3 -\u003e grenade). This repo uses the\ntemplate but also has for irrelevant files an extra entry.\n\nRename the job following the template change to avoid duplicate\ngrenade runs.\n\nDetails:\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014602.html\n\nDepends-On: https://review.opendev.org/725148\nChange-Id: I52ebbcae0d77c0420848a37babc73bb75dd58e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b4e53792d75bc02a47b0caf2b53d3cb55b53737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b4e53792d75bc02a47b0caf2b53d3cb55b53737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
