)]}'
{"id":"openstack%2Fkeystone~725806","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I52ebbcae0d77c0420848a37babc73bb75dd58e0b","project":"openstack/keystone","branch":"stable/ussuri","topic":"native-zuulv3-migration-stable/ussuri","hashtags":[],"change_id":"I52ebbcae0d77c0420848a37babc73bb75dd58e0b","subject":"Switch to new grenade job name","status":"MERGED","created":"2020-05-06 09:27:13.000000000","updated":"2020-06-02 16:44:46.000000000","submitted":"2020-06-02 16:40:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725806-1591116045156-6506d538","meta_rev_id":"af23d110e0e7b8aff869044db574c249eabbe1f2","_number":725806,"virtual_id_number":725806,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-05-07 12:29:17.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 16:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2020-05-29 18:12:43.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-06-02 09:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-05-07 14:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2020-06-02 14:56:27.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:10:27.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-06-02 09:31:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-02 14:56:27.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-06-02 16:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e5763856833a813c76d8e8652632a3f93809a66","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 09:27:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"256c2958ed1f409a52f4f0afc1300449cc9d17d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:06:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9d7a9674944e009b64bf54d8f804ad : SUCCESS in 20m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7ebd9a9dc734e8e87aa3e93518a73ae : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe2780ad7ea4e938e9dd28f63a965d7 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d25a1257c0f4f9f82b2fca08fb15267 : SUCCESS in 15m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e5434e94ecb4a2caa10c3e02ec3c1af : SUCCESS in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b0676995e84f179fe9b663ae1d626e : SUCCESS in 16m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e6fa53c1933a4c8995a5a83eb47bc4d6 : SUCCESS in 1h 21m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be0237dbc1d24ee3b2be303d352add02 : SUCCESS in 1h 32m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/abf06b1e8f8e43098b7f8ed8dde613f2 : SUCCESS in 48m 27s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/04d1ac4815974f8e96cf7e740f2c3d5c : SUCCESS in 39m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/340a8c185d9842bf9044c6b2f2a6cae1 : FAILURE in 40m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5e426980fd9d4ed3b293dffc0fc30bb4 : FAILURE in 19m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0aed575d551e46999d3690eaaa92a2bb : SUCCESS in 56m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d35e86ab308f4b3b9128a300f030e90e : SUCCESS in 1h 04m 56s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/54b81c9fd49e4eb4bb32928cd3757fed : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1646d2edef3e4f8b3562dd380096d6086e2076ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-07 12:29:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"275445c7556f3207bc22213d58442d65b862e00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:09:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73f44e6709ab401990a15761ca5ca8dd : SUCCESS in 20m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cade9cf920204ab18f8cb44a02d05c5f : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70e22f18c954cf9b8feb5f52a188b15 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf1d9d05462410489fb73ce0281911e : SUCCESS in 18m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e1a113cc55549c4b75df1fac17cef33 : SUCCESS in 17m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f51b1453d764bf3869800358fe59200 : SUCCESS in 17m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/19923de18dbe4cd2a46de64df84c0f5c : SUCCESS in 1h 27m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99558e1553e747c58b4e58af4eb4a950 : SUCCESS in 1h 29m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/973f5056299d458a9d6900d1652b56f7 : SUCCESS in 38m 23s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/819cd2b783794bad86df00d3acb6c73b : SUCCESS in 38m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/ff8306d6dfb4426cbc552f0f70de537f : SUCCESS in 41m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/63b6c325d8a04f71b011a08f4f64dfff : FAILURE in 21m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c8b3f055a3e64da89984708e41f09cf4 : SUCCESS in 41m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a222752f2db40e8824857f0804c0327 : SUCCESS in 1h 03m 57s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f5b85d8bcb1b4b23afad9dff7108460e : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":1},{"id":"52fa1b982c777df9cfd1ea7e626f5b7d9a6b6de2","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-05-07 14:10:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2fc69afa3126145d8716018c90d1a2590f40d1a0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-29 18:12:43.000000000","message":"Patch Set 1: Code-Review+2\n\nPatch is merged in master","accounts_in_message":[],"_revision_number":1},{"id":"5baeb575a170780be6393faf61b7be4697d55092","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 09:31:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5385ae0a9dd910ab0bd397024c0d0f0f3059b237","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-06-02 14:56:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"903dbce6a493493da3f63f63eec98e8940307021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 14:56:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bdce429061e01390284139a47fed8e16795df6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:40:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cdc2a8f2d114f8b9eeecedb63fd4b6a : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573d68383f044df8bfe23afc251e5c47 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44e23ecd65754965ba2fc3d32e6a9624 : SUCCESS in 15m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c18f3a7e6e94bb496700e2d742e6173 : SUCCESS in 16m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76099096c0a4740a54238168c5f0b85 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c18c137767b74679936145387b4aa5ed : SUCCESS in 1h 15m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d5417f947e5436fb40c2d35f6618f89 : SUCCESS in 1h 37m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5c1b8c738d454a3d9cb65776039aaadb : SUCCESS in 49m 19s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/cb20c78575ac4fddb8bb5d84a9087243 : SUCCESS in 50m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5141f86528d4a689a403a2d7b6a1563 : SUCCESS in 1h 12m 32s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f530bb0488394db59a15a93e35d9a863 : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"38e0ad5f79adb006a1ca74fec172c340fa2837a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:40:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d0793e32ca98a5ca9a9b79f26c7e47da981f184d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:44:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68788bfd0cfa460fb1de09ddde7ebee0 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f0603598f8f648a8920c419fc56c0bbee2ce7fe","revisions":{"1f0603598f8f648a8920c419fc56c0bbee2ce7fe":{"kind":"REWORK","_number":1,"created":"2020-05-06 09:27:13.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/06/725806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/06/725806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/06/725806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/06/725806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/06/725806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/06/725806/1"}}},"commit":{"parents":[{"commit":"0e463a78d468e78edaeb7cc5f41d70e2a0c77a50","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0e463a78d468e78edaeb7cc5f41d70e2a0c77a50"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:14:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-05-06 09:27:13.000000000","tz":0},"subject":"Switch to new grenade job name","message":"Switch to new grenade job name\n\nThe integrated gate template (integrated-gate-py3) has been switched\nto the new grenade name (grenade-py3 -\u003e grenade). This repo uses the\ntemplate but also has for irrelevant files an extra entry.\n\nRename the job following the template change to avoid duplicate\ngrenade runs.\n\nDetails:\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014602.html\n\nDepends-On: https://review.opendev.org/725148\nChange-Id: I52ebbcae0d77c0420848a37babc73bb75dd58e0b\n(cherry picked from commit 2b4e53792d75bc02a47b0caf2b53d3cb55b53737)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f0603598f8f648a8920c419fc56c0bbee2ce7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f0603598f8f648a8920c419fc56c0bbee2ce7fe"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
