)]}'
{"id":"openstack%2Fkeystone~725885","triplet_id":"openstack%2Fkeystone~master~I02f9836fbd4d7e629653977fc341476cfd89859e","project":"openstack/keystone","branch":"master","topic":"SEC-bug-1873290-1872733-1872735-1872755","hashtags":[],"change_id":"I02f9836fbd4d7e629653977fc341476cfd89859e","subject":"Ensure OAuth1 authorized roles are respected","status":"MERGED","created":"2020-05-06 15:15:47.000000000","updated":"2020-05-07 04:39:49.000000000","submitted":"2020-05-07 04:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725885-1588826271431-d95910ec","meta_rev_id":"dd05b4e8d799cb0e8083595a5906b582fb6cb1b9","_number":725885,"virtual_id_number":725885,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"date":"2020-05-07 00:46:46.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 04:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2020-05-06 20:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":2,"date":"2020-05-06 15:31:35.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":1,"date":"2020-05-06 15:40:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-05-06 17:06:26.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-05-06 15:20:41.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2020-05-06 15:20:41.000000000","_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":1,"date":"2020-05-07 02:15:33.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-06 15:20:41.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 15:20:41.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-05-06 15:20:41.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2020-05-06 15:20:41.000000000","updated_by":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"reviewer":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"state":"REVIEWER"},{"updated":"2020-05-06 15:40:40.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-05-06 17:06:26.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2020-05-06 20:48:37.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-05-07 02:15:33.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-05-07 04:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d384bdd0d5929339a287833a378de780104528d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 15:15:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2092614fce859b6a571701119549558a134d45c3","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-05-06 15:31:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83f3b4b2e9311d8ff0f7797def123c095244eed4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-06 15:40:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf212aeca529c03e92f7906e56cb283b7a4ff284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 16:33:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2837171e7c4a41a4b92fbe915ff55e : SUCCESS in 24m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/671fc584591d4727be364a8ffae5cd77 : SUCCESS in 23m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e455fa328244d393fb8ed29aab5cb5 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7cc92e0a01640b293138fb31279fbbb : TIMED_OUT in 41m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9d8e1335f8741198dcc18d160640b54 : TIMED_OUT in 42m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8fafe6b71994baa9814f33feec80b29 : SUCCESS in 22m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/760583af031d4e4dbd025960b51bbe44 : SUCCESS in 1h 02m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdb3ffe41e5f4fb597a3f802c03c5eda : SUCCESS in 1h 15m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a366dc775ce4305b81322c9b5b7489b : SUCCESS in 18m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/efff25b1a9c94f379ea00c518be0f631 : SUCCESS in 30m 46s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9931db7e3a0e495f820a6c3c5c06e157 : SUCCESS in 31m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d640461f9ca3406bbc9a13d27e33e550 : FAILURE in 41m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2377c6464c7842169b6e9736aca61b80 : SUCCESS in 16m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/67fa674ccb584d85b4f8443457bbe8f0 : SUCCESS in 46m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8428c1a28af944c586d43511c3aca974 : SUCCESS in 1h 12m 57s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/bdcc1b3f2577430a9907432ac3278f32 : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9fe127b6e84ea498b9e9098416ccba1da63f96c9","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-06 16:34:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"deee881a1928da3b40db04076f1fc8035cdcbfd5","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-05-06 16:43:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a7313d173f7f315fb1618a549c03a065013bbe9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 17:00:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eccabc62368676c1a832ccbc09b92b3e64f7a036","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-06 17:06:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"adb322202d0924fc9ba383ffa90d40155be3fae6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 17:33:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68a30b2a2fb3afe890e8fecda4bb7de2446129ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:01:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b1383fc246a4273a0f2b0651df4c64e : SUCCESS in 16m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/669506b5732043f8b6ce7186bae9a43e : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995acb4ca1254f58aa95194862118435 : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7539b3f83aa440a7bec3a231cd0957eb : SUCCESS in 28m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515e918e7cdf4f13a4e505ee7c96c6fd : SUCCESS in 29m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c4ca2829bb48168e63df98f37a8b15 : SUCCESS in 15m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/fcbe93c7b75c462eb6fb11a82e9affdf : SUCCESS in 1h 03m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89a35f835a8c402e90d0c0d3a02e3570 : SUCCESS in 1h 17m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e032047334fa468d913d6f2caf654761 : SUCCESS in 21m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a9144d46bbdd4b54a5f25b6f76e6838a : SUCCESS in 34m 10s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9bc1f8ed859a43129e277ef378b54ae5 : SUCCESS in 36m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/72403db927ae44b0b52692c6ad20af5f : FAILURE in 39m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/589ecec400384092aafc34a688a6818d : SUCCESS in 22m 58s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d5a0adc0646541e8bd1a25322263804c : SUCCESS in 41m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/816d1f2a5e914e159f9af0999437bf57 : SUCCESS in 1h 03m 20s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/6ee7f568d1414f96863191f5d43d567c : SUCCESS in 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"10a2127e0b66334cca2cd583c7e8273de33affc0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 18:46:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c390032d60efc15c18a8a280ee17f32e995392f7","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-05-06 20:48:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd1424ee287d9f354d18c756be872228b9c5678e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:54:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd62076bac494037bf48fcd5d56dc5f3 : TIMED_OUT in 51m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbc392bea45a465a9650783d9fa48425 : SUCCESS in 27m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75d056c74df414a9024ab64c152879d : SUCCESS in 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60523df956704b0bbb9a974ee948ef0d : SUCCESS in 15m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d57d8f3eb4fd407dbc21e90a8c315c4f : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7df7d6047e45d1997c738113cd9fd3 : SUCCESS in 16m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/b19b9003127744398c3a5b97113ae2cc : SUCCESS in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/874d000f8954424fac34a38a2ade2a3d : SUCCESS in 1h 14m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/601cb920024949089259140b54d1a4f2 : SUCCESS in 20m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0dd1354d608d44e08f9bf4805f1aafe5 : SUCCESS in 37m 26s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/660b2b3419f74892ae92eca87b29b025 : SUCCESS in 41m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/83730793dd004f1ba414fc4fb41f0c63 : SUCCESS in 41m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6fbbc3e651e34718a11fb17733f4687b : SUCCESS in 20m 40s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9b66fc40725545ae82853e15d53c87de : SUCCESS in 51m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdf58d47478344a387dd0a5eeb91e28f : SUCCESS in 1h 11m 36s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/29d14a0da2e24268955d10b7ebef319c : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d08e525a8d748e44d2acdf16dff442f0df6b224a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 21:19:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd97f6964a7c457d4520dfbd8db88af7ddf221eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:36:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3c7ecd88c9544dbb28c9aa9c88eb13b : SUCCESS in 21m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7edeab1c27374ecf84d6ae45fc6b4b22 : SUCCESS in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1176c2d9b540aea3793cbeadf88e60 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f159613ab434159b754a45150294177 : TIMED_OUT in 41m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/307c98b2006c465f958a2bc6c50513c8 : TIMED_OUT in 41m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f6a33c4aa34568b1beab4b68ddc0ce : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b16d377befd940c79eced6720ddd4a9a : SUCCESS in 1h 05m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0315a86be8d94271a0f88e43ecd5d0ff : SUCCESS in 1h 15m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c375be92fbd64a8bbac42fef2195237e : SUCCESS in 14m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/14b4a6d7feeb47c4ab0e8480b525ab8b : SUCCESS in 29m 30s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2536d4a3edc941189808d270d5715ee2 : SUCCESS in 35m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/95dcf01ae74847eb92fc2069601e8e57 : SUCCESS in 38m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/10e28b9a0aca4c9e98efb6c6d6748b4c : SUCCESS in 14m 58s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/010e53a82c8340e18309c0ab85050fcf : SUCCESS in 38m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfae4030331a4056b99aa71daf6dd9bb : SUCCESS in 1h 04m 36s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/c28e885fd7ea4399855597fd1a6e8bd6 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4724fc73f60c29af2b0960569bc6938d34d9ff94","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 22:45:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"31a12d243d3cab5f0b6cdf27a7391b246dd6af1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 00:43:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2135f66ce84f459a80ed7e20234e7f23 : SUCCESS in 24m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d7ee0ca43ba4d8bbe8e09b4c6f64095 : TIMED_OUT in 41m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4120b7262d44e6481ac895900667f47 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95aaaad1c29549d5ab549771ace457cb : TIMED_OUT in 41m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41fa2e8cea2743f88646b9d003038a2b : TIMED_OUT in 41m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d32d5b20504a27b25a9ccdde850d7f : SUCCESS in 27m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/23e03f18a6c8412ab16b6c6b7dd54195 : SUCCESS in 1h 32m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d065e872517401b897f0ca3a19d34a2 : SUCCESS in 1h 55m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd9a8dbeca7d45a8a79480babde2cf87 : SUCCESS in 23m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4d2be358c2bf47f1a82f59c4a1110c6c : SUCCESS in 1h 01m 31s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/95ced713c8b8414bbab62a39faa209b5 : FAILURE in 57m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e1f2bdda0024423bb7fcb147350c203f : SUCCESS in 34m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8ac5f68e398a4b54b4fadfa577f4e06a : SUCCESS in 30m 16s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/714546d3d73942a0839179fe0b16ac15 : SUCCESS in 1h 09m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82e84364f4b04bd8937c629b9a662e1d : SUCCESS in 1h 30m 59s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d473135cf42d4b6a87e79a022fea65b6 : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5cb1241696348d9168693461b0e7be770b6a2d6c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-07 00:46:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d47e36abd0faafcf371c8a139f2156d1338b0fee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 02:13:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d6765dda6bd4fcba0686a1f30eef1c5 : SUCCESS in 23m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a0a3962a0ac4efa8ec97bd9e332d45f : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88065a5588534d008fba53d6042819ea : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3abb2e28c0d5431db99b872ccf976be4 : SUCCESS in 13m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/336eb09753de4e5b911ec9a60486a6aa : SUCCESS in 16m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af19c0b1177f43ac8af0614a3fda9051 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/cd83c3e8aea247eb8491992dd9333883 : SUCCESS in 58m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74c38a76cdba483e8b37c101c05294e8 : SUCCESS in 1h 21m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f30ba676c6a1438381714d43d34bb71b : SUCCESS in 23m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/be30fbff6eaa4309b9f52e6ccc7d88d3 : SUCCESS in 30m 53s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/bfb69a4b13e440e898b674e24aaab9c4 : SUCCESS in 42m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/a457a64dd3864abe8263dde0991cf41e : SUCCESS in 40m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a0e1db5086604cbf98062ad57541ef65 : SUCCESS in 20m 46s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/81f7fff573314f939f8504b37f14f6bc : SUCCESS in 45m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5520f58d8b744aa8b1e869be608d0365 : SUCCESS in 1h 08m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/4a00113070794cfa8003d07ac93102ca : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5136c468b41024cd724e84f0377ce2dd03d6cc8d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-05-07 02:15:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aca53d768287fb152cb08328aecee407662453bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 02:15:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abd8ecb466c6e74e8ff0e8c8c812e722845803b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 04:37:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48824f89007e4aa3b3d0e2bba381a873 : SUCCESS in 25m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c95797de1824dc88efef17fb0e69420 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3feebaaed71844beba7771e1a9c9b35e : SUCCESS in 24m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dfc974be39048b8a7b7ed48c3f68006 : SUCCESS in 21m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e046666f580435791e37acb99c0cfc6 : SUCCESS in 18m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/4814bfad19ba4e3bb19c58bc9811a827 : SUCCESS in 1h 05m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f50e726a19384122a88ed67e20f9f2df : SUCCESS in 1h 45m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6908877118d242e9944866b061d0124b : SUCCESS in 17m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9279287f5fab44ac92297e66d55b26a7 : SUCCESS in 35m 53s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/b96a1c4ee1b04171887a262615eead45 : SUCCESS in 34m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/837971eb2e7e424b9d2796c811186890 : SUCCESS in 1h 02m 41s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d5e16a0760d34ccf95e55e62060b5aa4 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a3e4aa9ce36e247be5bf05b5102acc32ef2af076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 04:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51f5e2467fa8f6a81bc6a928ddd56efeb7f747e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 04:39:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0105b619bca4294855767de8ee8edca : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a411ef6da0404d46afcc4fbc2eaac279 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c73690f779a42a5c62914b6bc37f0ac2f41a3e3","revisions":{"6c73690f779a42a5c62914b6bc37f0ac2f41a3e3":{"kind":"REWORK","_number":1,"created":"2020-05-06 15:15:47.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/85/725885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/725885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/725885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/725885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/725885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/725885/1"}}},"commit":{"parents":[{"commit":"3ec99407109fdff8c10b1bb6609c01a6fc5a6ad1","subject":"Merge \"Correct help for unified_limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3ec99407109fdff8c10b1bb6609c01a6fc5a6ad1"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-04-17 03:35:46.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-05-01 22:48:58.000000000","tz":-420},"subject":"Ensure OAuth1 authorized roles are respected","message":"Ensure OAuth1 authorized roles are respected\n\nWithout this patch, when an OAuth1 request token is authorized with a\nlimited set of roles, the roles for the access token are ignored when\nthe user uses it to request a keystone token. This means that user of an\naccess token can use it to escallate their role assignments beyond what\nwas authorized by the creator. This patch fixes the issue by ensuring\nthe token model accounts for an OAuth1-scoped token and correctly\npopulating the roles for it.\n\nChange-Id: I02f9836fbd4d7e629653977fc341476cfd89859e\nCloses-bug: #1873290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c73690f779a42a5c62914b6bc37f0ac2f41a3e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c73690f779a42a5c62914b6bc37f0ac2f41a3e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
