)]}'
{"id":"openstack%2Fkeystone~725887","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I02f9836fbd4d7e629653977fc341476cfd89859e","project":"openstack/keystone","branch":"stable/ussuri","topic":"SEC-bug-1873290-1872733-1872735-1872755-stable/ussuri","hashtags":[],"change_id":"I02f9836fbd4d7e629653977fc341476cfd89859e","subject":"Ensure OAuth1 authorized roles are respected","status":"MERGED","created":"2020-05-06 15:16:17.000000000","updated":"2020-05-07 19:54:50.000000000","submitted":"2020-05-07 19:52:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725887-1588881164945-c134243d","meta_rev_id":"5cb0ade8498982010f081db758ffb91550b8a2fe","_number":725887,"virtual_id_number":725887,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"date":"2020-05-06 19:11:52.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 19:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-07 12:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-06 20:50:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2020-05-07 13:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":1,"date":"2020-05-06 16:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":1,"date":"2020-05-06 18:15:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":1,"date":"2020-05-07 10:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-05-07 13:55:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2020-05-06 15:21:17.000000000","_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-06 15:21:17.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 15:21:17.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2020-05-06 15:21:17.000000000","updated_by":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"reviewer":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"state":"REVIEWER"},{"updated":"2020-05-06 16:28:51.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-05-06 18:15:28.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2020-05-06 20:50:38.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-05-07 10:26:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2020-05-07 12:06:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 13:55:58.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-05-07 19:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9df48216446080400404be73fc9261c2d8db0848","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 15:16:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b63bdcd1815270d94efc2ed51e32bf60422aee0e","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-05-06 16:28:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"45015e512c1c21f23d6865e96673919dbadbfb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 17:09:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3395b9049654286abe81657dd111e28 : SUCCESS in 21m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a39aa0319f5f4cd89687a1cac3ac2090 : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3abc5bcdde400183e300a3f35b5e02 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8614cc25803149baac81eec1b851bf9d : SUCCESS in 25m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0f1e20117214c0287bc055924dc3c71 : SUCCESS in 17m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a67403cd0494e36a983d935029cfc0e : SUCCESS in 16m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/fc46d45a885d498c9fc1327e13175440 : SUCCESS in 1h 48m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7fd62fa19c2452182d30491f68fcff2 : SUCCESS in 1h 19m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53484a6c4e5d47729d455c778ca96a12 : SUCCESS in 17m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a08ee960179e44fabe439bc2c66cb5a7 : SUCCESS in 41m 11s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/55f6ce707bcc4af5b3573ca5ee28ba14 : FAILURE in 36m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/9e15e9961313418589166427b6fa9ef9 : SUCCESS in 34m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/47906c3705d645208e76525c684a8d92 : FAILURE in 16m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1c448fa91daa43afaeca86b2ac2eada0 : SUCCESS in 41m 54s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8de53c9f45a3420fad576502f7f78cbb : SUCCESS in 1h 40m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5a0f1c4d6184e0ab8c11d8814948da0 : POST_FAILURE in 1h 09m 11s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/5068ce3b57c94dae94a13cd9821913b5 : SUCCESS in 46m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b9c9db091ab33882b2b566a66d874a79e9846712","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 17:22:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8000c77fc4c5300e956474e2f066ef7eb247e2a","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-06 18:15:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"656f8e02c6465263643a43ae87d1fbf976efc06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 19:10:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37d92f553aee4c779b95830b2cff7638 : SUCCESS in 16m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/831ba2548853479f9a6a4d5c685d2f11 : TIMED_OUT in 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5570e70a1b4e3eacbcc81dd742ecfe : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1457224d9a94cdd91381571f9025b27 : SUCCESS in 15m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/558fdc464e114c4daecf283635ff6abc : TIMED_OUT in 41m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b466b50a3f08434bba15f51fcd534bf1 : SUCCESS in 16m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/f98ffdc202f84dc78a047b8936257886 : SUCCESS in 1h 08m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1172a3bc8422439194a2a4bd51df480e : SUCCESS in 1h 25m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b51a13fa58de4aa9bf80117d35ec6f06 : SUCCESS in 20m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/04737ea0014c41939cc07b8d9d969461 : SUCCESS in 42m 27s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f1e3ad17d0a0488dbe9af6b28e54b46e : SUCCESS in 39m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/026025f0b6184a2aac7f7fcdb6b78d1d : SUCCESS in 34m 05s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0bef3745793a4843a857ebda4bc4b7f6 : FAILURE in 15m 55s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4f4ac0e411384e1b903f0dbdcc7d8db9 : SUCCESS in 42m 51s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a86e9c178bd74cb38dd867c68b31ad5d : SUCCESS in 1h 41m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66ba1e0e39904f4fa528b10cd0d71181 : SUCCESS in 1h 03m 36s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/9447a6f610b34b5498c50c3f13513893 : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5fb6f4f70102e5d3a35280ad28e08bee9e8721cc","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-06 19:11:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"114539cb416552a7b48e0487a6197590d910b79f","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-05-06 20:50:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"81d0ae641f6b4f3ee9aaa624af067076dcdabf09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 21:04:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0636ac0495c64897bdc7039388897e30 : SUCCESS in 46m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6513665d0213453aa59755493e0d6bcf : SUCCESS in 29m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e698c4bd43ea482286a53ffc04d86eec : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6dfef0543da46628f7f6e189c27d12d : SUCCESS in 21m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f67bf289c5e5467593053dfbc477916f : SUCCESS in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37055afd3ab4cb9a7bd839c163406ed : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/f5e422d2a8ed426cb63fef06df560d70 : SUCCESS in 1h 19m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be0825f57b78476eaf3fd1ce11289182 : SUCCESS in 1h 27m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e8127fe7a8a4e28873ae9e125923b32 : SUCCESS in 20m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8f7562df5ae0457a8f27d2676e7fa888 : SUCCESS in 35m 59s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6e96620d5a6e4a01a3074039802570b8 : SUCCESS in 35m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/6039f30ee2e845f68d58e1980f7df1c2 : SUCCESS in 35m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/dcb2335a87ea441696dfe91e8d69af42 : FAILURE in 18m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d3d4eb7ea75f456ba10a9c07e6f8d315 : SUCCESS in 45m 40s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/82062fd896604a2aa9cdef1ab6f32f81 : SUCCESS in 1h 19m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72a95cd3989e44f588d463a8a327e9dc : SUCCESS in 1h 07m 01s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/4e364527c05d4442b9d2564b9a213429 : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2c1b84b7e48b3e676ca1f971afccb4f7c5594e34","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-07 10:26:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0570c1bcbaeb362b4559f8b7e4aaec6e06418919","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 12:06:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"787eec34523211353db73600b2796214e593ed51","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-05-07 13:55:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"270a6b592c1364775a524d13ac15bb72638979af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:56:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a16493e9aa3d30f5693f1bbb0579e399af1cbc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 19:52:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad89ae27b9284b67828745ec0921b8f7 : SUCCESS in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b465f3b2b54b9ca3ec7bbab7c30c32 : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d14e52845c34414a7508b00dd32afef : SUCCESS in 39m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ae76a0e4684427ab22fa4ae53e89436 : SUCCESS in 40m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc9174d890647809f7703662498f046 : SUCCESS in 17m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/83255bb2ae5e4880b138dbe0586af531 : SUCCESS in 1h 10m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c217827a406f4f92babf94c52f314b99 : SUCCESS in 1h 09m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa7e3ec8c4f9459f9d05988f99c7f4ab : SUCCESS in 17m 28s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2ab99cc6940a4b4792aa8059ab471299 : SUCCESS in 32m 47s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/8cbe370892b4447dba7a465f986fdc2c : SUCCESS in 42m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6e587d98efde4cf2af02eb40e62e9690 : SUCCESS in 1h 11m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92eb464aaa384ea598039b3c98da3e1d : SUCCESS in 1h 05m 14s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/00112b0131a04d699da06c6eaed7059d : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4204586ae7baf7c74010589537f8029741d367ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 19:52:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0ca2ebd3de6e2b5748b25f9bd1fcf2f1f5dfa760","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 19:54:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41ac33029ab430ea6a04cd7cc302e7a : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d81aea84313549419cea359b0c4a7b2f : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba89d27793c2d3a26ad95642660fa9bd820ed3be","revisions":{"ba89d27793c2d3a26ad95642660fa9bd820ed3be":{"kind":"REWORK","_number":1,"created":"2020-05-06 15:16:17.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/87/725887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/87/725887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/87/725887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/87/725887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/87/725887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/87/725887/1"}}},"commit":{"parents":[{"commit":"9e00e99866572435e71f5d825e035a84a1c5d7c8","subject":"Merge \"Check timestamp of signed EC2 token request\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9e00e99866572435e71f5d825e035a84a1c5d7c8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-04-17 03:35:46.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-05-02 19:35:10.000000000","tz":-420},"subject":"Ensure OAuth1 authorized roles are respected","message":"Ensure OAuth1 authorized roles are respected\n\nWithout this patch, when an OAuth1 request token is authorized with a\nlimited set of roles, the roles for the access token are ignored when\nthe user uses it to request a keystone token. This means that user of an\naccess token can use it to escallate their role assignments beyond what\nwas authorized by the creator. This patch fixes the issue by ensuring\nthe token model accounts for an OAuth1-scoped token and correctly\npopulating the roles for it.\n\nChange-Id: I02f9836fbd4d7e629653977fc341476cfd89859e\nCloses-bug: #1873290\n(cherry picked from commit 6c73690f779a42a5c62914b6bc37f0ac2f41a3e3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba89d27793c2d3a26ad95642660fa9bd820ed3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba89d27793c2d3a26ad95642660fa9bd820ed3be"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
