)]}'
{"id":"openstack%2Fkeystone~731783","triplet_id":"openstack%2Fkeystone~master~I331e6024b6d6c618d7c29ab09bf9d183ee74f85b","project":"openstack/keystone","branch":"master","topic":"doctor-memcache-log","hashtags":[],"change_id":"I331e6024b6d6c618d7c29ab09bf9d183ee74f85b","subject":"Doctor symptom for enabling caching logs","status":"ABANDONED","created":"2020-05-29 14:58:20.000000000","updated":"2020-06-29 13:10:29.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c7591a76ede8b65bfde58e5b30aae57e0bc55fd","_number":731783,"virtual_id_number":731783,"owner":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-29 17:34:27.000000000","_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"date":"2020-06-29 12:59:28.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-29 16:35:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 16:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-29 12:59:28.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"287dc2c45d7299bfee50c2cba831331a1b3ced07","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-05-29 14:58:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81107f3f75a39bc13af4703c86244fac7419e9f2","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-05-29 15:11:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"797c7661d9e12472e0767192092db2c9c7e13afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 16:35:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a3c8839523c4c529b1bebee79206f78 : SUCCESS in 19m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/398f3bf2081042889057f494a9ce9da9 : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727eec6663a941b6909596dd4137cf03 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e2d3eeb203c4b08901cd8665b08d251 : SUCCESS in 25m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2315d22d8cf4e78928b8b39c8b8888d : SUCCESS in 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8419a189e04e9dbb37e9bd8830f186 : SUCCESS in 16m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/836a8eb2120e457181d42cb564206e85 : SUCCESS in 1h 20m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0694da174b1e4cb4a7b9910b4a0584d0 : SUCCESS in 1h 13m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c3d28bc5e1e14389a0c318175cf3f17f : SUCCESS in 34m 40s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/4cf4f8f4fd0c44cf906e0d02aaff7056 : SUCCESS in 37m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/ab263e226ac74a03b2343cee9bc8d8c3 : SUCCESS in 36m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8e224e4f255c4ae388b5e60d0bde9d08 : SUCCESS in 18m 11s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ce4449c9662740598dfcb1e5d54c99f4 : SUCCESS in 38m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4dc191018e54d79b4c2150a09d46f59 : SUCCESS in 1h 06m 02s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/cd695964969b4b80a656765150922bac : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5a6c5d44db6c6b37b8ec8f205cbedfb1e8644952","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-05-29 17:25:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7098e264304ca001328476bbd4c4f785b059b05","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-05-29 17:34:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45415e6df0d89f2674f1aab24284166c3fa76b39","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-06-18 15:16:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29d4f058f1a3cb6d6098882adec55852faf7ca8a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-06-29 12:59:28.000000000","message":"Patch Set 2:\n\nRaildo, are we still going to consolidate these two approaches?","accounts_in_message":[],"_revision_number":2},{"id":"616e26dbdf7f040a97b27cb6d6f866026423e7d0","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-06-29 13:10:29.000000000","message":"Abandoned\n\nAbandoning in order of https://review.opendev.org/#/c/737579 It\u0027s an optimal solution for memcached backends, we can restore this later if needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"772b807b22ff5c57cc05cc803a8096552f75bd89","revisions":{"533b7703f5f059dcc3d9984f3ee318bc9a503e1f":{"kind":"REWORK","_number":1,"created":"2020-05-29 14:58:20.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/83/731783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/83/731783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/731783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/731783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/731783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/83/731783/1"}}},"commit":{"parents":[{"commit":"73a66464f38d03c4716d7d90c4980508e5da1ee8","subject":"Merge \"Update keystone Making an API Change doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/73a66464f38d03c4716d7d90c4980508e5da1ee8"}]}],"author":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-05-29 14:54:34.000000000","tz":-180},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-05-29 14:54:34.000000000","tz":-180},"subject":"Doctor symptom for enabling caching logs","message":"Doctor symptom for enabling caching logs\n\nIf the caching driver is improperly configured and keystone fails\nto connect to it, it doesn\u0027t appear to be logged out even with\ndebug\u003dtrue without enabling the debug level for caching itself.\n\nAdding this doctor symptom so operators can know exactly what is\nneeded to see the caching logs in the Keystone side.\n\nChange-Id: I331e6024b6d6c618d7c29ab09bf9d183ee74f85b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/533b7703f5f059dcc3d9984f3ee318bc9a503e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/533b7703f5f059dcc3d9984f3ee318bc9a503e1f"}]},"branch":"refs/heads/master"},"772b807b22ff5c57cc05cc803a8096552f75bd89":{"kind":"REWORK","_number":2,"created":"2020-05-29 15:11:40.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/83/731783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/83/731783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/731783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/731783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/731783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/83/731783/2"}}},"commit":{"parents":[{"commit":"73a66464f38d03c4716d7d90c4980508e5da1ee8","subject":"Merge \"Update keystone Making an API Change doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/73a66464f38d03c4716d7d90c4980508e5da1ee8"}]}],"author":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-05-29 14:54:34.000000000","tz":-180},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-05-29 15:11:06.000000000","tz":-180},"subject":"Doctor symptom for enabling caching logs","message":"Doctor symptom for enabling caching logs\n\nIf the caching driver is improperly configured and keystone fails\nto connect to it, it doesn\u0027t appear to be logged out even with\ndebug\u003dtrue without enabling the debug level for caching itself.\n\nAdding this doctor symptom so operators can know exactly what is\nneeded to see the caching logs in the Keystone side.\n\nChange-Id: I331e6024b6d6c618d7c29ab09bf9d183ee74f85b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/772b807b22ff5c57cc05cc803a8096552f75bd89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/772b807b22ff5c57cc05cc803a8096552f75bd89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
