)]}'
{"id":"openstack%2Fkeystone~740219","triplet_id":"openstack%2Fkeystone~master~Iedde135b9de03229c27ed57638d0c404169f43ab","project":"openstack/keystone","branch":"master","topic":"drop-os-testr","hashtags":[],"change_id":"Iedde135b9de03229c27ed57638d0c404169f43ab","subject":"requirements: Drop os-testr","status":"MERGED","created":"2020-07-09 10:15:42.000000000","updated":"2020-07-29 14:20:08.000000000","submitted":"2020-07-29 14:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740219-1596032301614-07a05074","meta_rev_id":"c7881f32fc78d03861ac271318ea1fdb0086cc65","_number":740219,"virtual_id_number":740219,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 14:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2020-07-16 00:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2020-07-09 15:36:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-07-10 08:24:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-07-24 04:40:49.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-07-24 04:40:49.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 15:36:50.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-07-10 08:24:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-16 00:35:47.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-07-24 04:40:49.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-07-29 14:18:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1020e7ed4eb4e05e23d188f4dad08ba5ba8563f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-09 10:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24d1bde8ed798b9e31116e7a2ecebd5c115efcd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 13:27:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8809544778034385bdf3f05bc8e29d0e : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aee9295291c746498432e6d6972f2150 : SUCCESS in 16m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2e83734a6cf40c2a2d5a97a62125c1d : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed03c51b53404fb98e4deb4ce8983ca2 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc966a917ee4fe9b5493b2895f6cc8e : SUCCESS in 15m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3dc4c0b59904ddf8f5f31a1afd0e5cc : SUCCESS in 15m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5b875b2976498f9055154d4db9996d : SUCCESS in 14m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/270d7f3b9fe145ceaafb31d9ced68cff : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/6c4c9e27ec3a41aa941caca3879a85ca : SUCCESS in 45m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4ecc0201f234abebcb99e6fc2b984f0 : SUCCESS in 59m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e43ff7bdba404372902cb13a95d739a6 : SUCCESS in 5m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c10a05f9694044bba675242d3116549a : SUCCESS in 24m 47s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/268d50c38dbc4544a12dfa54a7b9a47f : SUCCESS in 24m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/714dd0416f964b9d8ff32f5b32b9ed6f : SUCCESS in 40m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8903914c22314c8092281aec7b2d4e56 : SUCCESS in 17m 15s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1059d9228ac34708815ae569d2bc4282 : SUCCESS in 41m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdb1f65544314737abed1e8619bdf206 : SUCCESS in 56m 56s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/2d1e0f7bc05a4bf4bade8f2723bbb6f0 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c9abce2eb33f579e9123e19c6838ec62aa2fc03f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-07-09 15:36:50.000000000","message":"Patch Set 1: Code-Review+1\n\nMakes sense.","accounts_in_message":[],"_revision_number":1},{"id":"fbc34a2ac66dcb9216b1a5a269c86d66ea2d09b1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 08:24:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2bbed59a56f597d472243119eaf9cf876f698b37","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-07-16 00:35:47.000000000","message":"Patch Set 1: Code-Review+2\n\nnit: I\u0027m assuming the commit message should be \"we migrated to stestr\"","accounts_in_message":[],"_revision_number":1},{"id":"a414a73d9a58cdadfcdf963b3da2799178167200","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-07-24 04:40:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a29301d40af5fbb6fcd149dc7e18a5f13deada87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 04:41:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dde018b6208a02ca403da224c86e82fa12e5bfcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 06:25:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4af019e437ee4ee1ae9d280b2ee100c4 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dda0528e466d427798944556740213b7 : FAILURE in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6935c7011d6b481f93897b2008431726 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d830b6506ff45a6bd41819b0c7690a0 : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b17f46d0455a4198aa67688b8586c3c0 : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd913c90ef2454aa8b1199f6278a41c : SUCCESS in 12m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/57f1d3f3902b46a5a5901da8c0bc1c5b : SUCCESS in 3m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/f2963606aeda4eee99fbb1d8df41bd34 : SUCCESS in 55m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f83e46130a348538aceff55695e4a07 : SUCCESS in 56m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63f846c629541dbb6626fceb68702f6 : SUCCESS in 6m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6dc2422962c44e12bcfcbe68d43004f6 : SUCCESS in 41m 15s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/5bc2cf017e95422b9e038d3679d65181 : SUCCESS in 28m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b188858fb23f4305950a5f6cb0373bec : SUCCESS in 1h 00m 56s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/323e71f1d2084effbade1ac49dc16ec3 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":1},{"id":"dc753e8fe337cbbaad121500714a95b965bbca93","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-07-24 09:08:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99860c90bb16202b11232787f2c5f12471eba82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:40:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/754f3fe59ea648b1836c43ddc1a22773 : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d32ad723df14ccd81a4bac40dc3f009 : SUCCESS in 20m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56c540a34bf24a85b28bdacd6cdcb033 : FAILURE in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d87712914df48ce8f34553b5d703232 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffdd230c7b8c4a5e84774ae308a1124a : SUCCESS in 16m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2e703de57a3463a8e8f9b06c9bf5fab : SUCCESS in 15m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe385271be745be92c2fb836d12bafd : SUCCESS in 18m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2e9f191ce07468286526c69e7dc1560 : SUCCESS in 3m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/8c68cd1c02ce496bbb200bc77b0ef5ab : SUCCESS in 1h 10m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b29352bdd73044d3a2bf5e0a0ad8daef : SUCCESS in 1h 30m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/882582af42b24a0784e9dead49f27217 : SUCCESS in 6m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1994c6e6a69845d6922687bbd1b5b97b : SUCCESS in 35m 18s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cda636d80f6d48a5ab52a4be3b434895 : SUCCESS in 38m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/ced27eebc8dc4bd2b15a06c25dfd64f5 : SUCCESS in 38m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9ba4cbb5803747bfb624679994b9dd1a : SUCCESS in 17m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/19295939e5204cff9a1a22cf462ce917 : SUCCESS in 39m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21719fec0d52400dae96f52bdbb2c1da : SUCCESS in 58m 54s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a71e24765f104b83a1d18da6de3e0877 : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":1},{"id":"19bbd1a7723b80395fd6b3304137c494f2ce28d9","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-07-29 07:35:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44ef89c72a552c3a05151d4c4e5211b1fbe87976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:14:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ad84737724144878d28c9d10e5e2ad5 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdd2eb0b0a124f99847b6bd419dcf1ff : SUCCESS in 15m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f98d548a5094da5902c5a40bbb3b52a : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139482fc9c044ca4af0f881d90c80de8 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c0b689fa1d94f79a8efbbeb86ad9e93 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1bed6f25aa54b85beaa8b00927e08a5 : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cc79a93b6441c69e9f585af4923d1e : SUCCESS in 13m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5dc7cbf9b65e4b81b745595ac2e8596f : SUCCESS in 3m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/ff04c6ba619440a3a1e536ee3cbc5e86 : SUCCESS in 56m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17e27015aa614eec913612d0a19f4182 : SUCCESS in 1h 27m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/756907e102d74c089ec5b19049cc01ae : SUCCESS in 7m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/88a60b0edf144417a67236201e202f65 : SUCCESS in 28m 19s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b87798a6b254742bb6aa8d9635c4b0a : SUCCESS in 48m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/37f34f111b204b62b53a5c0ee667a5d4 : SUCCESS in 37m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/721a82ce62f34c809d945fb7dd0a4dab : SUCCESS in 20m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/de56439dc9814fbb8dac53729d4100e1 : SUCCESS in 42m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12ebedea462640c385740ad3a0229e4f : SUCCESS in 51m 36s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/6908fe313fde40c6971552e233bfafd2 : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5804e68005dfd9b724efe1379581668f19cad6a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:14:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bbab4925f899cba115e5a132fad522f546c6544","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:18:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fdc4df6252e947548a7363e7fcbdf477 : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6059832f33b948cfb82a1915e8e2ac3c : SUCCESS in 35m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308aab3ae52d4b17a4b44ebcf9c1f456 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce034d31d7804e89a694be0dac90c754 : SUCCESS in 18m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/395eb844d329430b9ec4dd1b2603d391 : SUCCESS in 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c4a63581224796a2408b47173e8e3d : SUCCESS in 10m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e9b2005b9f941caa4ed21b9d659e844 : SUCCESS in 3m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/14a65a9fcca4489c8ba804fe1bd274b5 : SUCCESS in 56m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc2d02b5b88d40d484dc8a383a837695 : SUCCESS in 1h 43m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9722cc27ffd1454094e3637914bff406 : SUCCESS in 5m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/78c2775afcc8497ab674697ce3730f59 : SUCCESS in 33m 19s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/d6e6148ecbe847b18691b8cbd8d4d53c : SUCCESS in 40m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d8ea1d5cc834abaa584732174714324 : SUCCESS in 58m 31s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7154affdda2c4b9ab6a27b9d292ae3ac : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b3b9c9abc517fb61c08faa56e49c59889fdb4955","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:18:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"370d64b036289e6ab2c0af6d53724c88414d62a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:20:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a52d9e18fa3547f8911c14ce0d4c475a : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d9eb5d77404c2cb085b2b2a78e728b : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7869e8f7745d4256a193fa9b4fda63ec : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b552d878599ec69d7fa7a6a104ca1d07ad78cc9","revisions":{"5b552d878599ec69d7fa7a6a104ca1d07ad78cc9":{"kind":"REWORK","_number":1,"created":"2020-07-09 10:15:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/740219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/19/740219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/19/740219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/19/740219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/19/740219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/19/740219/1"}}},"commit":{"parents":[{"commit":"05e9ac007ff6659d44497e7c6173249ad1be4d93","subject":"Merge \"Add an enhanced debug configuration technique to caching guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05e9ac007ff6659d44497e7c6173249ad1be4d93"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-09 10:15:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-09 10:15:04.000000000","tz":60},"subject":"requirements: Drop os-testr","message":"requirements: Drop os-testr\n\nWe migrated to os-testr some time ago. There\u0027s no reason to keep this\naround as a dependency.\n\nChange-Id: Iedde135b9de03229c27ed57638d0c404169f43ab\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5b552d878599ec69d7fa7a6a104ca1d07ad78cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5b552d878599ec69d7fa7a6a104ca1d07ad78cc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
