)]}'
{"id":"openstack%2Fkeystone~743727","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I5d5d31c8d89b0d2812a5e9be18a81611041e6da8","project":"openstack/keystone","branch":"stable/ussuri","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I5d5d31c8d89b0d2812a5e9be18a81611041e6da8","subject":"Port the grenade multinode job to Zuul v3","status":"MERGED","created":"2020-07-29 11:24:06.000000000","updated":"2021-03-10 01:38:37.000000000","submitted":"2021-03-10 01:34:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743727-native-zuulv3-migration","meta_rev_id":"0bf2c14969d7bb761fcd33fed6a301b66b1e1619","_number":743727,"virtual_id_number":743727,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2021-03-09 21:04:54.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 01:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-11-20 02:31:34.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2021-02-23 22:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"date":"2020-11-18 14:10:51.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2021-02-23 22:19:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 23:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 14:10:51.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-11-20 02:31:34.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2021-02-23 22:19:56.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"bca7f8fae9ffe567e1c1b7b13cf6a8d082b18aaf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-29 11:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00b9aeeae3a3f19e6f60e74389004ff0ad49c57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:33:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14fbdff207084d858f941cbbf78da406 : SUCCESS in 24m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09dcd247b3054d10a9b129adca00a199 : FAILURE in 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8f406b2cd3465086caa646d34f56e4 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3cc7e567eb549e89c3f329a085c1731 : SUCCESS in 32m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60018812772b4dac8edbc4207f2dac70 : SUCCESS in 16m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b6b41baca14f69b6dfa4b4c63d681a : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/968af0c0ff424ca9b14142958ed045dc : SUCCESS in 1h 10m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e84eaf7cbd5348889cdb0f49b0662170 : SUCCESS in 1h 24m 47s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1271a86350e74a6d908ef26a8eaaec05 : SUCCESS in 44m 05s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e1b794bd3c21439494b9ecfd2af838ec : SUCCESS in 43m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/6c7e47411fe1465fbae8534587a92ddf : SUCCESS in 39m 27s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c287b809ad7d424e97ee30f7ca0e52d8 : FAILURE in 17m 47s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7748d7792c834aed84a1f1d5f4d9620e : SUCCESS in 25m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf2d1a8acb8b450ebdb68264d9cd273c : SUCCESS in 1h 03m 14s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d58bf4b64af34c0a8ccf94baf200c6ee : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"67ee7f9b46684b6ee710664d178777c89a5975a7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-17 19:09:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4ccf7993fc060978e9c3603c2d5d75cf8a07480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 23:12:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22013666fa3a4610a0eab0db2d409aa5 : SUCCESS in 15m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dcaad40076d41e9b6fbc4b27aa3516c : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc7470dac1442e2b15826d3663e6d98 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc5f619fe58d4ceca4bf9ec2af44eaad : SUCCESS in 12m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c90149746730416d9d4363c7a71642dd : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fe74c84c3f4c5988f33b2044ca31d1 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/bf4b918af801428bbed38de1a3353b33 : SUCCESS in 1h 21m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf4c7ab5d57f4ed8a3f399041aa0621f : SUCCESS in 1h 20m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/929ef47b9eaa43698b58428bd2964865 : SUCCESS in 33m 06s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0caf4ef0ea6b45b7897e75d3cde85032 : SUCCESS in 38m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/a0c3bd454ccd461fad5cbd66d3663272 : SUCCESS in 41m 22s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/50cfa3b1192b4c5d86cd8bf00a04ecb8 : FAILURE in 18m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5a55f1c73b2946539ca09687d11cb94a : SUCCESS in 39m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1da97b14618643978c47c7288f25c53c : SUCCESS in 57m 53s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/faf5d369eeb346d4a3b75a2eef9832e0 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0db7864d85887605b953bdcc842a3dd69abb15a4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-19 23:29:46.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"7436d3aff3262427220d1b262cd9a4387f1fc235","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:31:14.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- keystone-tox-patch_cover https://zuul.opendev.org/t/openstack/build/f9bfef802f5147da873645c893cf7d79 : FAILURE in 4m 26s\n- keystone-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1d1293bc29d40d59efda019076c8517 : SUCCESS in 1h 55m 30s\n- openstack-ansible-keystone-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44606ac99e68435a9e43c19d77e26d18 : FAILURE in 5m 10s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/5f37518cfc7142beac45687e0a05c107 : SUCCESS in 1h 24m 33s\n- tempest-full-py3-opensuse15 https://zuul.opendev.org/t/openstack/build/7930eccd9c5043648c660ec34adb559a : SUCCESS in 1h 26m 32s (non-voting)\n- keystone-dsvm-functional-federation-centos7 https://zuul.opendev.org/t/openstack/build/97b9e9ed5d154f2ea2f3ebca0fb2d6e1 : FAILURE in 5m 42s\n- keystone-dsvm-functional-federation-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a335f4a4e35342de8c17396b8efba499 : FAILURE in 11m 39s\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f89cb05d5a2c405a9415bf1b7c2fe905 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c6652ba878d33b12741295aaf7f5117c46ff2842","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-20 02:31:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0691db84fcbfdb64b578e348c499d1a3c26dca0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 22:19:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e157116a66134949a5f73d9514be230ed917b77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:21:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed8177e7253746994eb9517853ed221b61b083ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 01:17:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/004f85f64d8e4c7b92c079f8e7b90702 : TIMED_OUT in 41m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4071e1167d344feae06a30074913fc0 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbf3f82f7d52400b925e13165a6cac8f : SUCCESS in 14m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c899f45f06e74de08c09415a1ba5fff5 : SUCCESS in 14m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ada88b001d4ace8067a8223f96eb8b : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/d94207657dae4197a3e3364b23c106b4 : SUCCESS in 1h 17m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55f6906b46854f2ebe07da78940ae5d2 : SUCCESS in 1h 21m 17s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/16421900c1e64c9b953741b6f13ca875 : SUCCESS in 28m 21s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/fb817c8d22764d84bdc5130d602806cd : SUCCESS in 44m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47bf8f030e014753b1e672b31f1fb175 : SUCCESS in 50m 51s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/80643d85368a45fd9bdbce5d9b60e8de : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2939cd53016825a46b8d2efd078c346c08430826","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-24 10:00:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e3d9a1aa10030fb35990cd5f4763a1475f6f496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:04:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7787640cbd05425997a38da6b3a3b189 : SUCCESS in 16m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c6feb355e814a9799ad69f9f78bbc33 : TIMED_OUT in 41m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a295918952dc420db4ef1c359319b98c : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2844691671464c35be1d6d142f36391e : SUCCESS in 14m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8ede088012a48ac843452aba3511442 : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e466dcbfd93b4cc39eb8c27bbbb57525 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/c25e3c09b6f740e19a1c33b7f08cee24 : SUCCESS in 1h 03m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1505e1c4046b4234a08d9ab5eecc57cf : SUCCESS in 1h 24m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/14b759f207b343c89e1654b81de0ad60 : SUCCESS in 28m 01s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6dd25615ff344ea88994db49cf57a8c2 : SUCCESS in 33m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/aa7e01fb225c4fb0b97fc4e598b80fb4 : SUCCESS in 48m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/be8e2ecccaa84378bc67f6613ec8c2b3 : FAILURE in 19m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/dbbb5f991b384bbab8d663805dd03ba1 : SUCCESS in 38m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11e1409c8f824bbd94014bd894f547a6 : SUCCESS in 1h 00m 59s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/4deb7884aead488b8eefe4ebfc8099c1 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":1},{"id":"856734b656f726c8b655794d3cb30ef68d4fef0d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-24 13:08:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a802704edbbc446c8bd04ffc396b5f54b07ad8e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 19:41:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88c68c3f51eb4824840febfa24f7e634 : SUCCESS in 14m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301196c6af4046738d9b364a3b2bd3da : TIMED_OUT in 41m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f35c012e1f4b4ba1e37a9ca60b0579 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1983829eb9224345b033a5df7ad824b8 : SUCCESS in 11m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/feee95c8c5a14e4ca3acf995234ee8e3 : SUCCESS in 27m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fbe3b69e6641e0840572708d18550b : SUCCESS in 15m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ba5e9a111f29496dbc5d6f89972416b8 : SUCCESS in 1h 05m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c260603e931f421cb3699f1e43dab6b2 : SUCCESS in 1h 29m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5423ef08402e4e75b46e6b8424f41949 : SUCCESS in 28m 26s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/22e1b93265e24f33bd2c2ea4a76b7928 : SUCCESS in 40m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/029b5f3ade6043e9b9d3319bf11f03e6 : SUCCESS in 42m 50s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d855fd5ac73e4fbda7ec466b859a1d0e : FAILURE in 17m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fba9859a317b47f89c0a6532024bc32f : SUCCESS in 47m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c517e66a71145d58dad9d17d7bd8c17 : SUCCESS in 53m 37s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8392cd656fb44206822c33f3f0ad3bfe : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0d5801c90f65ca5e3e628a017651ad39c37b7aae","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-24 23:53:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0d28631f1da93687a902b4d833e7a8a63f8f8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:08:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3473402bf8ba4f46a8c3174781a1cb96 : SUCCESS in 19m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3a90706ad184c83bd0c9c8d60156413 : TIMED_OUT in 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0fe6a424954b0b8bc62516b93a7abe : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5118a2d3c704070b05422991ac2c7f8 : SUCCESS in 18m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9449273ee9674f8a925a92570490bab5 : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a877242f046f41cdbf3b26931c550d78 : SUCCESS in 18m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8ad8a22125704cac82804f17490b941f : SUCCESS in 1h 06m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab70010415d74f02a09067bc2a00e9f1 : SUCCESS in 1h 26m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8c8407c0029f4125b96809be69eb2ac1 : SUCCESS in 29m 32s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9dc24f94ac604021949235a0b28cbd5f : SUCCESS in 52m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/401533d341c14d9c8a5495b21c06083a : SUCCESS in 42m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9a2cc21aa29b41bcb372c4a11ab97e3e : FAILURE in 13m 06s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/84e00b123d6540d28a52b95a658849bf : SUCCESS in 38m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3545a7d3244e4f8bad68457b7c5e6c6f : SUCCESS in 1h 15m 00s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/4f737b2f81014624989c768f7573c1e1 : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0b1c4cbb5f272d13f5ab29d170d1b3fb84493fee","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-09 21:04:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6d0dc2437f967cba19a7819635fbd629c9adc50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 22:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/442a844d73a34117866d9a28de64ff75 : SUCCESS in 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318f6002c31049b6906b0b717d66c336 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a95af58c3454d8a8e5d3d8252368ca1 : SUCCESS in 14m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aebc173c1b874dd3a28b05b67b7bd847 : SUCCESS in 14m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1036106e27f4abcba90e2dc8b597a32 : SUCCESS in 18m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c1010fb8ea734a499a03507da9f129b9 : SUCCESS in 1h 24m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc1b94d9ee5447338b6bb176e00d8ea8 : SUCCESS in 1h 27m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/646f83d500b14c349ca364d1102393e8 : SUCCESS in 31m 48s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ca253f96f7cd4f18a2c5548fd93addca : SUCCESS in 38m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/b98a9cf930c1432496a8b5ce2cd7c1da : SUCCESS in 48m 24s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ddd28d490a69457cb190c9dabe8bee96 : FAILURE in 17m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/83790ee1a3154025a474930e9174b4da : SUCCESS in 54m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/042ee00d8018416eaff2ac5894d90c95 : SUCCESS in 57m 15s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d052a90207584b00b97cf2787b6bce83 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b9af497794af761f4668fb7634d16e37a88c2d52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 22:38:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7e7856e85c2c599c25c7651cc2f8f40599470bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 01:34:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96647c3dc6a74dec876b7e7736d6f52a : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afbe30b1ae434382a5cc5d4866c6ff77 : SUCCESS in 13m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf96918b9e584514a2f3f33ac9f9b9e0 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa8276f499c4e199a43893ab16504ba : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/132a8ca21e214c4384b6e91ac4a37881 : SUCCESS in 1h 13m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74bba89f9af445d7b32eb383691229ec : SUCCESS in 1h 33m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1a9cac4d95164cc7ba07424f7d01c523 : SUCCESS in 34m 31s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/a0f324be7e764f159df70543e9888ebf : SUCCESS in 43m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e77d4ff2991a42dabef59e085008e030 : SUCCESS in 1h 03m 37s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/262fffd3a3a74f6b910b433029244d89 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9bc2d88365add461c7d8cde7a0d62072f7773a24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 01:34:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0bf2c14969d7bb761fcd33fed6a301b66b1e1619","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 01:38:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c485709f36143e6b4adee7277e9e4cf : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c37797c51f735c3bf2ac35a534b54f1d97f3819","revisions":{"1c37797c51f735c3bf2ac35a534b54f1d97f3819":{"kind":"REWORK","_number":1,"created":"2020-07-29 11:24:06.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/27/743727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/27/743727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/743727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/743727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/743727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/27/743727/1"}}},"commit":{"parents":[{"commit":"624c2101ce2919431f93cfab4db58249c25ebd00","subject":"Merge \"Switch to new grenade job name\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/624c2101ce2919431f93cfab4db58249c25ebd00"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-06-05 12:25:26.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-29 11:23:49.000000000","tz":120},"subject":"Port the grenade multinode job to Zuul v3","message":"Port the grenade multinode job to Zuul v3\n\nIt implements the same behavior of the old one.\n\nIt is worth noting that the existing job the MULTI_KEYSTONE does not\nproduce any effect because this change hasn\u0027t been merged yet:\n- https://review.opendev.org/399472\nand when that merges, the following devstack-gate changes should be\napplied directly to this new job definition instead:\n- https://review.opendev.org/394895\n\nBoth changes are out of scope for this porting.\n\nChange-Id: I5d5d31c8d89b0d2812a5e9be18a81611041e6da8\n(cherry picked from commit 3be8d40fc56f876f3dea30812e6352e1e482bb86)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1c37797c51f735c3bf2ac35a534b54f1d97f3819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1c37797c51f735c3bf2ac35a534b54f1d97f3819"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
