)]}'
{"id":"openstack%2Fkeystone~746043","triplet_id":"openstack%2Fkeystone~master~I06b02b2ebfed35d4e82c5fc35ce8eb0bb20b2fc5","project":"openstack/keystone","branch":"master","topic":"follow-up-745752","hashtags":[],"change_id":"I06b02b2ebfed35d4e82c5fc35ce8eb0bb20b2fc5","subject":"Follow-up for bug-1891244","status":"MERGED","created":"2020-08-13 10:16:00.000000000","updated":"2020-08-21 15:14:51.000000000","submitted":"2020-08-21 15:11:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746043-1598022718205-f63fd487","meta_rev_id":"c4ae9d6519d9b8ca4f3c66f1b6de75b470b09e43","_number":746043,"virtual_id_number":746043,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 15:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2020-08-21 13:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2020-08-21 13:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"date":"2020-08-13 10:29:19.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-08-13 10:29:25.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 10:29:19.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-08-13 10:29:25.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-08-21 13:42:14.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-08-21 15:11:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cc3dae5ef9bd43437801d3498b00e90db401717","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-13 10:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84a03eb251a2ae12e316cdf07761ea1e524a9185","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-13 11:14:25.000000000","message":"Topic set to follow-up-746043","accounts_in_message":[],"_revision_number":1},{"id":"ff2c09ed3f41bea155d66ebeb156dcf99bf4bf4a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-13 11:14:45.000000000","message":"Topic changed from follow-up-746043 to follow-up-745752","accounts_in_message":[],"_revision_number":1},{"id":"e5d3d009989b873d3384937a06a319ae2373a81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:48:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c08a17d2e3d0467890f15a65f9a403c3 : SUCCESS in 16m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5349e001e5e24be8b9fe4308c217a111 : SUCCESS in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce8172b0480b431580996d93f4f16752 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50cdae0fa7f443cc9f88c08a853c37e5 : SUCCESS in 11m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3b3a0478d2448ed9fcf073b70468949 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded9e48f01814de1a85fe5b08f9373b2 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/6e859cf39c474385874c8d4c2743a655 : SUCCESS in 57m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad597610de5d4e5585269bf61b40b41c : SUCCESS in 1h 28m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/af37e7a0da7a4aedb08ee2b8b590f102 : SUCCESS in 21m 10s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c0962456a8d4c0badd1cb62ee50f315 : SUCCESS in 22m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/dae55d8517e24129aa3c0b2679e32318 : SUCCESS in 33m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/df08a5ecf8ce4d7194c9cb45cf9d55fa : SUCCESS in 15m 26s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ddf440e9a5184d649c5302ba08efb18e : SUCCESS in 38m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7589530dbe49c8ad5544cf31cb714b : SUCCESS in 1h 10m 54s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/e4dcb8667dbc4f05b72a7aa439f7a517 : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7988307f0beab1aabcda236206317bb10ab64ef6","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-08-21 13:42:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good to me. Only minor changes to the wording of a comment.","accounts_in_message":[],"_revision_number":1},{"id":"4f05d9663dc959c858b285c7312fa27b3377414c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 13:42:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1f7efd727c81168c6e128824105fecf94103c7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:11:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b484d797f2654a84957d39f2475210d3 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb37619955b143159d44c4467e5d024c : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e7283913abb4695a26025496c464ef7 : SUCCESS in 12m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e0bf38b6964d21ab887e10a04a69ca : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc165063a9e4a47b72d3990bb12049f : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c94bc25b79b24658ba4a7299242a7806 : SUCCESS in 59m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3179d77a77324e1ea0d7bc0172a86aa8 : SUCCESS in 1h 23m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f006a6f14a3b4db883e7162043abe38c : SUCCESS in 33m 33s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/f8ab1efdb18c4216abed09fc7cf5c926 : SUCCESS in 32m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7192c8553f6c46eeb9e6bfce1f1d81f6 : SUCCESS in 1h 02m 53s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/69dcba67ad1e4d9fbd957a8d5748b207 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c0d72d1e422097128221a642a4442c4c05b79bc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:11:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c4ae9d6519d9b8ca4f3c66f1b6de75b470b09e43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:14:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c089d665ca4d8a80c03276678baf85 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88197d2175a6f56b012c3f8a3254dee5f540ada4","revisions":{"88197d2175a6f56b012c3f8a3254dee5f540ada4":{"kind":"REWORK","_number":1,"created":"2020-08-13 10:16:00.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/43/746043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/43/746043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/746043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/746043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/746043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/43/746043/1"}}},"commit":{"parents":[{"commit":"7d6c71ba26694c21110280e741b9ffe2d36a94ca","subject":"Support format for msgpack \u003c 1.0 in token formatter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d6c71ba26694c21110280e741b9ffe2d36a94ca"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-08-13 10:10:35.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-08-13 10:10:35.000000000","tz":330},"subject":"Follow-up for bug-1891244","message":"Follow-up for bug-1891244\n\nThis patch closes the review comments of [1].\n\n[1]https://review.opendev.org/#/c/745752/\n\nChange-Id: I06b02b2ebfed35d4e82c5fc35ce8eb0bb20b2fc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/88197d2175a6f56b012c3f8a3254dee5f540ada4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/88197d2175a6f56b012c3f8a3254dee5f540ada4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
