)]}'
{"id":"openstack%2Fkeystone~747924","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~Id80b17bdff015e10340e636102576b7435bd564f","project":"openstack/keystone","branch":"stable/ussuri","hashtags":[],"change_id":"Id80b17bdff015e10340e636102576b7435bd564f","subject":"Properly handle octet (byte) strings when converting LDAP responses","status":"MERGED","created":"2020-08-25 13:04:25.000000000","updated":"2020-09-01 18:51:06.000000000","submitted":"2020-09-01 18:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747924-1598986151937-436c11b4","meta_rev_id":"1efa2ce118cee9b33011c0df8c1541a835c99434","_number":747924,"virtual_id_number":747924,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 18:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"recommended":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"all":[{"value":1,"date":"2020-08-25 13:50:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":1,"date":"2020-08-25 13:21:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":2,"date":"2020-09-01 17:13:46.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-08-25 13:10:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"date":"2020-08-25 13:04:25.000000000","_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":1,"date":"2020-09-01 17:13:46.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 13:04:25.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-08-25 13:10:53.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-08-25 13:21:33.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2020-08-25 13:50:45.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-09-01 17:13:46.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-09-01 18:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3139491e06e208fed29fb9932d7b984e25d28837","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-08-25 13:04:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c0bad3066ea64ed38ad5a8f7b752753514eedd7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-08-25 13:10:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa84d56edc9efd2cfa98d1a53a1fefd7ced70bf6","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-08-25 13:21:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"107e851d29873f7df6f8690d3d38d62b50eaf4ae","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-08-25 13:50:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8cd9860ac3578f8af4cc033a3fca5c9fab45e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 17:19:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5065d954a2b04e61b064b620ceeffc0b : SUCCESS in 32m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b461d283c8fd402ab5958241f1402fc4 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19bddc9107014c70a22c54b4262d03b7 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bb29399328f4b53842de091cb703446 : SUCCESS in 13m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5158481d2c2d4db29be23bd556f7814f : SUCCESS in 14m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2f45b3e3604205bc288e3f155b3950 : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/5fc88a8359a84ae69905c1ca8cf9cdc3 : SUCCESS in 1h 13m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db21801429c34cd6b96b87484ef067b6 : SUCCESS in 1h 29m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad9276fe483d43fcb59c37a302836acb : SUCCESS in 9m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f8276c40656b43c0a45adbe43541e01e : SUCCESS in 29m 06s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2ca40459af634cbfbe838590a89e90af : SUCCESS in 35m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/9b7db412ea0447beb00977a154605c4e : SUCCESS in 36m 42s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ae1237321cfa405fa770c6569dec4e73 : FAILURE in 21m 56s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1021d7ed7cfe493cbeebcd78459215ff : SUCCESS in 43m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20e852b576c0474cbda6f1f4c322df89 : SUCCESS in 56m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/48e6ae9e29b349d0bf17eeea83cabde1 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"70b295de8aea1feeaab41e66c37c51f542831cc0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-09-01 17:13:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFixes a bug, proposed by another stable core, has PTL +1, merged in master","accounts_in_message":[],"_revision_number":1},{"id":"97417cbc6e6c8378d169075eb23c77f9ba3f43a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 17:14:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7597019eb442caf5a7439071b4d58a2b75e8471b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:49:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f04cd6e08a34593abdaf4e5fdd9e844 : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28e3409ff35b4d6b8a9b6bbad8512882 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ff9f7f393814745b817cb1579d85339 : SUCCESS in 23m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c003b4f0a94549da9cb741f3287396c2 : SUCCESS in 20m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf06f962f8cb48ed91eeedf287a1bd14 : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/b0ad4ec2b9ca486fb6c6d23b80ffeb5b : SUCCESS in 1h 16m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1049d41f89841ee96ea4a985e87a8c8 : SUCCESS in 1h 31m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8d78c415964da2ba9002bdd5183a60 : SUCCESS in 6m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e75d3c8965c847798efc06f93a1c50ae : SUCCESS in 21m 57s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/83a4d9025a844c42a7f05e4d3739f206 : SUCCESS in 25m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/344f855619e34b28afb25202a98e286d : SUCCESS in 52m 49s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/db253f38c3e84ae397393ee5db7da602 : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9f82f93835d454b43014bd8e134337c0251cd0be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:49:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1efa2ce118cee9b33011c0df8c1541a835c99434","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84687c52bafd42979bfff261adb90ee8 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/852712a6c5724b088f7c1ef49043d51d : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5870f69c12c034dd97b164345e85e6259ee7abe","revisions":{"d5870f69c12c034dd97b164345e85e6259ee7abe":{"kind":"REWORK","_number":1,"created":"2020-08-25 13:04:25.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/24/747924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/747924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/747924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/747924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/747924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/747924/1"}}},"commit":{"parents":[{"commit":"300e79e93b9966e988ec92a74bf9b49b71d77d1b","subject":"Fix lower-constraint for PyMySQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/300e79e93b9966e988ec92a74bf9b49b71d77d1b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-07-31 20:28:15.000000000","tz":-300},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-08-25 13:04:17.000000000","tz":-300},"subject":"Properly handle octet (byte) strings when converting LDAP responses","message":"Properly handle octet (byte) strings when converting LDAP responses\n\nIf LDAP returns a UUID as an octet string the LDAP driver will fail to\nconvert it to something meaningful. The error usually looks something\nlike:\n\n  ID attribute objectGUID not found in LDAP object\n\nMicrosoft AD\u0027s `objectGUID` parameter is stored and transmitted as an\noctet string [0]. If you attempt to use the `objectGUID` to generate\nuser or group IDs, you\u0027ll get an HTTP 404 because keystone can\u0027t decode\nit properly. This is unfortunate because `objectGUID` are a fixed\nlength, UUID format, and ideal for generating IDs in keystone. As\nopposed to using the object\u0027s CN, which is variable length, and can\ngenerate hashes that are larger than keystone\u0027s database table limit for\nuser IDs.\n\n[0] https://docs.microsoft.com/en-us/windows/win32/ad/reading-an-objectampaposs-objectguid-and-creating-a-string-representation-of-the-guid\n\nChange-Id: Id80b17bdff015e10340e636102576b7435bd564f\nCloses-Bug: 1889936\n(cherry picked from commit 8bf222ac5d390e25d306d35f69bd958b18bee4d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d5870f69c12c034dd97b164345e85e6259ee7abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d5870f69c12c034dd97b164345e85e6259ee7abe"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
