)]}'
{"id":"openstack%2Fkeystone~750788","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~Icad9383c8bc963843ef36fc24d745f29fc8798ed","project":"openstack/keystone","branch":"stable/train","hashtags":[],"change_id":"Icad9383c8bc963843ef36fc24d745f29fc8798ed","subject":"Update amqp lower constraint","status":"ABANDONED","created":"2020-09-09 18:54:47.000000000","updated":"2020-11-03 16:27:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00986742e034b67088a7c1a98e17aa7ffb639e78","_number":750788,"virtual_id_number":750788,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"date":"2020-11-03 01:17:53.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"date":"2020-11-03 00:17:58.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"date":"2020-11-02 20:23:59.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-23 22:56:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 18:54:47.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-10-19 14:47:52.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-10-23 17:53:25.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2020-10-23 22:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-03 00:17:58.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-11-03 01:17:53.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"}],"messages":[{"id":"a58f24c3183476ec183e3930df38f0c70e39b4a1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-09-09 18:54:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db17eb24f48e66a18a13081f3d6ffd99618fc401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:48:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8879993225ed4254b16243e5bf5c7c9d : SUCCESS in 17m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f4902a3e2d64b42959a51c2d3de9710 : TIMED_OUT in 41m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38085cfce5704f2ea30c1e78eda210fa : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea01897227e74cd4bad5e0ca975e4997 : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ebb8076e20484891c16e33600ad1d5 : SUCCESS in 11m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c8e597a5282480c955724b873be0f6e : SUCCESS in 14m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e9dd4085d344a4820734fdf9b999b5 : SUCCESS in 11m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/825f777a217748ca9b7022ca9555275c : SUCCESS in 3m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0decb060179b491199904cf0bebc3cf8 : SUCCESS in 1h 41m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/064161c6b0d04b9194d594eae2297a97 : SUCCESS in 1h 09m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/a8a9021ed714453885db840bc8a2599f : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e81fcd6d77f444399e5a5055eedf29b : SUCCESS in 1h 15m 55s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/375e0e6aa9cf466e90debe3ebbc3b6bd : SUCCESS in 39m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e39ee1589b9e480db5aeaf4e055bddd6 : SUCCESS in 37m 04s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/4472bafcece047cdad555b9623b1fb35 : FAILURE in 26m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cfdc4409124b4136b0adf033ce87d6f6 : SUCCESS in 43m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/60186201586445249f2bc58585719a35 : SUCCESS in 40m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d7fc59d3bab54adca478dc1f2a66ae04 : FAILURE in 19m 40s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/748ee31e74434ff1a6a8cc170bc1f2ce : SUCCESS in 49m 16s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fc4b79c9728f4d1c948e54e639f6563a : SUCCESS in 57m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79b2cc0e0be8459e8dc3fd4039599278 : SUCCESS in 58m 20s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/d8d35ad865e1448aa064cda452b127db : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":1},{"id":"054c0ff59dd1c2df404b1d8d70b380e3cdc9760d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-09-10 12:38:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b90d964e69af639c664efc8970a7974455d1941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 14:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ae032f70474c6fa67c9833297d5231 : SUCCESS in 16m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e179f4e455b4f4b880895d972555d03 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d506aac32dae43908fcba8c3c1d6273a : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eee885abbbb44f3b81b7e64b2544e3a0 : SUCCESS in 16m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1103cdd23fc9407297e6e526bf99b598 : SUCCESS in 14m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6318c2f8a136433f9d892d0d85ffd3d4 : SUCCESS in 18m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e76be55f7404b6d84a559eaa3990f24 : SUCCESS in 9m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2b98f7f28364494f9921ba19cc0e1333 : SUCCESS in 3m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4195ce5129704df49ad1d475c01e322e : SUCCESS in 1h 23m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7e6c63e7a1e24052a77cdba3dc825a42 : SUCCESS in 47m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6fb7af3cd25a4570b0f898488dcd0edc : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40424d33c834491d808633fd0ee8ba31 : SUCCESS in 1h 06m 19s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/e777f46d002a4cbb85b37472c77bcc43 : SUCCESS in 32m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/16bbd83a7e7044619bf1ff5f4fb4e755 : SUCCESS in 26m 56s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5bdc074c31ba4f01815904eba9051707 : FAILURE in 20m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/158caeae9e284d5099bd529b4d2dc8f8 : SUCCESS in 47m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/9272ebfb96b9466b9c2da3345063c389 : SUCCESS in 36m 25s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/93a5f72c12a64fbaa6b6b9bdd4df4bdd : SUCCESS in 13m 20s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/796f02e516dd4b33903235452bbe3c7c : SUCCESS in 36m 59s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1f23785a387e403c8d6f810a4213d6c4 : SUCCESS in 1h 01m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61d5e2ee04c44ce08b0d492664960937 : SUCCESS in 55m 05s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/31f39f1933db4bccbaf607b104e0758c : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b0aedae66c3183ebda452985c60eb276597a3af0","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-10-19 14:47:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"158d03d169ef4c8932d42347722a727848a6cdc3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-19 17:15:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSelf approving since we have sign-off from PTL.","accounts_in_message":[],"_revision_number":1},{"id":"06b8131987013d2d36cac1d141b1312e5f9293e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:16:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c779899ade348bb10f4fb57594cd5a04cd471810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 21:11:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a55a602bb21245d6bf7a5a08e732eb55 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811adb5f85414e4ea1cb50c52b320b06 : SUCCESS in 10m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f6febdad33e4cb0ad323973d53bb8f8 : SUCCESS in 30m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa19bc3675343babe22c9f1067b5470 : SUCCESS in 24m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da83e9d8d80543a7934079cab857468a : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0865f8b290d440e3a1f40888680f79eb : SUCCESS in 16m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1c3b9c8fe8348e9bbcdf5420273b51c : SUCCESS in 7m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eff75a15900c4881aad4ecc95c06a394 : SUCCESS in 1h 58m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e72269d591d2481f86cbd4725a997300 : SUCCESS in 1h 02m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e0aa4423a4694992997be682592f61e9 : SUCCESS in 1h 05m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7298ce14ca654d11bfb2405482f29463 : SUCCESS in 1h 29m 41s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/cba661efcbc149b4a52cf0d44c9ca550 : SUCCESS in 31m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c72cc428c54e4fa3a82e1c9ea07fd0db : SUCCESS in 32m 30s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/ffb17eb31c284e32b89c2e66380c7014 : FAILURE in 10m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e0c8fc5ce9a54849b583877367a48e5c : SUCCESS in 57m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9caff56058e742b191f42744cb1e34bc : SUCCESS in 58m 01s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/46981954a2244b7892a3913a87c3963d : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b17b51aea5860a3e4fd839088360ec09ae913df7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-21 16:01:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cccfb00ce5ce5c5cfbdde7fbc0e473ac03601916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:21:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49eece32d9ad46e3b205980fb6f22732 : SUCCESS in 16m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03392473a6034343aef8cdaead43fbad : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58380b80951a413b836bdd53d40b534f : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/992e6684025547e1ae2a8c133c136e20 : SUCCESS in 15m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d46be9d08cd4244be113616b074e0d7 : SUCCESS in 14m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72c693d56f824d4ebdbdcf5ee94853ff : FAILURE in 15m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f5f6429d7f411c87ac3d875ebe6428 : SUCCESS in 14m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ad23393b95746009659af2a6610548c : SUCCESS in 5m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dac42f246c6344228e17c3fe8f1cf769 : SUCCESS in 1h 27m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1df86e673c144bb4a0e8a257f53e35f2 : SUCCESS in 58m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/ccd06a300b9141caabe270818f5b1dd7 : SUCCESS in 1h 03m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84fe86511b844dfabec171a0a126a963 : SUCCESS in 1h 12m 08s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/c13ec743aa1a4c61af206cf6f1b5926b : SUCCESS in 37m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/76723eecd1464761b7a6f6a854249c82 : SUCCESS in 38m 31s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3cdb8d7db7a0485b8e0052107668409c : FAILURE in 13m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2ccf9023c43d4615bc1b86312cce824f : FAILURE in 8m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/831e571e45614474b5e69d20f1884fbf : FAILURE in 8m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3161844f2f744937996471be0967bede : SUCCESS in 15m 55s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/01ed0db22b0f4236b77c8fa71d4cf4bb : SUCCESS in 36m 06s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/48f9059617a84912b6b5eb52987f1c0b : SUCCESS in 59m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc27ba8cdde54d46b76348fef26ed256 : SUCCESS in 54m 34s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8b7029326b4d46399334e5f96a7465a7 : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f7d301a03951e098b4f0535658dd017ee87b3d4e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-21 21:22:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"462394f52530111dc0cb14649ca80464a2382afc","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-21 21:39:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93166554bc263c490086206d8c8770c5b32d5875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 23:07:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4634085a51745abba4bf247c0c86afc : SUCCESS in 16m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e71aa2466218434ca4299097ee29d8dd : SUCCESS in 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e472c6e98174d0091dad117566fdb67 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f89d95cbfa466dac17c425bde216cd : SUCCESS in 14m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc13223e422b46a3ac743c0fc3b544db : SUCCESS in 13m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/289a7fbbd7ef4c359a46c9b7030fab18 : SUCCESS in 14m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348af19ee3e24e2aa8b844be122a055a : SUCCESS in 14m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6d78731493d4d758361482c5efcbfef : SUCCESS in 4m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eb6213efc61d45f8be4a5a7a405d2aab : SUCCESS in 1h 31m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d16869ed109641ee8d546aecdb97b93d : SUCCESS in 1h 01m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/a1b16a11ecbf41c9a3537582a06434b6 : SUCCESS in 1h 02m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46c436cc533e45ffab3859d7a6be5419 : SUCCESS in 1h 27m 52s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/78882f899e0244aaa40b7c2284500da0 : SUCCESS in 31m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bc8d7f09008a422480320536f5b531c5 : SUCCESS in 30m 28s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c886fbbc3c834f999b5f27ca6ef97a70 : FAILURE in 8m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/95be9c27dff947d5a3c262ad44126392 : FAILURE in 10m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d1d51f42db9f46638fdd373b7b9d42b0 : FAILURE in 8m 25s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b319ba19949a4497acd12acf4415a260 : SUCCESS in 17m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bf1d13f9c50345ce949dbfd3c497a29c : SUCCESS in 30m 59s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2987f557f8f34a84934e3e492b1ec0fc : SUCCESS in 1h 00m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68a07392063b4313b95e2b364d8e09a6 : SUCCESS in 1h 03m 24s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/da163d29b2a545088d082bbbc3defdce : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":1},{"id":"52627564e58650ca23a307efa74d6f347505b859","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-22 01:29:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be279d33a8f71c029fc523222851afc0b444ac34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 03:09:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1060fffc894d2bb133e4f622e11fd4 : SUCCESS in 19m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e3f445ddd4743798660a7c2477bba49 : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e3ea87114142b9b67b41034d98e89f : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39a6045892354a628e9869d1f1fb2733 : SUCCESS in 16m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5756b0dc77241e18fdaefea1021aeb4 : SUCCESS in 13m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76dead94138744439b27998700628725 : TIMED_OUT in 42m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c695caf8e0e411b9a6e7c14ab62b800 : SUCCESS in 14m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f790706317c943338aa57cc8b6334f20 : SUCCESS in 4m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/33edba025ffd433792bc496f415a67d9 : SUCCESS in 1h 33m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5f87b420158741938d4087cf83ac4936 : SUCCESS in 1h 00m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/c1d8f6b36fd144328c388e3a798f399d : SUCCESS in 1h 10m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/244aa929242e4b308e6c08dc4c650a1c : SUCCESS in 1h 18m 21s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/05014fde22264cb1a96d60697bf7366e : SUCCESS in 36m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b9f733b2d9bc43af8285318fdd8feb38 : SUCCESS in 41m 13s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d9c8977d8df849eead16943b553882cb : FAILURE in 11m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/182b603b354842e691f84961e4369c84 : FAILURE in 13m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e53b3ceffe544450858e7397945eaa56 : FAILURE in 12m 57s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8089e7c5b24c4b45b07e99c1fc32beef : SUCCESS in 20m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ab944fd1b69d4e8699c36983449b3478 : SUCCESS in 43m 07s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2459d5e090b94d22abc04243bce305d0 : SUCCESS in 1h 05m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87377fcfcb374204a7f34850ca779333 : SUCCESS in 58m 05s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7ba5460bf25d4e7c81fdd57a577d2289 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d1fd43401688bb5edcda65c5649b9a43ee9d7a8a","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-22 08:42:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54734fea68d5a74cf3e10d4a1b75a9d3b5bf3345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:24:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afe7942b151f4aacbcafc0a22657ecf4 : SUCCESS in 14m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/462e703e4b3f45589b226c4d4130b278 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4ea758de7947adb4f12ebad355cbb1 : SUCCESS in 9m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afaa7c4327224db1aff40183427fe873 : SUCCESS in 14m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e06fd7e2ea4315bd3ae7ba23c6da5e : SUCCESS in 13m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07b7f64d67594f06a865fe14f9927d4c : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c337665d34e4c9ab89c772356ffc8d0 : SUCCESS in 21m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/43e48f89d0df43128e35d20473d7baab : SUCCESS in 4m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e5c57c5147a64c09983dc4352c05b34a : SUCCESS in 1h 32m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b2002d663a8b4d728981477af4c90d8a : SUCCESS in 1h 04m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/11949274d2ce4ee9ac36d6f08f438b17 : SUCCESS in 1h 05m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7068405580041bf8fc78551a8f5cd25 : SUCCESS in 1h 17m 31s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebbd001ef42548ad9900517662cea3e7 : SUCCESS in 35m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/71c67f64ead3460f893ecdcf0a51dcef : SUCCESS in 31m 48s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9737e6788f6d4244a03609940d0ba59c : FAILURE in 7m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3f5d758390154a49b2be32d1e26340f6 : FAILURE in 8m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/56b610b68722468a906fc006bf1cecd1 : FAILURE in 8m 39s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1cb4c4a023a241b8b8d84e1237c26780 : SUCCESS in 14m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/14a7d38444064cc2873618c936f7e69a : SUCCESS in 40m 51s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db83384a32c5432a8e45fc71c8477227 : SUCCESS in 1h 11m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71a3dc6215c14594b157eb999a117944 : SUCCESS in 1h 03m 56s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ef9399f42f974c6c93dc91fa4160b9f3 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b665da561fdae57fa37de00159f01f3a9654a33c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-22 15:25:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbc4ff7fd66e3c44f85999dd8bf607066702384b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 18:44:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdeeb22aaea04d1298c597a10e551ad3 : SUCCESS in 17m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fc374a2b7f843118793db15650a9cc5 : SUCCESS in 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82df5bb12a36453b9f47f36b214a1c00 : SUCCESS in 7m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfed39f32e4e4c11be9dd7e1abd2ae88 : SUCCESS in 14m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0713a661788f44e7a848f8b5833d8b10 : SUCCESS in 17m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1aa5b1de431c4859a299e57cebf7f471 : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270a326ebfe74090a7e20fb06154f803 : SUCCESS in 14m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a9929b4ebe945a782bf92eed20c297a : SUCCESS in 4m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/072e3dff4d95455093dd53c1a29a90a1 : SUCCESS in 1h 31m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/734d2f345c4a46759f44e1bdb77aa0f9 : SUCCESS in 55m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/dd1272cfed1f4c39bc9210818360a33d : SUCCESS in 52m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c01626eee8f42d2a3ec022e1a674fdf : SUCCESS in 1h 16m 53s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/21f033712caa44f39e28f00dba2279c9 : SUCCESS in 35m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/765bf3b2970041d3bc73081dcab282b5 : SUCCESS in 31m 37s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9d887edb61fb48faad88a4893733f20d : FAILURE in 12m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ed61db46a0724228aebf542bcb4df695 : FAILURE in 13m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/4ecede0ef40a429fa6a5502466dbc414 : FAILURE in 7m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/661ea995466e48a498d53aef1907db84 : SUCCESS in 23m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f96c34032f314a29bfc90eb15daf39e5 : SUCCESS in 40m 33s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/87c37319f53f44e083ceaf1006ae1a42 : SUCCESS in 55m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c07fe35141ea44e9aeac9c1912af30bf : SUCCESS in 1h 00m 33s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/632906d01af14754a8f6edfb600c03bb : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ebdd20c5a2e33e7706aae3f7c98bf6fda8b3c034","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-23 07:33:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95994387b03fbc08c617081c11d62bd3d318510b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 10:28:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc4a7e1b268d485f88e051530d069292 : SUCCESS in 16m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f395e0b8892e4501984315d7ff271c3f : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5548dd1c56ed4d2baa179404729e23c3 : SUCCESS in 7m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da5eaded83b347a7b614d30eafeb441e : SUCCESS in 27m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/038d315324c94e7da30b5929e025bc1b : TIMED_OUT in 42m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2de012d18fd24043ad4a4e85c2e60562 : SUCCESS in 28m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf22c6dda58e4632976f3a09d7f688dd : SUCCESS in 13m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a29fafb4dcc4ea7ad83dfbb53593704 : SUCCESS in 4m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a9b4bee590a423d845f6f9d8dae7f81 : SUCCESS in 1h 42m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aed3adc8a5df49bba859f4234c11cb2a : SUCCESS in 1h 06m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/dde43e0e2bc24d7d8ab7b75c0aba77cb : SUCCESS in 1h 00m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc4dcf4256b641c89e348212c01fb444 : SUCCESS in 1h 23m 18s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/fed7c893f0e04820b0177596073293a8 : SUCCESS in 33m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4db8cb78f25d42e2ad79d56809dae186 : SUCCESS in 37m 00s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2ea1e56057d94cd69f5f0ed2a8103e93 : RETRY_LIMIT in 4m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3bd5ff1868e047978eaf85d034880fc8 : RETRY_LIMIT in 4m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/3886c95af18740d0918a53dbf7137d66 : RETRY_LIMIT in 4m 12s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3082e7fa0c40490285f938c09d2e6c66 : RETRY_LIMIT in 3m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9f05190548a54f6282c5a1fa97545122 : SUCCESS in 39m 41s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9e84b23599814f3aa3b25090ea50883b : SUCCESS in 1h 04m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed1612f4f48849308be1ae9f8e0347c9 : SUCCESS in 1h 05m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/c9e217db58fe4e7496e28847fb635c0b : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5e09f8f10b886d5870ead69b620e0df6c8ab912c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-23 10:41:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"203e052d6ddf74dc0c1576b5790630874b84d61c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:29:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff929befd0b4de1a27c02c445eaacba : SUCCESS in 16m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac0955d46d0142619524a591be1a416d : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15a290b28f4427daa3912d681e6a38b : SUCCESS in 7m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d2fdd13f1d340708b47a9cea6647d8f : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d5cd389c36b4b58986e4c5948527728 : SUCCESS in 14m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91bc3e263f1642419fa9a32c739fd543 : SUCCESS in 14m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/701b95ccd0584b459e278f843d966ac3 : SUCCESS in 13m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0293932a5944f1b9282dfc39a0af57f : SUCCESS in 4m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dceb4a7dc5eb4f068cdaab72e9f49f6b : SUCCESS in 1h 37m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/12d3245fa2c840b3ac59e87d739a61ef : SUCCESS in 52m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e6cfdb92341c4e87a519fc6149e38577 : SUCCESS in 1h 07m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ab57a0da7a84b3ebaa8464288f19797 : SUCCESS in 1h 20m 43s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/409ec4ad4d4e4593b61612cc9b1254bb : SUCCESS in 38m 28s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c1f2f02aed3d44b8a3e830706f21a8d6 : SUCCESS in 36m 30s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/847143464b5141e7a8c17a86291a43d6 : FAILURE in 13m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/85c74852979041d98611c47cc4954cf3 : FAILURE in 14m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/bc5c636376354dc1ae83ee3c3403177d : FAILURE in 10m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9ba98d93a0754f32a74c81ba0a99b05e : SUCCESS in 20m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/17b69e6ee3f0480a8464fcacba46c501 : SUCCESS in 46m 03s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79bd31e02cef4ddf944c8ddb16f1c83e : SUCCESS in 1h 01m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af9245f427214f889db1ad46fb479479 : SUCCESS in 1h 01m 03s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/6ac0ea2616254cce9a1ccfeaa556c449 : SUCCESS in 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b76caf60c20e126af7330088ce0294fd93d81565","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-23 16:20:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"641493cb80be837c4f604f83d8b9124979685fc4","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-10-23 17:53:25.000000000","message":"Patch Set 1:\n\nPlease don\u0027t do anymore rechecks till this one is landed.\n\nhttps://review.opendev.org/#/c/755406/","accounts_in_message":[],"_revision_number":1},{"id":"f05e6e662cd4c5453a002e12a2b9fa0cdea0ba71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 20:06:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d370ea309544897a4d0074990b3bde2 : SUCCESS in 21m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7c4d4f1b7d446558bcc12448bb630ee : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22091953755b49bbacd97a25b182c0d7 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66fb3f97ffc64de0b3eb9049d608eacc : SUCCESS in 14m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/471de51d0622433db297af33373d7e3b : SUCCESS in 14m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/709274f537d8474f9fcda472a41acb50 : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c3b8134bee41048b35e882b8567530 : SUCCESS in 15m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c04e493d85b54bd287bd7683aa9cbd83 : SUCCESS in 14m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f4967d4f5dbb4f309a5e1a40bfa2b78c : SUCCESS in 1h 33m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/217ab77f111a48dfb6b5c1d71121da28 : SUCCESS in 1h 03m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/c74625d2362245d0b6ac731b1a88e9ef : SUCCESS in 1h 01m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/341efbc3327d4f678fcf9b15d1cd36fa : SUCCESS in 1h 27m 48s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/581bb34f69f54db19868527c25b4a16d : SUCCESS in 34m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/51262a7c83f8499a81ed1bc2cae5d847 : SUCCESS in 37m 52s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/25ba60a6882c432c856de4239c1a37e8 : FAILURE in 13m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f377e9e4a50e4d9aa72ea0deaad79795 : FAILURE in 14m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/28511acc00c6445f84bd61ff0197c53a : FAILURE in 14m 22s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c145c6e85de546bba497927eec231399 : SUCCESS in 38m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/00aab435d41d4e3a9bcb894b0a5a86a3 : SUCCESS in 39m 54s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/78baf8c1214443059f09059cd9356902 : SUCCESS in 1h 07m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/209f003f014a450ca49dbc0d8cec454d : SUCCESS in 1h 09m 34s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/e3893026624d4d74947ee2de190d155b : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3dbcff1f6581160014b976a621420f61f633bdde","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-23 20:38:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"35be1d20086ab6c175c91e676b98891ab40c6f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 22:56:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe769a3f094415492fcb6d4f9e175c5 : TIMED_OUT in 52m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bea78b09aaa344a3b86b2ebfeebece6d : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07831e1f50d54aa08a66d0a9680cdc7b : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a9b85bd8d0e4cecb3379b32b5ba90d9 : SUCCESS in 15m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/792f8d16a5124e2ead54b9927d469673 : SUCCESS in 15m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8566418c343c4cb9b271b80f9084a8a2 : SUCCESS in 16m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528c62d7f13042fa95bfde2b0dd0d8eb : SUCCESS in 13m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/29d8297d2de142f684af21e8908d1394 : SUCCESS in 5m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0190fc10f47419cac6c72d743562d3f : SUCCESS in 1h 29m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe9f3cfb97914f1fb6398d81dcf25853 : SUCCESS in 2h 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/9c077f77d7ac4675ae677f395a52efd7 : SUCCESS in 1h 01m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b162da97f2e445bb89f494ba87f93043 : SUCCESS in 1h 19m 35s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/abb60ad9ec734722ba0ef245178f7ed5 : TIMED_OUT in 1h 13m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/24fb9c576df549eb808e8c6afb713e15 : SUCCESS in 31m 49s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0fb1ad1e0ab14f1a9e47ef808a722909 : FAILURE in 49m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ddb74a4d52ba4d588f18601d0ffe84f9 : FAILURE in 14m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/df37dfc33ed64fef818c909f45352029 : FAILURE in 49m 05s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/29d8fab0fcd24f1988e2394563937e44 : SUCCESS in 17m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0a083fd61e384a04bd02c32690875135 : SUCCESS in 1h 48m 16s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5ae61ee640fa4ee1994c42b7adac4ed0 : SUCCESS in 53m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7328127fe7847378eb77f638fd0c864 : SUCCESS in 58m 58s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/c837f3e6a7f145fbbf47b181dac82102 : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e6a93effe727f66bfe31d356b1c2ec547d2590f3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-02 20:23:59.000000000","message":"Patch Set 2:\n\n\u003e Please don\u0027t do anymore rechecks till this one is landed.\n \u003e \n \u003e https://review.opendev.org/#/c/755406/\n\nAre we good now that https://review.opendev.org/#/c/759750/ merged?","accounts_in_message":[],"_revision_number":2},{"id":"fc30105cb5d613856ceab713733aa5fe9031030f","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-11-03 00:17:58.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/#/c/759750/ already fixed the amqp lower constraint. I rebased the other fix to skip this one. I think this patch can be abandoned.","accounts_in_message":[],"_revision_number":2},{"id":"dea716521b3cf8949aa1b5ecb2a5a4aad2b8b124","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-03 01:17:53.000000000","message":"Patch Set 2:\n\nI rolled this into https://review.opendev.org/759750 because the gate was broken without both patches","accounts_in_message":[],"_revision_number":2},{"id":"00986742e034b67088a7c1a98e17aa7ffb639e78","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-03 16:27:12.000000000","message":"Abandoned\n\nThanks for the clarification. Abandoning.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d435a915160a9876b58f2a04696d9430790e5cdc","revisions":{"b7c3458b6f11ade0ce54889ae5f782fbac4a9a2e":{"kind":"REWORK","_number":1,"created":"2020-09-09 18:54:47.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/750788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/750788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/750788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/750788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/750788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/750788/1"}}},"commit":{"parents":[{"commit":"fb7d54543fd69e046a5136ca4028f4e128b947c2","subject":"Fix lower-constraint for PyMySQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fb7d54543fd69e046a5136ca4028f4e128b947c2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-09-09 18:48:30.000000000","tz":-300},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-09-09 18:54:39.000000000","tz":-300},"subject":"Update amqp lower constraint","message":"Update amqp lower constraint\n\nNewer versions of kombu rely on updated versions of amqp [0]. Keystone\u0027s\nlower-constraints job is failing consistently across multiple branches\nbecause it\u0027s referencing amqp 2.2.2 when we should be relying on a newer\nversion (5.0.0+).\n\nYou can verify this by using `pip3 check` with the lower-constraints tox\nenvironment.\n\nWithout this patch, you\u0027ll likely see the test_notifications tests fail\nwith something like:\n\n  ModuleNotFoundError: No module named \u0027vine.five\u0027\n\nWhich spirals into subsequent failures mocking messaging drivers.\n\n[0] https://github.com/celery/celery/issues/3547\n\nChange-Id: Icad9383c8bc963843ef36fc24d745f29fc8798ed\n(cherry picked from commit c495defc2bd4f2ef5a38d2a9e4d5aac5653fb406)\n(cherry picked from commit 12891a039fb1dae7ba22884cf78ad7d1faa8569d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7c3458b6f11ade0ce54889ae5f782fbac4a9a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7c3458b6f11ade0ce54889ae5f782fbac4a9a2e"}]},"branch":"refs/heads/stable/train"},"d435a915160a9876b58f2a04696d9430790e5cdc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-23 20:38:00.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/88/750788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/750788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/750788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/750788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/750788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/750788/2"}}},"commit":{"parents":[{"commit":"fb7d54543fd69e046a5136ca4028f4e128b947c2","subject":"Fix lower-constraint for PyMySQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fb7d54543fd69e046a5136ca4028f4e128b947c2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-09-09 18:48:30.000000000","tz":-300},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-23 20:37:41.000000000","tz":0},"subject":"Update amqp lower constraint","message":"Update amqp lower constraint\n\nNewer versions of kombu rely on updated versions of amqp [0]. Keystone\u0027s\nlower-constraints job is failing consistently across multiple branches\nbecause it\u0027s referencing amqp 2.2.2 when we should be relying on a newer\nversion (5.0.0+).\n\nYou can verify this by using `pip3 check` with the lower-constraints tox\nenvironment.\n\nWithout this patch, you\u0027ll likely see the test_notifications tests fail\nwith something like:\n\n  ModuleNotFoundError: No module named \u0027vine.five\u0027\n\nWhich spirals into subsequent failures mocking messaging drivers.\n\n[0] https://github.com/celery/celery/issues/3547\n\nChange-Id: Icad9383c8bc963843ef36fc24d745f29fc8798ed\nDepends-On: https://review.opendev.org/755406\n(cherry picked from commit c495defc2bd4f2ef5a38d2a9e4d5aac5653fb406)\n(cherry picked from commit 12891a039fb1dae7ba22884cf78ad7d1faa8569d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d435a915160a9876b58f2a04696d9430790e5cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d435a915160a9876b58f2a04696d9430790e5cdc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
