)]}'
{"id":"openstack%2Fkeystone~750789","triplet_id":"openstack%2Fkeystone~stable%2Fstein~Icad9383c8bc963843ef36fc24d745f29fc8798ed","project":"openstack/keystone","branch":"stable/stein","hashtags":[],"change_id":"Icad9383c8bc963843ef36fc24d745f29fc8798ed","subject":"Update amqp lower constraint","status":"ABANDONED","created":"2020-09-09 18:55:13.000000000","updated":"2020-10-05 13:41:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8cf02a2cc78f453e516f99ebc7cacf16fb9f0639","_number":750789,"virtual_id_number":750789,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"date":"2020-10-05 13:35:55.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-09 21:23:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-09-09 18:55:13.000000000","_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 18:55:13.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-09-09 21:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f942ddff3d9ce26f611712fed15061005324bfbf","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-09-09 18:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f3c2f41d1d11c2fb01f8faed212054cd461bf86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 21:23:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/858112938d2a4c6cad9a3e0fdbe297b7 : SUCCESS in 31m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c820cdf3711d4c96890ac9a62e22e9a0 : FAILURE in 34m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc416fdf8ab4204b1eeb2d08a0b61e0 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73ebb457e2e44268bdf58f5eb754eda6 : SUCCESS in 20m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2c52a9540754b27b8cd1a61c48660b7 : SUCCESS in 31m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f33375e3c2ba4e86ba31078f07d852df : SUCCESS in 30m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b156d8d0423d42d48bbda6fd7cf1e61f : SUCCESS in 11m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3315be4e24644408a8993cacf5885ca : SUCCESS in 2m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1be49004c2f43bd870314c69096f4d0 : SUCCESS in 1h 16m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2120696a65bf460bb5c3d8104b1225b6 : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb413cb334ec406090fcf271d16f0c89 : SUCCESS in 1h 17m 23s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/83a8080e14974f0dbad6420d11db73ef : SUCCESS in 32m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/303e52b468104ea882aebfdc9954ffdf : SUCCESS in 35m 33s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/735cfe5062fb4870928572bd00433187 : FAILURE in 12m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1f81d36214c745c3b9013291724faaf5 : FAILURE in 11m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/2183759d22484c1ab080c242ef6a1546 : FAILURE in 11m 55s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a42fb1eeb0d2403099ef36fced3e3197 : SUCCESS in 24m 52s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e8b977dbdd1f4f76afa4f51ff840c2e3 : SUCCESS in 40m 14s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6214595fbc6644479f6d01dbe786c2a8 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"08fc0e187f44cf18fa904f0dfb3297de8c64eeee","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-05 13:35:55.000000000","message":"Patch Set 1:\n\nThis patch is failing becuase the PyMySQL version is 0.10.0, which is fixed in the next patch. Consolidating this into a single patch since we have a circular reference.","accounts_in_message":[],"_revision_number":1},{"id":"8cf02a2cc78f453e516f99ebc7cacf16fb9f0639","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-05 13:41:35.000000000","message":"Abandoned\n\nRolled this into https://review.opendev.org/#/c/748474/3","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"255b6cdd70db654ce8112cef0b86361c34d9993b","revisions":{"255b6cdd70db654ce8112cef0b86361c34d9993b":{"kind":"REWORK","_number":1,"created":"2020-09-09 18:55:13.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/89/750789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/89/750789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/750789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/750789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/750789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/89/750789/1"}}},"commit":{"parents":[{"commit":"353a656dc40dd3bb4a33b3cc4b6d33c1d1dfa7f7","subject":"Merge \"Refactor some ldap code to implement TODOs\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/353a656dc40dd3bb4a33b3cc4b6d33c1d1dfa7f7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-09-09 18:48:30.000000000","tz":-300},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-09-09 18:55:06.000000000","tz":-300},"subject":"Update amqp lower constraint","message":"Update amqp lower constraint\n\nNewer versions of kombu rely on updated versions of amqp [0]. Keystone\u0027s\nlower-constraints job is failing consistently across multiple branches\nbecause it\u0027s referencing amqp 2.2.2 when we should be relying on a newer\nversion (5.0.0+).\n\nYou can verify this by using `pip3 check` with the lower-constraints tox\nenvironment.\n\nWithout this patch, you\u0027ll likely see the test_notifications tests fail\nwith something like:\n\n  ModuleNotFoundError: No module named \u0027vine.five\u0027\n\nWhich spirals into subsequent failures mocking messaging drivers.\n\n[0] https://github.com/celery/celery/issues/3547\n\nChange-Id: Icad9383c8bc963843ef36fc24d745f29fc8798ed\n(cherry picked from commit c495defc2bd4f2ef5a38d2a9e4d5aac5653fb406)\n(cherry picked from commit 12891a039fb1dae7ba22884cf78ad7d1faa8569d)\n(cherry picked from commit b7c3458b6f11ade0ce54889ae5f782fbac4a9a2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/255b6cdd70db654ce8112cef0b86361c34d9993b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/255b6cdd70db654ce8112cef0b86361c34d9993b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
