)]}'
{"id":"openstack%2Fkeystone~760614","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I4df19c45c870ff3fb78578ca1fb7dd0d35da3c82","project":"openstack/keystone","branch":"stable/ussuri","topic":"bug/1878938","hashtags":[],"change_id":"I4df19c45c870ff3fb78578ca1fb7dd0d35da3c82","subject":"Delete system role assignments from system_assignment table","status":"MERGED","created":"2020-10-30 16:03:14.000000000","updated":"2021-03-11 05:56:50.000000000","submitted":"2021-03-11 05:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760614-bug/1878938","meta_rev_id":"3f78dcce9810c2d223450fda935e89d29cedbaf3","_number":760614,"virtual_id_number":760614,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 05:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"recommended":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2021-01-26 16:01:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2021-01-26 17:49:45.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"date":"2020-10-30 16:03:14.000000000","_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":1,"date":"2021-01-26 17:49:45.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-10-30 16:03:14.000000000","_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 16:03:14.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-10-30 16:03:14.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-10-30 23:59:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 18:52:59.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2021-01-22 18:52:59.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"282e7b184f6810a5625178beae2501c5539f46df","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-30 16:03:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e75db1cbf12e3e68e515a0f278edbf7e634e1767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 23:59:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cc02558472842ef973b28b57e701abe : SUCCESS in 29m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d627ac5f2ca54a50889f2fce30847519 : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd6ba9a8b1240dcab90de8cdd5d76e0 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95cef943277d4fac96d75089af986c9b : SUCCESS in 12m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38e3210d32da43fea1ef6e7d8ef0a652 : SUCCESS in 17m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b72582da5af4b6a92f1dea8b5778719 : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/afdf9ec492ea47fe85bd8bd371c206b3 : SUCCESS in 1h 09m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a179968a78614f5c91303a1f1086aac9 : SUCCESS in 1h 08m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8435c65a234b406f9337a8e7892f1816 : SUCCESS in 7m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/40135798cf864b93872508fbbd7c0f39 : SUCCESS in 33m 45s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/cb99eaa99d8448878021a52ce40604c0 : SUCCESS in 32m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/79ee9c885f074df4b5749fb488c86239 : SUCCESS in 35m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ec3bbc05de1b4fca8f8d9ad7d031d9d1 : FAILURE in 17m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fb29d82131f44b78a5f1ccd6cc5483f7 : SUCCESS in 46m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c340e0cb2054aee8ce29170c4666e96 : SUCCESS in 57m 39s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/dfda9594b00f4753a42a1f8d6dfe8041 : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2df5f1fa9e10aaedfa7412da69a01185fb03e73e","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-01-26 16:01:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba3b1ae41ab7f6b7a6d98e8a0a5e3d47e76231a8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2021-01-26 17:49:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b04833347f70866036ea5e4a75b6294cb8efee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:50:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00c4360549683d47d5253fd2a1553a4dd12b61a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 22:27:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/498d2ed14813498b8d3480b0f8faa817 : TIMED_OUT in 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1352c205720b4607a907963d8ec46271 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/014ca561ffb043e089211c1674ccde4d : SUCCESS in 15m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45cb00405c7949019391064a51bf1894 : SUCCESS in 23m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c337cafd454778bf0ac8836dedb090 : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/1e4ebc02138c4b5e9121451228161faf : FAILURE in 10m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f599c5ca4644e2ab2119e479d9fe798 : SUCCESS in 1h 19m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a295782245347aeac293d590baa3482 : SUCCESS in 6m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3674d164da39411eade8abcb003f4f6e : SUCCESS in 25m 50s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/3faeeb6000b247f1a68a84714dafe5f4 : FAILURE in 6m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbc9958448f14e95993cc3bb11eaa6cd : SUCCESS in 50m 52s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/4fdd76b4695a430dbb80b067ed72495c : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5273e2defd87e5d1015e2dbd71c95275eee87473","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-27 13:04:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"988020d6266594daf3b7ede62c8a813e6296fc38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fa6a60bbfba4cb5bb05d39ca4e27635 : SUCCESS in 21m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d95880b96113499eace804084fead90f : TIMED_OUT in 41m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6622399fcfc47e987edfd2048a672ad : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/788abcd1e7c14c66a2e554e7b1b836bf : TIMED_OUT in 41m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6a671b590624eb388d4e7128856b421 : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb751bab07a48d3bde80089637dd4d1 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c8b32ba5985949879ef795211fbb72ac : SUCCESS in 1h 11m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe4e5d6dc08d46349bc6a6088190b33d : SUCCESS in 1h 20m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b6af534df6d411f960cfaf0d2c25fe4 : SUCCESS in 7m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8973412cd46a45bf84d5cc74b48c1681 : SUCCESS in 31m 24s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/6b0ba1e5884242a790b516d2ac5548e4 : FAILURE in 13m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/15fc6d21c6024b4e8aacd17481448908 : FAILURE in 13m 41s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fef9707299144719b00ab9dbfa07c4e5 : FAILURE in 16m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5752e3a1ca7649579cfa01ae043f9c48 : SUCCESS in 38m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcb2bbf9fd37475c86bbe3de00c88746 : SUCCESS in 53m 39s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/befb56e98d7b4db2ab74cc998f5378dd : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3c746db15aeecdf76a416201e4d06b84cbf7d7c7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2021-03-10 15:41:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce3a68cb868ff803dcdfe9ab51d3a37db5ba3dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 22:34:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da46c1f919f54fd18c347c13afce1deb : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1186600d0046b28244e2e2fa03bbf7 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44db9b3d0cfb4b4cae811dfe8ed7f7b1 : SUCCESS in 16m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca8cadf367d84ff28597999bcb996061 : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fbd65b79bcc47d8965b710bcaedb21d : SUCCESS in 15m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/5bbd2629e37d45f7a1f1e8b00ef76d40 : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ff0bb10bac141efa3689fa08a91389a : SUCCESS in 1h 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/874eabb3aa2c4756b1d13ddd152c4424 : SUCCESS in 6m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7a3d3232aa52414a90c26a1a493928e0 : SUCCESS in 32m 00s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5a312e2471e04bb78dd4636dc9baff65 : SUCCESS in 37m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/0d0baf7d315749bf9395ad6f50813b9b : SUCCESS in 37m 55s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/317bb7c97c2242fea436850ec661994c : FAILURE in 17m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9412a58167a948869f7c48ac6d96c788 : SUCCESS in 35m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1867b71b0b4644cca569c3ae4912fb64 : SUCCESS in 52m 01s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/2caef33d10554184ba201ab15d95dbe2 : SUCCESS in 46m 31s","accounts_in_message":[],"_revision_number":1},{"id":"26a1936f2390186bb80c5368076258abb8dc1c57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 22:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e49bc313110fcfa169d6c31a251271431867bee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 05:54:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8f7c4ff4674e5f8e7c9a4797c9c192 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/586a56d1b5a74f94afcd703b1405d2c5 : SUCCESS in 14m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05ee2cf7996541e09aff5d78f85eb2f5 : SUCCESS in 16m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7c97e4a79442afa3140e9ca0388e35 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a6f86866af1f4a859a237c48fca0ee67 : SUCCESS in 1h 08m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c01ba93efdf430aa6d8d2b972d7aca6 : SUCCESS in 1h 02m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb251d18cdb54db3803eeeb79f70eb5b : SUCCESS in 7m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b6fc5e2c31dc4304936a397f945a51a4 : SUCCESS in 32m 24s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/acaeeddc2ab948688b07ed12e732caee : SUCCESS in 32m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7de71eff40fd4779be9a75be516933b0 : SUCCESS in 49m 56s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/204459ce2bf3431084245f47db9a5681 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d702551d7371a9fa029b3e21d1a31cabb4e984a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 05:54:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f78dcce9810c2d223450fda935e89d29cedbaf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 05:56:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b55418d7baf4e0db788a4fcb597ea44 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a42ec2641483433cbc5da88876823f75 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ac0891375656d156f1a9601e3520052dddb39d0","revisions":{"7ac0891375656d156f1a9601e3520052dddb39d0":{"kind":"REWORK","_number":1,"created":"2020-10-30 16:03:14.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/760614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/14/760614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/760614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/760614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/760614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/14/760614/1"}}},"commit":{"parents":[{"commit":"5a7ebf53bc797524ea90400bbea5bcd75f82ddd9","subject":"Merge \"Implement more robust connection handling for asynchronous LDAP calls\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a7ebf53bc797524ea90400bbea5bcd75f82ddd9"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-27 06:38:41.000000000","tz":330},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-10-30 16:02:51.000000000","tz":0},"subject":"Delete system role assignments from system_assignment table","message":"Delete system role assignments from system_assignment table\n\nThis patch ensures to delete the system role assignments from\nall the assignment tables in keystone after deleting the role\nuser has over the system.\n\nThis also make sure of deleting stale role assignments before\ndeleting role for the deployments that are already in this state.\n\nCloses-Bug: #1878938\n\nChange-Id: I4df19c45c870ff3fb78578ca1fb7dd0d35da3c82\n(cherry picked from commit c1dcbb05b4488f1fa3e7af4d9171d11702d94119)\n(cherry picked from commit b83170a386ba8da2195c7494d04d832ce9b6d7b0)\n(cherry picked from commit 6f93063ff95f3c65af106a09281427e411d01850)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ac0891375656d156f1a9601e3520052dddb39d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ac0891375656d156f1a9601e3520052dddb39d0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
