)]}'
{"id":"openstack%2Fkeystone~764453","triplet_id":"openstack%2Fkeystone~master~I960379ceb435472cdc754b5f63243c70d552d9c3","project":"openstack/keystone","branch":"master","topic":"models_revoke_model_e225","hashtags":[],"change_id":"I960379ceb435472cdc754b5f63243c70d552d9c3","subject":"fix E225 missing whitespace around operator","status":"MERGED","created":"2020-11-27 12:21:45.000000000","updated":"2021-02-28 04:55:45.000000000","submitted":"2021-02-28 04:53:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764453-models_revoke_model_e225","meta_rev_id":"c48af65e48cb1167ea089d87cabf663e5d2ef101","_number":764453,"virtual_id_number":764453,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-28 04:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-02-26 06:34:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-01-06 14:13:07.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2021-02-27 03:57:22.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-07 01:54:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-02-27 03:57:22.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 14:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-06 14:13:07.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-01-07 01:54:32.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-02-26 06:34:17.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-02-27 03:57:22.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"ea63cbf7122cd7f07d10cf3a6d9d06d41b99037e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-11-27 12:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c0c3a7469fd73554fd8bc964033a236b40d26c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 14:29:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c298cf02d7144e2a1520eeb1d900dab : SUCCESS in 19m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a53c2316e1d64ffcb2608f3d50467011 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50a9d5bbc7e41bf9b6504674d23e0c8 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5beb05d03bb47cf9c90bf19994a7e2d : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/406c2dd95ef8419ebba7d1749936164a : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7546b3c3b35344b5bf0c88d24c8aeb93 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/dfd474e923e04284b2db2eb79c31bf11 : SUCCESS in 1h 09m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/699d281f729f414ca6e6328e200baca2 : SUCCESS in 1h 33m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c22e01e836c240dd9862ea820473fc7b : SUCCESS in 34m 40s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fd5591b580b4ae69c809c1ce8198f24 : SUCCESS in 27m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/acf0f6843f634422a57a0503281647be : SUCCESS in 34m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1c3b2e75ba7c4aa0bd4bc4e59c6f4aa8 : SUCCESS in 15m 47s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8286f4aa69844122b24c4d1246b4dfe7 : FAILURE in 9m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7482df3adcf14a63b3fe51d915f45e30 : SUCCESS in 55m 11s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ae6d2475e5c8428280b1e4fc6c70d545 : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c9d9c4295332925fa0e5107af3300b1ea385575d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-06 14:13:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"883bd650245f5dced7206ed3526a42b6bd0a1094","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 01:54:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"204c971d826802e297647fb40ecf862b350f8849","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-26 06:34:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3b7cfd5474448cc9626cfb8eb64d3d621257783a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-02-27 03:57:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"024e488e3f5a383f9c263e172ef9b9bf4615edfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 03:57:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dcb45739d54b286a8edf31492ed5f3d4791f1d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 05:06:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f298f0a28e64b14ad879462fa7d30f1 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c1dfb861b4d41a188d27ff0c833b7ba : SUCCESS in 22m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41db789c07f4f09b015e5455e424682 : SUCCESS in 21m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66ead1c83424c86b8f32676a0666b64 : SUCCESS in 17m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/99c682caec4c4b05afd56d7be723ca13 : SUCCESS in 1h 01m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7f649a39bbd47b19cfe4890966ea74c : FAILURE in 1h 06m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/64a4d6683f514234a2b7dd60a76508a0 : SUCCESS in 27m 55s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/aba9a8c5035e44f5893e11b660793f2d : SUCCESS in 33m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42d4499e1ea44501ba9b165dab1328e3 : SUCCESS in 1h 02m 16s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/91014f072a334bc5bf85f18545747409 : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":1},{"id":"25550396387ee711104fd9abc66bd6fb0ea7bc9e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-28 02:19:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5532d7348333291ee58446e2dea440444e98b331","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 03:01:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ade07b85dfb441180e807824a2ccf6a : SUCCESS in 40m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb0678df801547bb95fc2caccf1d8135 : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c525f8ebe89e931712032dcd5bc8fa48ce433bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 03:36:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b146d25266154cb8a290279eff156f99 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1cbfb6b6bc447a954e122131fbbaf2 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab61da09f5fe49089b996fc4df96376f : SUCCESS in 29m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/972001f6ddbf46a794fa499c1564b4e2 : SUCCESS in 14m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be78d8ede30e4936bd79076ae356c5d3 : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/56f814b1aad04b9f9faa7c7f7c917f87 : SUCCESS in 58m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/557b0a56496f44e0b974033ea2b8d95c : SUCCESS in 1h 15m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a915b0f09b354a5092d9aa1a6e6dfe3a : SUCCESS in 29m 19s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/712f00ac079e4a24b8a152e842e0b82c : SUCCESS in 37m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/c6824ac0a8294d48bd27c19a468a0138 : SUCCESS in 32m 45s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/aa5ba4bc393f463cabbbc7057915b2fb : SUCCESS in 12m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4628be7b4df948e784e7c40c024c18be : FAILURE in 14m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87c918ea949c4bd4bb0338670641866a : SUCCESS in 53m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/1ab67f9428da44b9932c66b1ea9b4a61 : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f9800cc27ea765cd8e0f0be39f885df70e58843c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 03:37:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d1ff6c2b50d7d71c96fc2a0c356d9145dcca43e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 04:53:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333c9541bd324ec99b635029f71fee1e : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe4b8a7c23224318948fa7f295b56284 : SUCCESS in 18m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b03438e1d9e44183b120857e54721d5c : SUCCESS in 16m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8309053200534dc5b4c47f1b3d879649 : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/684b9f7de6d54064a6fc939f83bde2c4 : SUCCESS in 51m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18644197c7cf45f2be907e228b4cd449 : SUCCESS in 1h 15m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/aab8e43f139143b0a4654ec1b57dc445 : SUCCESS in 27m 00s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/dc4b59e633f64caaaeda2b0aca37bb24 : SUCCESS in 26m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/666368dcae50490cb774f37315ec7ec4 : SUCCESS in 49m 57s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/371d6c3396384cef8b6fe6c040c2e76e : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c04d867409a2b0ad495a3362cbc7fe0ddfe34325","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 04:53:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c48af65e48cb1167ea089d87cabf663e5d2ef101","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 04:55:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732dab51784f499e96e6e24c86a32c50 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e7369f2e4e25055121570637de31bcd23fb0cf4","revisions":{"6e7369f2e4e25055121570637de31bcd23fb0cf4":{"kind":"REWORK","_number":1,"created":"2020-11-27 12:21:45.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/53/764453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/53/764453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/764453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/764453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/764453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/53/764453/1"}}},"commit":{"parents":[{"commit":"b0b93c03986f3bb40c5a2ec31ee37c83014e197a","subject":"Merge \"Generalize release note for bug 1878938\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b0b93c03986f3bb40c5a2ec31ee37c83014e197a"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-11-27 12:04:20.000000000","tz":60},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-11-27 12:21:26.000000000","tz":60},"subject":"fix E225 missing whitespace around operator","message":"fix E225 missing whitespace around operator\n\nChange-Id: I960379ceb435472cdc754b5f63243c70d552d9c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6e7369f2e4e25055121570637de31bcd23fb0cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6e7369f2e4e25055121570637de31bcd23fb0cf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
