)]}'
{"id":"openstack%2Fkeystone~774678","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~Ifab609fc42622c7d8f350e699275b541848f1a8a","project":"openstack/keystone","branch":"stable/ussuri","topic":"switch_ussuri_ci_to_xenial","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2024-02-07 12:21:59.000000000","reason":"Change was abandoned"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-02-07 12:21:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifab609fc42622c7d8f350e699275b541848f1a8a","subject":"Switch stable/ussuri federation job to use ubuntu xenial","status":"ABANDONED","created":"2021-02-09 15:18:05.000000000","updated":"2024-02-07 12:21:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1dcaa94cc8228b103e82a6d29cd2d299be7bac3d","_number":774678,"virtual_id_number":774678,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-02 19:49:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 17:41:08.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-02-09 17:41:08.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2021-02-09 19:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-25 09:46:48.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2024-02-07 12:21:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"507cfdd56e489fa6bbe767a9f4dcb7f23edcd8ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-02-09 15:18:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc4f626ef745fe78c287b75d8e3f72ebf8d30328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 19:08:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39628d5e440c4867823c845cc8c788a6 : SUCCESS in 15m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8953faa43d744841aa8ed0c23a557e70 : TIMED_OUT in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429cf1c37701415cb9ebe5ec6c45f6a7 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57b695e250ed4b448ffedb0b39cb8f6e : SUCCESS in 13m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96fe2d21ce9f4e9cb351f531cafff47b : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a599e2472c42359832ea87a482f14c : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/2b3b2208ddf94f33b82e606d9bf6a295 : SUCCESS in 1h 07m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9ed18d94c2745aa8512a1a02c8075b2 : SUCCESS in 1h 25m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/29ba70ebdf6d4f24a6e8d36466341749 : SUCCESS in 44m 45s\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b910323898144fb8adbaefa4028193c0 : FAILURE in 9m 11s (non-voting)\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial-k2k https://zuul.opendev.org/t/openstack/build/d3adac11eb51426c8b05584a2f1f42d3 : FAILURE in 11m 25s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5b1a00f98327449f8a0ca2adbf446a09 : FAILURE in 17m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/012d6b4dccdf40619ee7a63d9d788091 : SUCCESS in 40m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24ee6678b6834968a9efab83875ee6b6 : SUCCESS in 47m 08s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a28cea49d3b14a7d9f4e8622bd782f85 : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"af2b659cd24cf25c92514e3113ad5333236f8141","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-10 13:33:36.000000000","message":"Patch Set 1:\n\nThis is failing because of the lower-constraints job. I rebased https://review.opendev.org/c/openstack/keystone/+/772082 on top of this to see if that helps.\n\nWe might need to squash these two together.","accounts_in_message":[],"_revision_number":1},{"id":"eb6822c2c7c86abc98e6af848aee32339c707c06","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-10 13:34:18.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e This is failing because of the lower-constraints job. I rebased https://review.opendev.org/c/openstack/keystone/+/772082 on top of this to see if that helps.\n\u003e \n\u003e We might need to squash these two together.\n\nBut - this is also failing federation jobs, too.","accounts_in_message":[],"_revision_number":1},{"id":"1d36aa283f3ed98e822599b9a92e7c2b17e0418d","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-02-23 14:46:24.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e This is failing because of the lower-constraints job. I rebased https://review.opendev.org/c/openstack/keystone/+/772082 on top of this to see if that helps.\n\u003e \u003e \n\u003e \u003e We might need to squash these two together.\n\u003e \n\u003e But - this is also failing federation jobs, too.\n\nYeah, looks like xenial is not supported.","accounts_in_message":[],"_revision_number":1},{"id":"d9af3db9a3b033f8a1522193604a870430bdf132","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-25 09:46:48.000000000","message":"Patch Set 1:\n\nMay I suggest to switch the job to non-voting for now? A broken ussuri is not good in itself and it prevents people from merging patches on all the other earlier branches.","accounts_in_message":[],"_revision_number":1},{"id":"047ba741cfeb567d31c041306174daa640c6f5f3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-09 21:13:12.000000000","message":"Patch Set 1:\n\nDoes this change need to move forward? Looking at \nhttps://review.opendev.org/c/openstack/keystone/+/772082/\n\nit seems the two opensuse jobs are passing:\n\nkeystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/38f7e4d9110c454c9f21874474491ce3 : SUCCESS in 45m 35s (non-voting)\nkeystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/4fb97625fa2c4ccd95412c7ca523e171 : SUCCESS in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"379432b72c683873c74b940927b7b52ac82d5b07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-21 14:44:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22aa174999b05084f650ef19b110d84bb122f83f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:06:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb4af48016094e9bb51f305ab9f38669\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76afc0620b64457e9e918752d12e48c9 : SUCCESS in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4ebbaa87414163b81ff17604d60dce : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0e9740b4dfe41688f80cacc93db5e52 : SUCCESS in 19m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1983383ad6844ed8a5fb9e81c00cca7f : SUCCESS in 14m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e04abba482440919ab2c50f83e3b9fd : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/1b7fdcbebd1b4e398af3d72dfe1dcacf : SUCCESS in 1h 10m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93eb481d8b0a4a6ba02cd56756814e32 : SUCCESS in 1h 20m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7bc1238ba32545b6a126b6c6b7c47608 : SUCCESS in 34m 06s\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/44b3ee2aa6834fd9a941d760079d5b4d : FAILURE in 9m 42s (non-voting)\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial-k2k https://zuul.opendev.org/t/openstack/build/b05512875cfc4777bba40e2855cbcaf1 : FAILURE in 9m 15s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ff148a21d7ad456d8dcca1ed650e09ed : FAILURE in 15m 40s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/04655708f23b4baa91dcd57aada96fa6 : SUCCESS in 42m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66df3156465f41dfb1788a85b2e18f43 : SUCCESS in 56m 28s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/3966900172df431ebcfca167f80e4673 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":2},{"id":"592439737de9cea1ac92c440520fec5f43a91f63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-25 15:18:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"74a39f89eda6f5ad08b2b86184b642fc2a4aa612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:57:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb0f7d8634ec46fa93e3a99fcbcab174\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f83c0dac45849a1b9d6a3cb49e9ca6f : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc685edf39f46588d41d985052a489d : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce559d4b37614edf94eb7cf916a84543 : SUCCESS in 13m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdfbb0235c2042828a21a93d75c6fa16 : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fed6c2cbffa741ed9e41ada72289b095 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/50ff4297e9584a67b511d3111046d2a6 : SUCCESS in 1h 15m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebc59f9b663f4b77959c3a151d4f043b : SUCCESS in 1h 32m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/36881457e8ea4472bdc2a6f782503ca2 : SUCCESS in 38m 20s\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d7c223f0f7ca4543a99597e4e1d64b0e : FAILURE in 8m 12s (non-voting)\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial-k2k https://zuul.opendev.org/t/openstack/build/8a9cd2acc36e48248f06af93be25885c : FAILURE in 14m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/453273f3e4f84c07b34f51bfc3214020 : FAILURE in 21m 22s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/079e052a715a42cc8e5e268cf25e341c : SUCCESS in 45m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64f34c2f29aa46da988a959a846060fa : SUCCESS in 59m 00s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/35d5d5ccbcb5476cbd07bba445c87d8e : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":3},{"id":"bd06615d73926d8bc5e0fc327845da86376d0cfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-06-02 18:06:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"017d18daa703c051fccff3fb621a00ea4cd5141a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 19:49:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c23864ec85db43ad9a6854ec24ad3ee4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd4cadaa1b364fa58bfa2b23ecd4a96e : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289a68bb4e3440b2a4f137446935d30f : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8945a1ca3c0840198d8f3b45c8469601 : SUCCESS in 13m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e03098921df42078268a6a3c62f20e7 : SUCCESS in 14m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae5d30d63b94fad8e302e0439eaca7d : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/c0ff7036a1ce406ebbcee81a96dbad1a : SUCCESS in 1h 06m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4131b2b8aa1148119c74276f682c7e62 : SUCCESS in 1h 36m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2ddef236a9b54ac883038a375bf22e5f : SUCCESS in 24m 43s\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/773900eecf8f467fb98c88b4e69ee172 : FAILURE in 5m 22s (non-voting)\n- keystone-dsvm-py35-functional-federation-ubuntu-xenial-k2k https://zuul.opendev.org/t/openstack/build/5446324fc8964701aefd374affbe4977 : FAILURE in 5m 01s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a07d4b623f3d4ac5aecd6c3afc35dcff : FAILURE in 10m 27s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c0f47598c5364cffb3da8b46e2b50a4e : SUCCESS in 32m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aac6e997bf3b4a0ab9927b9d710beae8 : SUCCESS in 46m 15s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8516897e3f4a4dd79ff191d8e724abcd : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":4},{"id":"1dcaa94cc8228b103e82a6d29cd2d299be7bac3d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 12:21:59.000000000","message":"Abandoned\n\nstable/ussuri branch of openstack/keystone transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e3badfe5a3f0cc64f6debb1b348ac2aff0ca89f2","revisions":{"8ad053a1eb0c3e1c3933137b81a13b9906add9d4":{"kind":"REWORK","_number":1,"created":"2021-02-09 15:18:05.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/78/774678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/774678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/774678/1"}}},"commit":{"parents":[{"commit":"5a7ebf53bc797524ea90400bbea5bcd75f82ddd9","subject":"Merge \"Implement more robust connection handling for asynchronous LDAP calls\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a7ebf53bc797524ea90400bbea5bcd75f82ddd9"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2021-02-09 15:14:41.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2021-02-09 15:14:41.000000000","tz":-300},"subject":"Switch stable/ussuri federation job to use ubuntu xenial","message":"Switch stable/ussuri federation job to use ubuntu xenial\n\nThis is being proposed directly for stable/ussuri, since the\nchange only affects and makes sense on the stable/ussuri branch,\nand it only touches the testing infrastructure.\n\nThe opensuse base which we are currently using these tests has\nissues.\n\nChange-Id: Ifab609fc42622c7d8f350e699275b541848f1a8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8ad053a1eb0c3e1c3933137b81a13b9906add9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8ad053a1eb0c3e1c3933137b81a13b9906add9d4"}]},"branch":"refs/heads/stable/ussuri"},"c395cf30f1bcd9d9f2b941b214cd736434c87765":{"kind":"REWORK","_number":2,"created":"2022-07-21 14:44:53.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/78/774678/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/774678/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/774678/2"}}},"commit":{"parents":[{"commit":"9477584a396ca0e3352567847afe6f862718f625","subject":"Merge \"Fix issue with LDAP backend returning bytes instead of string\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9477584a396ca0e3352567847afe6f862718f625"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2021-02-09 15:14:41.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-07-21 14:44:51.000000000","tz":-240},"subject":"Switch stable/ussuri federation job to use ubuntu xenial","message":"Switch stable/ussuri federation job to use ubuntu xenial\n\nThis is being proposed directly for stable/ussuri, since the\nchange only affects and makes sense on the stable/ussuri branch,\nand it only touches the testing infrastructure.\n\nThe opensuse base which we are currently using these tests has\nissues.\n\nChange-Id: Ifab609fc42622c7d8f350e699275b541848f1a8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c395cf30f1bcd9d9f2b941b214cd736434c87765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c395cf30f1bcd9d9f2b941b214cd736434c87765"}]},"branch":"refs/heads/stable/ussuri"},"91fc2a873c09adc9b877eb64453e1fa6454613ad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-25 15:18:32.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/78/774678/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/774678/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/774678/3"}}},"commit":{"parents":[{"commit":"5de9cfcc2ef0863611099216fd1a7a1241c9b798","subject":"Merge \"Remove the note of training-labs\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5de9cfcc2ef0863611099216fd1a7a1241c9b798"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2021-02-09 15:14:41.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-07-25 15:18:32.000000000","tz":0},"subject":"Switch stable/ussuri federation job to use ubuntu xenial","message":"Switch stable/ussuri federation job to use ubuntu xenial\n\nThis is being proposed directly for stable/ussuri, since the\nchange only affects and makes sense on the stable/ussuri branch,\nand it only touches the testing infrastructure.\n\nThe opensuse base which we are currently using these tests has\nissues.\n\nChange-Id: Ifab609fc42622c7d8f350e699275b541848f1a8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/91fc2a873c09adc9b877eb64453e1fa6454613ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/91fc2a873c09adc9b877eb64453e1fa6454613ad"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"e3badfe5a3f0cc64f6debb1b348ac2aff0ca89f2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-02 18:06:16.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/78/774678/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/774678/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/774678/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/774678/4"}}},"commit":{"parents":[{"commit":"1ab860a08e527ca9e0c82a49fbf004d415fec991","subject":"Fix host:port handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ab860a08e527ca9e0c82a49fbf004d415fec991"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2021-02-09 15:14:41.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2023-06-02 18:06:16.000000000","tz":0},"subject":"Switch stable/ussuri federation job to use ubuntu xenial","message":"Switch stable/ussuri federation job to use ubuntu xenial\n\nThis is being proposed directly for stable/ussuri, since the\nchange only affects and makes sense on the stable/ussuri branch,\nand it only touches the testing infrastructure.\n\nThe opensuse base which we are currently using these tests has\nissues.\n\nChange-Id: Ifab609fc42622c7d8f350e699275b541848f1a8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3badfe5a3f0cc64f6debb1b348ac2aff0ca89f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3badfe5a3f0cc64f6debb1b348ac2aff0ca89f2"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
