)]}'
{"id":"openstack%2Fkeystone~789417","triplet_id":"openstack%2Fkeystone~master~I4bb54aac9b950d59082a4468203a3249790839d7","project":"openstack/keystone","branch":"master","topic":"bug/1926483","attention_set":{},"removed_from_attention_set":{"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2022-07-25 23:37:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bb54aac9b950d59082a4468203a3249790839d7","subject":"Only log warnings about token length when length exceeds max_token_size","status":"MERGED","created":"2021-05-03 20:43:49.000000000","updated":"2022-07-25 23:38:56.000000000","submitted":"2022-07-25 23:37:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789417-bug/1926483","meta_rev_id":"a76b2575366141caea35a94c451481deb5b6f9fc","_number":789417,"virtual_id_number":789417,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-25 23:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-07-25 19:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2021-05-03 20:46:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2021-05-04 13:56:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2021-05-05 17:01:33.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-07-25 19:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 20:46:44.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2021-05-03 21:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-03 22:14:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 13:56:23.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2021-05-05 17:01:33.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-07-25 19:14:41.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"a9334d0059ee5910f13471fc3ff6a993931a81a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-05-03 20:43:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"163ef4adc296716ecdddda1a6e149ba0012b14e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-05-03 20:44:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee267b0f4f38fe9da2ca61b23361a6fcf3eccc63","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-05-03 20:46:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"917b56c961f83d5a3a6366e322b14789746a22ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 21:40:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e643f1c0974849af92b9d9781d6250c3 : SUCCESS in 47m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/993a9947f27643d0a3d77cc43d6c87f1 : SUCCESS in 54m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"82fd456297b4fbf962528b808515d12e03534dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 22:14:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf78c149928649fab54642c0f3208003 : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b63ec9428ab4c6984e9321b51320491 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ef894e5cdb9408b940e95aaa06a7fb7 : SUCCESS in 13m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76a2a087f27f4abdaa36800395f02e00 : SUCCESS in 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921920efb2674e77a4e98ff45edc8200 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/47c29c711e314356b8212d3bcd3659c0 : SUCCESS in 1h 12m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d26677d9b4840e6b306b3609da8a090 : SUCCESS in 1h 25m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83420f6ee00646d9b2e6c9650f5a64e1 : SUCCESS in 8m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/96d961bd93934dd4b4d5aefddab1b0ff : SUCCESS in 32m 46s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e618cf2a52f406380ac172e4178bec7 : SUCCESS in 32m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/96b660d9a9d84113a100638dd10f5235 : SUCCESS in 30m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e7e354ef603e42f187280493b081d3e4 : SUCCESS in 18m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0ba10508c50e4bda85e40c262b0421e0 : FAILURE in 15m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a77d7671baeb4548bb4ef05f9e4b01ec : SUCCESS in 56m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6bb75da3f20540e79fb420d5921e6214 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1be80eef58112e3150c9f28c555f3e1a0da4ed88","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-05-04 13:56:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e2451ce99d39c000195718839e4a1a16ba219ba7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-05-05 17:01:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0485af4a94cecb501c524facda7f6c533f5a33f5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-25 19:14:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4ff8898e0e8821ed1d344eac4ef4916f2ee3428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 19:15:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a2aeb2a569c574f9ef4b978c7d0dc0e8cbb452d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:37:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7730797d7a734427a82c838e4fa2a32d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84721088e6dd48749f0bec7979c1d56d : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40966d85e94247ab850329964cde20ba : SUCCESS in 17m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59e15998ea5c4db0853e52ce59c86d0a : SUCCESS in 18m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74a3fa3e6804d60ac3473688f03fead : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f322c43e87294e8d98857c194fd5b377 : SUCCESS in 59m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bec1018be67d4ce2adf2be88a5693034 : SUCCESS in 1h 36m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c79e22fe5264baa8c8b5cd0f3ff1ab1 : SUCCESS in 8m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e9e700d3d54c49b0bf6b9831e97bd1dc : SUCCESS in 28m 26s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/ca849fefcf8a4c3ca28625872be5c3e0 : SUCCESS in 28m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec85cf38132491ea20456875968bbed : SUCCESS in 52m 18s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f5fa83c968e3429d953ddd370097953a : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9cd728713f840225fab93667633c043eafa3cc47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:37:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a76b2575366141caea35a94c451481deb5b6f9fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:38:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/895bdfefb95b4b3a9aa1702c60928c5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f042fd7ae7f6446b98493dd71492a4bf : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06e94cd33e5a4b479e4732da7d8611b3 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68bfb685d12937dde11d1a335bd992203ec7c293","revisions":{"e3c31bfd88ac478a6d35df7b1ebbc72c36951253":{"kind":"REWORK","_number":1,"created":"2021-05-03 20:43:49.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/17/789417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/789417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/789417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/789417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/789417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/789417/1"}}},"commit":{"parents":[{"commit":"10057702ac361213e74472ec1d0d4e4c4a041f09","subject":"Merge \"Retry update_user when sqlalchemy raises StaleDataErrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/10057702ac361213e74472ec1d0d4e4c4a041f09"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-05-03 20:37:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-05-03 20:37:35.000000000","tz":0},"subject":"Only log warnings about token length when length exceeds max_token_size","message":"Only log warnings about token length when length exceeds max_token_size\n\nPreviously, the fernet token provider would log warnings when a fernet\ntoken exceeded 255 characters, which is common for LDAP-backed\ndeployments. The warning is always issued, even when operators configure\nkeystone\u0027s max_token_size to a higher value, causing confusion because\nit appears the configuration value is silently ignored.\n\nThis commit fixes that issue by using the max_token_size configuration\nparameter consistently in the fernet token provider.\n\nChange-Id: I4bb54aac9b950d59082a4468203a3249790839d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3c31bfd88ac478a6d35df7b1ebbc72c36951253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3c31bfd88ac478a6d35df7b1ebbc72c36951253"}]},"branch":"refs/heads/master"},"68bfb685d12937dde11d1a335bd992203ec7c293":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-03 20:44:17.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/17/789417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/789417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/789417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/789417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/789417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/789417/2"}}},"commit":{"parents":[{"commit":"10057702ac361213e74472ec1d0d4e4c4a041f09","subject":"Merge \"Retry update_user when sqlalchemy raises StaleDataErrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/10057702ac361213e74472ec1d0d4e4c4a041f09"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-05-03 20:37:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-05-03 20:40:31.000000000","tz":0},"subject":"Only log warnings about token length when length exceeds max_token_size","message":"Only log warnings about token length when length exceeds max_token_size\n\nPreviously, the fernet token provider would log warnings when a fernet\ntoken exceeded 255 characters, which is common for LDAP-backed\ndeployments. The warning is always issued, even when operators configure\nkeystone\u0027s max_token_size to a higher value, causing confusion because\nit appears the configuration value is silently ignored.\n\nThis commit fixes that issue by using the max_token_size configuration\nparameter consistently in the fernet token provider.\n\nCloses-Bug: 1926483\n\nChange-Id: I4bb54aac9b950d59082a4468203a3249790839d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/68bfb685d12937dde11d1a335bd992203ec7c293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/68bfb685d12937dde11d1a335bd992203ec7c293"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[]}
