)]}'
{"id":"openstack%2Fkeystone~790444","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~Id51241989b22c52810391f3e8e1cadbf8613d873","project":"openstack/keystone","branch":"stable/train","topic":"bug/1688137-stable/train","hashtags":[],"change_id":"Id51241989b22c52810391f3e8e1cadbf8613d873","subject":"Hide AccountLocked exception from end users","status":"MERGED","created":"2021-05-10 15:53:27.000000000","updated":"2021-06-04 19:14:52.000000000","submitted":"2021-06-04 19:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790444-bug/1688137-stable/train","cherry_pick_of_change":759940,"cherry_pick_of_patch_set":8,"meta_rev_id":"a909583c2b5a190168e4eab7e6a6325e75d36b39","_number":790444,"virtual_id_number":790444,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 19:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"recommended":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2021-06-02 14:51:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2021-06-03 15:57:07.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2021-06-04 17:35:08.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 17:45:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 14:51:25.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-06-03 15:57:07.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"}],"messages":[{"id":"f10451624687dc2a49b5a928e6ec03e2cc87f300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-05-10 15:53:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1c69e377e8624d9ca15d6410c0bc3d2790609471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 17:45:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a86fc469d5465ebe96764003cbb1f3 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664d6c76e2c74056bf191925950ebfec : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c54ba01b3dc4be19cf7be38ace5eaf4 : SUCCESS in 19m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f44769dcea24939b0840eaf57ee9364 : SUCCESS in 18m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e693e891c17a4ef2b5275edabca606c2 : SUCCESS in 18m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b320a163b34e4941bee8a68722dcf73e : SUCCESS in 12m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cfb1102f39d74ce9897cfbdeecd0a2ba : SUCCESS in 1h 44m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9e6b09caccd14b0cacb168293a17bf4f : SUCCESS in 1h 00m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c05fabf0a2224a718e8a0ecbcd3e6551 : SUCCESS in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5efda48a2eb544679cdf6f7ce342d7d7 : SUCCESS in 1h 20m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac2bda4336924c0ab2b94d34a78ede76 : SUCCESS in 8m 14s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5cfc7d876bc4dfba27a237168618d57 : SUCCESS in 33m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e0e20cb7181a403b8fd98adb59e28231 : SUCCESS in 35m 37s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d54ef4aafc534554b2d639e96bc3cbca : FAILURE in 7m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/41731ba3d62d42e399a282fee5661ce0 : FAILURE in 12m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/1d502f04762f4235b2c67aeafd0bb11c : FAILURE in 13m 47s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/dbd480d33be74a41ac66769eaa49cc9f : SUCCESS in 20m 07s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d8b0944d13d348e493f8b8b44fba2eae : SUCCESS in 39m 42s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/159cef938b1c4873b1183aaae4e0cc6b : SUCCESS in 1h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1291f2f0cdfe4580977b7db4449d150f : SUCCESS in 51m 49s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/e505bfbbdc8e409f806b4e14d882f1a2 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8ab8c311a0e4cd7d040e4d82debee188e7bfdc70","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-06-02 14:51:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"af2f43e69a30d0322ff8802db7d733b847bdbc2b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2021-06-03 15:57:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e111739798f90aec38ba0ee879d3a5f1ecb87d96","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2021-06-04 17:35:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"008281db063cf25c4798c2bde4a2dc0f9aef5944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:36:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2848705497cb8e36c133c0fe9193e76bd6255e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:13:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ed4029f1db48bf9f7f0d5fb889fc46 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22aa07faf2074bcea6096a7a826c2119 : SUCCESS in 13m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/712bb75049fd4222b893882acd60f889 : SUCCESS in 13m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01fb82bac69f48bfb8e4cf25dc70d565 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3ad8bb0e3f4cf49e859d7a8e89d645 : SUCCESS in 12m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aebf63dab2724b17943db02147e66eb0 : SUCCESS in 1h 31m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fb8c4e16e254442691ff73df8eca5dda : SUCCESS in 56m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/79c8a39851a748e2b7a7ee2060feebc1 : SUCCESS in 57m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1c2ae4ad5f742829abcd9281cd86ae8 : SUCCESS in 1h 25m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b323355e4a7c44b6b3b51a0286b87ecb : SUCCESS in 7m 44s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa9a707d339249248f7b3aaafe4af773 : SUCCESS in 31m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d92308b121754b50bf77219f4739ff6e : SUCCESS in 32m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/40b9167feb5842b9a84fb1c559d9833e : SUCCESS in 1h 00m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83e5f051b5f34e138f063b02f71ca120 : SUCCESS in 57m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/875b8b1517724d5bb1dd49dd72de58b4 : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d9057c42a94bf39d225ea5d6bf9f8dc11db8234e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:13:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a909583c2b5a190168e4eab7e6a6325e75d36b39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:14:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8c81237d7240f5bf4463a84f72271f : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58972bce01f4428ea89d1cb566890059 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b573ae7d1c20e0ebfbde79bbe7538a09589c75d","revisions":{"1b573ae7d1c20e0ebfbde79bbe7538a09589c75d":{"kind":"REWORK","_number":1,"created":"2021-05-10 15:53:27.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/44/790444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/44/790444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/790444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/790444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/790444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/44/790444/1"}}},"commit":{"parents":[{"commit":"fd5fa12977d07b4441307ac52dd0db7966ad3a4f","subject":"Merge \"Retry update_user when sqlalchemy raises StaleDataErrors\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fd5fa12977d07b4441307ac52dd0db7966ad3a4f"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-10-27 20:22:04.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-05-10 15:53:27.000000000","tz":0},"subject":"Hide AccountLocked exception from end users","message":"Hide AccountLocked exception from end users\n\nThis change hides the AccountLocked exception from being returned\nto the end user to hide sensitive information that a potential\nmalicious person could gain insight from.\n\nThe notification handler catches the AccountLocked exception as\nbefore, but after sending the audit notification, it instead\nbubbles up Unauthorized rather than AccountLocked.\n\nCo-Authored-By: Samuel de Medeiros Queiroz \u003csamueldmq@gmail.com\u003e\n\nChange-Id: Id51241989b22c52810391f3e8e1cadbf8613d873\nRelated-Bug: #1688137\n(cherry picked from commit ac2631ae33445877094cdae796fbcdce8833a626)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1b573ae7d1c20e0ebfbde79bbe7538a09589c75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1b573ae7d1c20e0ebfbde79bbe7538a09589c75d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
