)]}'
{"id":"openstack%2Fkeystone~799670","triplet_id":"openstack%2Fkeystone~master~I6a0a51fb2d68ec994f610f5de41c9760a4867730","project":"openstack/keystone","branch":"master","topic":"oslo.db-10.0.0","hashtags":[],"change_id":"I6a0a51fb2d68ec994f610f5de41c9760a4867730","subject":"Adapt to oslo.db 10.0.0","status":"ABANDONED","created":"2021-07-06 13:31:28.000000000","updated":"2021-07-06 15:29:06.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f3be3f69f9b6d6fd1b20811b4aff28d0ec1b5968","_number":799670,"virtual_id_number":799670,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-06 14:31:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2021-07-06 13:39:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 13:39:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-06 14:31:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e38ccc376dd88add07e7462f73cd2722a80ac56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-06 13:31:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2eb1cefe99dfe24e7973fc62db226396e1fad671","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-06 13:39:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nJinx [1]. Even the topic is basically the same, heh. Happy to proceed with this but I do think the tests should be removed.\n\n[1] https://review.opendev.org/c/openstack/keystone/+/799672","accounts_in_message":[],"_revision_number":1},{"id":"746bef0c3d94d4e63a5a1d3462839a96079220a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 14:31:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/880e4e103021442286b0ca95bf4b864c : SUCCESS in 26m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e06ffc6c1a45a7924847ea1cda0cbe : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c98cd8de5a6d49f49b834855381d3e55 : SUCCESS in 14m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/316386967eb54ba3935cb12c1555b3d0 : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5f5f29301d453eb87b8e77ae35da81 : FAILURE in 11m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/68b81b5eadbe4ed59fc231a911272769 : FAILURE in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"061627a272ba284c61c8a586f86f1056fc038eef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 14:34:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/893387d9cb8b4a858dcc0c0563d2819f : TIMED_OUT in 1h 01m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ba442c0930a434095df8092ea9efa02 : TIMED_OUT in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3be3f69f9b6d6fd1b20811b4aff28d0ec1b5968","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-06 15:29:06.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/keystone/+/799672","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e06b743d07383a00826e063b9886b928cdc8c59","revisions":{"8e06b743d07383a00826e063b9886b928cdc8c59":{"kind":"REWORK","_number":1,"created":"2021-07-06 13:31:28.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/70/799670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/70/799670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/799670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/799670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/799670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/70/799670/1"}}},"commit":{"parents":[{"commit":"252733dce7ae7d335b982d36e96b64e7a29d4ef2","subject":"Merge \"Make DB queries compatible with SQLAlchemy 1.4.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/252733dce7ae7d335b982d36e96b64e7a29d4ef2"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-07-06 13:24:18.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-07-06 13:29:01.000000000","tz":120},"subject":"Adapt to oslo.db 10.0.0","message":"Adapt to oslo.db 10.0.0\n\nIn oslo.db 10.0.0 the [sql] and [database]idle_timeout config option is\nremoved. This config option was deprecated in Icehouse and replaced with\n[default]connection_recycle_time config.\n\nChange-Id: I6a0a51fb2d68ec994f610f5de41c9760a4867730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8e06b743d07383a00826e063b9886b928cdc8c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8e06b743d07383a00826e063b9886b928cdc8c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
