)]}'
{"id":"openstack%2Fkeystone~802751","triplet_id":"openstack%2Fkeystone~stable%2Fqueens~Iaac49cd66fdaa2871b8c012b1c5c8685cb6453a8","project":"openstack/keystone","branch":"stable/queens","topic":"bug/1938323","hashtags":[],"change_id":"Iaac49cd66fdaa2871b8c012b1c5c8685cb6453a8","subject":"[queens-only] Ensure catalog is present in token validation response","status":"ABANDONED","created":"2021-07-28 16:13:46.000000000","updated":"2022-07-15 16:49:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc0288618ee1ed22eb60ee3a22726bce4f3e4246","_number":802751,"virtual_id_number":802751,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-28 17:50:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 17:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83830c0652f08cd0ad12eb22e7407086a89c3c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-07-28 16:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a671b7ec4a4781a5176b8a98cfa62485f3d79902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-07-28 16:15:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d18e9a5131eef701fd06c6da784fd7c4aa779f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:50:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f7c7e7b33d421491e1141dae128859 : SUCCESS in 11m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/132b33eb287e4efc9c0421c1b1737669 : SUCCESS in 14m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/59b5511fe0b54d1989480599bc56c7e0 : SUCCESS in 16m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7482cff9b38941b082a273fa58e0045c : SUCCESS in 10m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8a33b16d52b4f3ea7e9eac2c5f2d362 : SUCCESS in 1h 16m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cfbe0707fed84918a1675c027670da12 : FAILURE in 12m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe2fc61bc67742fab8ac1db4b31f03ca : SUCCESS in 1h 28m 45s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/acb0eaebf3ba425d805fcfba03690db7 : FAILURE in 17m 47s\n- keystone-dsvm-functional-v3-only https://zuul.opendev.org/t/openstack/build/3f51944771bc4dd495b3d17d50ffce00 : FAILURE in 10m 09s (non-voting)\n- keystone-dsvm-py35-functional-v3-only https://zuul.opendev.org/t/openstack/build/7a95cb807b3c4dd8963671ba23c3a299 : FAILURE in 14m 27s (non-voting)\n- openstack-ansible-keystone-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6b945ffde384a71965b2e907597de8c : FAILURE in 8m 25s (non-voting)\n- legacy-keystoneclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/9da124126f714e76804c3039c3990f99 : FAILURE in 15m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/269dcd56c7b34422a6b23c5926af9fda : SUCCESS in 16m 11s\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/adbd8138a3494f818491d37984ea3c34 : SUCCESS in 42m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cc0288618ee1ed22eb60ee3a22726bce4f3e4246","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-15 16:49:57.000000000","message":"Abandoned\n\nAbandoning unmerged stable/queens changes.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02c35998fd407e320fb85fd923154c5554fed0dc","revisions":{"ac6f0a6c9091da3b8f323f1c33199aec6586eddf":{"kind":"REWORK","_number":1,"created":"2021-07-28 16:13:46.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/51/802751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/802751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/802751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/802751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/802751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/802751/1"}}},"commit":{"parents":[{"commit":"0153ab781d387cc0fdb3c93fbd7be5c897bcf933","subject":"Filter by entity_type in get_domain_mapping_list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0153ab781d387cc0fdb3c93fbd7be5c897bcf933"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-07-23 14:28:19.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-07-28 16:12:28.000000000","tz":0},"subject":"Ensure catalog is present in token validation response","message":"Ensure catalog is present in token validation response\n\nunless explicitly asked otherwise.\n\nIn Queens, token caching includes service catalog also.\n\nIf token was created with `?nocatalog`, it will be cached w/o catalog,\nand on token validation response will return token infoa also w/o catalog\nuntil cache is invalidated.\n\nSuch catalog-less tokens break other services (e.g. Nova) as they rely\non catalog being present in the results of token validation returned\nby keystonemiddleware.\n\nChange-Id: Iaac49cd66fdaa2871b8c012b1c5c8685cb6453a8\nCloses-Issue: #1938323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ac6f0a6c9091da3b8f323f1c33199aec6586eddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ac6f0a6c9091da3b8f323f1c33199aec6586eddf"}]},"branch":"refs/heads/stable/queens"},"02c35998fd407e320fb85fd923154c5554fed0dc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-28 16:15:20.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/51/802751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/802751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/802751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/802751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/802751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/802751/2"}}},"commit":{"parents":[{"commit":"0153ab781d387cc0fdb3c93fbd7be5c897bcf933","subject":"Filter by entity_type in get_domain_mapping_list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0153ab781d387cc0fdb3c93fbd7be5c897bcf933"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-07-23 14:28:19.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-07-28 16:14:45.000000000","tz":0},"subject":"[queens-only] Ensure catalog is present in token validation response","message":"[queens-only] Ensure catalog is present in token validation response\n\nunless explicitly asked otherwise.\n\nIn Queens, token caching includes service catalog also.\n\nIf token was created with `?nocatalog`, it will be cached w/o catalog,\nand on token validation response will return token infoa also w/o catalog\nuntil cache is invalidated.\n\nSuch catalog-less tokens break other services (e.g. Nova) as they rely\non catalog being present in the results of token validation returned\nby keystonemiddleware.\n\nChange-Id: Iaac49cd66fdaa2871b8c012b1c5c8685cb6453a8\nCloses-Bug: #1938323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/02c35998fd407e320fb85fd923154c5554fed0dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/02c35998fd407e320fb85fd923154c5554fed0dc"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
