)]}'
{"id":"openstack%2Fkeystone~806205","triplet_id":"openstack%2Fkeystone~master~I54021986d17c57d7733d53caa4032c2767eaf25e","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"I54021986d17c57d7733d53caa4032c2767eaf25e","subject":"Fix typos in ec2 credential policies","status":"MERGED","created":"2021-08-26 14:53:29.000000000","updated":"2021-10-05 23:37:35.000000000","submitted":"2021-10-05 23:33:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806205","meta_rev_id":"7b345f84c4834f591ad143aab51cf726426e840e","_number":806205,"virtual_id_number":806205,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 23:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2021-09-28 15:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2021-09-24 20:10:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":2,"date":"2021-09-24 19:21:52.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2021-09-28 15:23:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 14:58:40.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2021-08-26 16:02:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-26 17:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 20:47:05.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-09-24 20:10:18.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"}],"messages":[{"id":"6ea518f28768b800e035708640b53d956c2c4683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-26 14:53:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b54ea87b72070f1060364e6aaf8532b4737ca173","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-26 14:57:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"656ae743ddab76af79170169bf9a5d596b9e96e0","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-08-26 14:58:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c5ec41efde6d326d409f6ccbff6d9ac8e598c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:02:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8143e7495b8e45299081e84903d3007d : TIMED_OUT in 1h 02m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58b6edfc3fb54f9ea66006d918ab3f51 : SUCCESS in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e530d064a3c4d6f426b2f553dded3dea138f4db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 17:46:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f7b5cc9751543f2842f66379e8101e4 : SUCCESS in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53222ded9f2a4ba18798edfd4a0dcdfc : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7970a639c08a4b99bfd89d3695318c90 : SUCCESS in 14m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e550d52c5e4eb6a4cdd2efa6175dc7 : SUCCESS in 14m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8752ae93c63a40718f2d14161e547d69 : SUCCESS in 15m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/30d8d5bbe2d1410983e888351dd3ca13 : SUCCESS in 1h 08m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5f629f177624d3087e5c7f10304fc9f : FAILURE in 1h 18m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/94be34175e06443a99de8b454c9c6075 : SUCCESS in 37m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/77bd880af8a64a81b4e30cc415a714ce : SUCCESS in 43m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4fd1aecd1994e0cb95e493f9d15ce86 : SUCCESS in 37m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/1fe2f1e6d5c0403285d4076dd598ce87 : SUCCESS in 27m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a2de6d772ceb4abbbfda95a41f448f14 : SUCCESS in 15m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/90cd8bd7d04548a386be69f343bac3a7 : FAILURE in 13m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fdeff9e90ca48edbbe1f990b20d7dfc : SUCCESS in 58m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/99b02f6c07e844419c912a61fa69c498 : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4aad746505bdae2386e7117f031d3767ac0017ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-27 02:33:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"466e0f57158bed71046e92adea827dd3db12a679","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 03:23:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/beb64ee98e9f42c7ad0ba4b23350b776 : SUCCESS in 44m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f1616588dc54d7abc6d7474bb8f8a73 : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"74fdff2be586d78b7438d9c7da31661f059c8733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 03:45:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5687612cee86403f99c48c716bdfd721 : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37672b61d8144624a9aaf35e077e30ec : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/348d84af19b84bc0aaca92d14089a19a : SUCCESS in 12m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/697e1f70b6bf4a5ea8b354b0a2850c95 : SUCCESS in 16m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd24e81fff024e809d80de654a7828b4 : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/18dec2882359411381f38da32edce288 : SUCCESS in 1h 01m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2671f4a8d057429c919863a7327bbc8f : SUCCESS in 1h 09m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f6eefadd455d433bac884129dff21a1e : SUCCESS in 26m 15s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/fb0aabaf1a924878bd13917ba7627f3e : SUCCESS in 41m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/199dd3beebc645bc982f43fbcdc63f1d : SUCCESS in 29m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/0c5d8c85bb664a388f1b1d7f4aa6449f : SUCCESS in 22m 29s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/dd3fe0d0ac2a4243bb63fbf9e1864377 : SUCCESS in 13m 48s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ba4afdf89d4f487d8999541feb40b496 : FAILURE in 14m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb95ed2fbe89440f943c37e454b5e28f : SUCCESS in 44m 50s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1e54640dd81b4f639915fea0d86a27f8 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":3},{"id":"9c76826fc930d51562ddd1ceba46367cc63f810a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-24 19:21:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"22263e0a46129c017f3c3d9a94571808d4af853c","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2021-09-24 20:10:18.000000000","message":"Patch Set 3: Code-Review+1\n\nI would +2 if I could","accounts_in_message":[],"_revision_number":3},{"id":"c19171b82f3eeba293d20b471c588892d635c5cd","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-09-28 15:23:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ded23ce9e59d0c82075b80a86e70b03a1088bbfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 15:24:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"770c2e465a98dd454b3290c8e631a49830bb4651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 16:28:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7000da76d99b49a7bed5d504a1e8963b : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b55759419ef4b0ea42ab7583185dea7 : SUCCESS in 15m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef9208b908864c99a4bb078562027ec7 : SUCCESS in 14m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f7f323414b4c7292fea092d1328e49 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a8724857325f4754968ac775e4a49e91 : FAILURE in 23m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04927e83ef3040fcabdc8d4acac3f37c : FAILURE in 24m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8becf9b645244f0688071ee51a387669 : FAILURE in 24m 03s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/23ed145a5c864d6da3fb41154c5b5807 : FAILURE in 23m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4bc804f12ba47cebb1005c82c760e36 : FAILURE in 24m 02s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d1c4d19356df4a3baf82e41f47bb1aef : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":3},{"id":"9f3cbb2d58322e9dc47047f10cdd1aef7f62db26","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-28 18:26:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fd8dcc50bd02c2d08e5e9067d2cdef7df41d1fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 19:22:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df769424d0734bc680762a716d43d0b3 : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706ab659d9cf41c7bd049872f1451168 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab932848c94c4db3ad2a3fb45025aafb : SUCCESS in 14m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1378cf3954d4b93b0c221a4e8d87fb7 : SUCCESS in 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aebb53b3fd341969d0f9351d5364cc8 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f1db956f3ec443dd8fda213f289e0af9 : FAILURE in 30m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85232f0836b741fd8748952547ed1c08 : FAILURE in 33m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8a804106cbac493fbe30accddcb6c42a : FAILURE in 38m 54s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/0ec2ffc6c3c74b97b35cd04b750cf9f2 : SUCCESS in 39m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/121b4e4e26594e36af680c91ffb30c0b : FAILURE in 29m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/a0b1488b282f46198557474e01d0918d : FAILURE in 46m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/16a99538d3054a1cb964b14c56165483 : SUCCESS in 25m 26s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0cb2c94a00e9458b86d2d559ca187c16 : FAILURE in 18m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18d92f5b008945bdad90983aed066e46 : FAILURE in 27m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6f9a74df9ebd4afe8b2ef51f1dd90f66 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4d141d3f941c52625407276ba821bc7842d90c2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 19:30:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6726ac0029264c0c84a8734e96b346af : TIMED_OUT in 1h 01m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd9c87e759ed4edba38aa6dfdd7381b9 : TIMED_OUT in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bfed5883245f369ab58fd22bf3199c046dabd1bf","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-05 20:44:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0064d9bf15e22b874b1830dd854ecc54ac1d947d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 21:48:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/603dff1ea4f9445c9f4f26cb2136cd15 : TIMED_OUT in 1h 01m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b039393d05ad4b87838aad94ccd8e19d : TIMED_OUT in 1h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"721c0ae6413aa36c420b7ad467a041ea0b2ef94b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 22:14:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/373139f6e34f4a37b0e888f69d42ae14 : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06e6114ae424fb38022c92c60f50df4 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6623296aa464b3b939c13134956a15a : SUCCESS in 13m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54062f34fc4748e08605fcb9aa90a540 : SUCCESS in 16m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206f6aa4a70f4168bc65e4f006b652cf : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/8027d4b6f66d47ac80813589812168ed : SUCCESS in 1h 00m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/327ad3b753f64cea853bec526ffcac06 : SUCCESS in 1h 15m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/18f9c78ff36049e193bad6cf14334067 : SUCCESS in 38m 59s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/289b473364dc4b0c82c0e862b4dd672b : SUCCESS in 40m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/858917089d9349dc889b4b1a672b4305 : SUCCESS in 34m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/99d1038fe3e8405ab6af08e2f66777b8 : SUCCESS in 32m 41s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ed2e56e24bfa4417bad5eb61150649e5 : SUCCESS in 18m 06s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c8ff8fbe444741f8aad70c8a24f19d80 : FAILURE in 17m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca28f0c463404ed8b660661ba00328ac : SUCCESS in 52m 41s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dc2a5d803d5847eeb3545cb096d767bd : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":3},{"id":"487de68f8e3df90fd540d8f12d93d52450e1b5eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 22:15:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"56d6e1b7b6216f37a90b50f0a1c284baee41b196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 23:33:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63217681b9f24010ae68729d6a9aee88 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a66619e663304f999ea1b9f6b49b2174 : SUCCESS in 14m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/081b53ef5dac45fba13a64843fc1526c : SUCCESS in 16m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95fc73b3dee44ab8a92713f6df5f65d5 : SUCCESS in 15m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/f724ba1b771246b6904c0931a44d6d38 : SUCCESS in 55m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dd9aff2188a4fb9932b898241eaaa69 : SUCCESS in 1h 13m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a76a56881b014197b9b92fc0e7435ab0 : SUCCESS in 32m 18s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/1febbb7ea99c490ea81f6cb3720dc0ac : SUCCESS in 34m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c8ddf66dcf3427cba3d041086a7a911 : SUCCESS in 43m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3445b399a42044d0ba1945cad6277e2d : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":3},{"id":"aab92f16ea3c72e49dd643d571e9dc498afda5cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 23:33:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7b345f84c4834f591ad143aab51cf726426e840e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 23:37:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9d6925e8344de9a28679bba9f4a354 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"82da8824df0f56ef4e137805bf32d647cef1ea59","revisions":{"aed56220ea75241f4e69de82df3c14f88eebf21c":{"kind":"REWORK","_number":1,"created":"2021-08-26 14:53:29.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/05/806205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/806205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/806205/1"}}},"commit":{"parents":[{"commit":"0c3d2d656371feb509aee5def1b73da438ace0bf","subject":"Merge branch \u0027master\u0027 of https://opendev.org/openstack/keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0c3d2d656371feb509aee5def1b73da438ace0bf"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:50:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:50:35.000000000","tz":0},"subject":"Fix typo in ec2_create_credential policy","message":"Fix typo in ec2_create_credential policy\n\nThere was a trailing s in this policy and it caused the policy names to\nmismatch, which causes confusion with the rendered policy files and\npotentially causes uses with deprecation logic.\n\nBased on the code, only ec2_create_credential is used, so this updates\nthe deprecated policy to be accurate with what\u0027s used in the code.\n\nChange-Id: I54021986d17c57d7733d53caa4032c2767eaf25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aed56220ea75241f4e69de82df3c14f88eebf21c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aed56220ea75241f4e69de82df3c14f88eebf21c"}]},"branch":"refs/heads/master"},"e14d673d69ab18910a59da17789f103e3a0905a9":{"kind":"REWORK","_number":2,"created":"2021-08-26 14:57:19.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/05/806205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/806205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/806205/2"}}},"commit":{"parents":[{"commit":"50f0a50cf4d52d3f61b64713bd4faa7a4626ae53","subject":"Add FIPS check job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/50f0a50cf4d52d3f61b64713bd4faa7a4626ae53"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:50:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:55:36.000000000","tz":0},"subject":"Fix typos in ec2 credential policies","message":"Fix typos in ec2 credential policies\n\nThere was a trailing s in two of these policies and it caused the policy\nnames to mismatch, which causes confusion with the rendered policy files\nand potentially causes uses with deprecation logic.\n\nChange-Id: I54021986d17c57d7733d53caa4032c2767eaf25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e14d673d69ab18910a59da17789f103e3a0905a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e14d673d69ab18910a59da17789f103e3a0905a9"}]},"branch":"refs/heads/master"},"82da8824df0f56ef4e137805bf32d647cef1ea59":{"kind":"REWORK","_number":3,"created":"2021-08-27 02:33:30.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/05/806205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/806205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/806205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/806205/3"}}},"commit":{"parents":[{"commit":"e057378b82efe27c2ac9dd277ead587836c759b5","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e057378b82efe27c2ac9dd277ead587836c759b5"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:50:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-27 02:32:13.000000000","tz":0},"subject":"Fix typos in ec2 credential policies","message":"Fix typos in ec2 credential policies\n\nThere was a trailing s in two of these policies and it caused the policy\nnames to mismatch, which causes confusion with the rendered policy files\nand potentially causes uses with deprecation logic.\n\nChange-Id: I54021986d17c57d7733d53caa4032c2767eaf25e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/82da8824df0f56ef4e137805bf32d647cef1ea59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/82da8824df0f56ef4e137805bf32d647cef1ea59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
