)]}'
{"id":"openstack%2Fkeystone~806456","triplet_id":"openstack%2Fkeystone~master~Ie588736bf43ee102d3a72aa9f44ced0523e58e1f","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"Ie588736bf43ee102d3a72aa9f44ced0523e58e1f","subject":"Add some missing parameters in docs of oauth","status":"ABANDONED","created":"2021-08-28 02:56:01.000000000","updated":"2021-08-28 03:18:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e74446c1600a0042a7b7db1a89b0656722684f39","_number":806456,"virtual_id_number":806456,"owner":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-28 03:13:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-28 03:13:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"124d34e9d4a1ec7e5a838a1fda006199ec5e5ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"date":"2021-08-28 02:56:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2987e4008d93962f1f0d32642fae1e4f8cc03de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 03:13:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/75fe1d79f12c48dfbc79c289b8cf033c : FAILURE in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f482020fc5a540d2b17c771db71a19f9 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add950aaa4e7423da61c8963cb7d3959 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e74446c1600a0042a7b7db1a89b0656722684f39","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"date":"2021-08-28 03:18:25.000000000","message":"Abandoned\n\nfailed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2871159298813b5dde4cb4373613f1211d33191","revisions":{"f2871159298813b5dde4cb4373613f1211d33191":{"kind":"REWORK","_number":1,"created":"2021-08-28 02:56:01.000000000","uploader":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"ref":"refs/changes/56/806456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/56/806456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/806456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/806456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/806456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/56/806456/1"}}},"commit":{"parents":[{"commit":"f03ff806c157fb0cf14895db3b31294692f544fd","subject":"Merge \"Update local_id limit to 255 characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f03ff806c157fb0cf14895db3b31294692f544fd"}]}],"author":{"name":"budebao","email":"531648642@qq.com","date":"2021-08-28 02:55:45.000000000","tz":480},"committer":{"name":"budebao","email":"531648642@qq.com","date":"2021-08-28 02:55:45.000000000","tz":480},"subject":"Add some missing parameters in docs of oauth","message":"Add some missing parameters in docs of oauth\n\nChange-Id: Ie588736bf43ee102d3a72aa9f44ced0523e58e1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f2871159298813b5dde4cb4373613f1211d33191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f2871159298813b5dde4cb4373613f1211d33191"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
